)]}'
{"id":"openstack%2Frequirements~828903","triplet_id":"openstack%2Frequirements~master~I4c644e473c0be569f36056e1505bf0eac76df51a","project":"openstack/requirements","branch":"master","topic":"coverage-py39","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-02-11 22:57:14.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4c644e473c0be569f36056e1505bf0eac76df51a","subject":"Constrain \u0027coverage\u0027 for python 3.9","status":"ABANDONED","created":"2022-02-11 16:16:18.000000000","updated":"2022-02-11 22:57:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"833cedc5fbf588f4861683dbe63bf4dcd55f59c2","_number":828903,"virtual_id_number":828903,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-02-11 17:51:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-1,"date":"2022-02-11 17:39:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-11 16:40:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-11 17:39:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-02-11 17:39:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-02-11 17:51:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a82e61c9fbe521fe9f432a471d3a38fa22ff563c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-11 16:16:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd7a79ee3e6f93f060a82c94140624118cd8cb5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 16:40:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/763e09a83fc24e20b9868456146654df : FAILURE in 18m 02s","accounts_in_message":[],"_revision_number":1},{"id":"c255f4830e92a99515b8a0c2dc37b9057c14714a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-11 17:39:42.000000000","message":"Patch Set 1:\n\nusing https://review.opendev.org/c/openstack/project-config/+/828909 instead","accounts_in_message":[],"_revision_number":1},{"id":"511b8c6fa720da8aba2605311ca8ecedcc2b1857","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-11 17:39:49.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"e6bccae2ea376360fa5ed345b5f4bbf8a0bae05d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 17:51:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03aa84cb4b434e869c29ac76ee6dfc6c : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c6ebb2d058a4eb68fc5d3aa808d8cac : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c6926c285d1489f8bfe7769f6af99d8 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91e1d0a8d0f94513bab300f00807dcff : SUCCESS in 5m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3c853b3c0da04dda8b11849bf5f72cde : SUCCESS in 4m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/864e6a9558d14246ad057c83573adc28 : SUCCESS in 4m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ebfca2c380a54708ae8efcc896f356fb : SUCCESS in 10m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3098249ae38948999fbbb1d4a3901d1a : SUCCESS in 11m 22s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/5c340b0be3254ff79e21f892d1797c88 : SUCCESS in 5m 06s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/bc475120e2c040c88eb23378b6c41f20 : SUCCESS in 10m 35s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/013f3cd2d91d46b18a1c835e9ed0a5bb : SUCCESS in 6m 59s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/498ae3b319b340c88bc1b9f256a09a89 : SUCCESS in 5m 42s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/fe0457383a9241d49c6360b8702a18bd : SUCCESS in 11m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/99549fe776bc448a84a43870082252b6 : SUCCESS in 8m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c9ec513b25b74c398188c36ed757ddd9 : SUCCESS in 10m 20s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9bc2efccdbde4570ae9ffac476edc627 : SUCCESS in 13m 29s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/88705dcae4a74f1d96717776717bbcbb : SUCCESS in 4m 19s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/436472d06478497d9e12ce914a72c6d6 : FAILURE in 30m 05s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/469b1936cf7d495693be8815723c3dab : SUCCESS in 5m 05s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/cc9637b564b7469a8f1a6f82bf3e7692 : SUCCESS in 26m 31s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/cfe191ca3190473fb952cacf72aa1514 : SUCCESS in 17m 53s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/787ed5505cd8445ca911916ebe930ac2 : SUCCESS in 5m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b4f1f9d5b31b4b0ea55d8c4effd118fe : SUCCESS in 3m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2b6890a1e2434dd890dfc5f7e78bfa89 : SUCCESS in 20m 43s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/79cc7d2b38014110b73f2bfee088d084 : SUCCESS in 5m 48s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/218e3b8c6e554729a9c060e64990cdcf : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/043c9720f4bd420c9820ce0a3a1d91d2 : SUCCESS in 6m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3370e42b3ec04461ba42dc152ce602e1 : TIMED_OUT in 41m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/130ab324b26c47e3a61d2df6aab063d7 : SUCCESS in 43m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e9b6104b20e64ffcb8c889ecf0613aa5 : SUCCESS in 1h 28m 05s","accounts_in_message":[],"_revision_number":1},{"id":"833cedc5fbf588f4861683dbe63bf4dcd55f59c2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-11 22:57:14.000000000","message":"Abandoned\n\nThis will be fixed in the tooling: see https://review.opendev.org/c/openstack/project-config/+/828909","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c6943e855c2849151f1aaedb580578682e7ebc5b","revisions":{"c6943e855c2849151f1aaedb580578682e7ebc5b":{"kind":"REWORK","_number":1,"created":"2022-02-11 16:16:18.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/03/828903/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/828903/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/828903/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/828903/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/828903/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/828903/1"}}},"commit":{"parents":[{"commit":"9c67ccf4853898709067d61f8a52a9ea439a210a","subject":"Merge \"Readd neutron-fwaas and neutron-fwaas-dashboard to requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c67ccf4853898709067d61f8a52a9ea439a210a"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-02-11 15:51:24.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-02-11 16:13:14.000000000","tz":-300},"subject":"Constrain \u0027coverage\u0027 for python 3.9","message":"Constrain \u0027coverage\u0027 for python 3.9\n\nYoga supports py3.6, 3.8, and 3.9 and testing is supposed to be done\nin at least 3.6 and 3.9 [0].  For the \u0027coverage\u0027 package, the\nfollowing is currently in upper-constraints.txt:\n\n  coverage\u003d\u003d\u003d6.2;python_version\u003d\u003d\u00273.6\u0027\n  coverage\u003d\u003d\u003d6.3.1;python_version\u003d\u003d\u00273.8\u0027\n\n(We have to do something like this because python 3.6 is EOL and\npackages we consume are dropping support for it.)\n\nThis leaves the package unconstrained for python 3.9, and the pattern\nis repeated throughout the file for other packages.  It also leaves\npython 3.7 unconstrained.  (Technically, OpenStack doesn\u0027t support 3.7,\nbut I\u0027m not sure that should mean \"anything goes\" in the u-c.)\n\nThe point of this patch is to get the discussion going about what\npattern we should be using in the u-c file to handle this situation.\n\n[0] https://governance.openstack.org/tc/reference/runtimes/yoga.html\n\nChange-Id: I4c644e473c0be569f36056e1505bf0eac76df51a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c6943e855c2849151f1aaedb580578682e7ebc5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c6943e855c2849151f1aaedb580578682e7ebc5b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
