)]}'
{"id":"openstack%2Frequirements~829152","triplet_id":"openstack%2Frequirements~master~Icbaf6dedd5884458536492acdf3b5931908866dc","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2022-02-15 05:14:07.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Icbaf6dedd5884458536492acdf3b5931908866dc","subject":"Updated from generate-constraints","status":"MERGED","created":"2022-02-15 02:47:33.000000000","updated":"2022-02-15 06:56:02.000000000","submitted":"2022-02-15 06:52:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829152-openstack/requirements/constraints/noclob","meta_rev_id":"394b3a149588a5e8b08dcedaf7d5d554417588aa","_number":829152,"virtual_id_number":829152,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-15 06:52:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-15 05:14:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-15 05:14:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-15 03:08:37.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-02-15 03:28:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-15 04:21:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"41dd4fe5aaba12a25aa1a08176b8bb7d7dd3184d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-15 02:47:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f6b39806c64b3586815775534fc09e54a30d58a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-15 03:08:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4d63e9d8a297b6daf035378f23992d2faae18ad8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-15 03:08:59.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"2ad67f1fef17653d06d49b8bcf01c94047f24653","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 03:28:37.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/28f6fc2cd2cf4895bf3f7a5f999b670c : FAILURE in 9m 40s","accounts_in_message":[],"_revision_number":2},{"id":"3ce73df2e806bc002edd14cc17f4e88b3949c789","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 04:21:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f213a03ee5b14a28a5c615af60cab94d : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4a7e3ac6cf541e3861a73e5ee031017 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb83d219ead54ddea70c12ebd3e7c680 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2480980c51f4b8e82355363d0c24c97 : SUCCESS in 5m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a224ae4c7f84444583bae8528db47bd2 : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eebcf3a33d00483582a9921502319aac : SUCCESS in 5m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f86832b18fd44860a73ef704f8b14e07 : SUCCESS in 8m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/aec983dc3522461faccfcd33740ad168 : SUCCESS in 7m 33s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/dfed9586ad9c4654819c8b17e464b0e5 : SUCCESS in 6m 34s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9f4dbb15b8124bc199f5f3ea89a101c2 : SUCCESS in 11m 50s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/89b7aea962cc496ea8a474d7f0e65259 : SUCCESS in 5m 32s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c48eee07db5542da93097d9cdfbd082e : SUCCESS in 5m 06s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/10fb0fa5f67b4172a7ca4deacf54da94 : SUCCESS in 9m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/68041e5b13324573964e13e6ba37a31a : SUCCESS in 7m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1851e61fb3634c46bf0bba8d888aa703 : SUCCESS in 6m 56s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b1e88f460b75473789dae2e37eaed207 : SUCCESS in 14m 13s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2fa3ab662c3d4184a38fd1e3b97fd417 : SUCCESS in 4m 02s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9b6949707216459ebf848e3efcc49c07 : SUCCESS in 16m 09s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/465dc4a443974dd79320e9ca2da2a718 : SUCCESS in 4m 42s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ba92aee3de1e4200860792a3bca0de4b : SUCCESS in 22m 50s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/fc69e2f1ee4a4144bdbacc6ba232e359 : SUCCESS in 10m 39s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/8d4ef55552fe4d6d861fbc0bb7191330 : SUCCESS in 6m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0f5c6ebbb95540c1b3b62e54ccee5d9a : SUCCESS in 3m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1e816dd3feb94f9c879031d5d5a9e744 : SUCCESS in 19m 07s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/9d1dd93b684843179d460b27090b9eed : SUCCESS in 6m 49s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2dbba1ee62c440478351f528db17fe70 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/da5966b79e054175aa581de078a25512 : SUCCESS in 6m 39s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/97a1a82375104865882ab53bba06252d : SUCCESS in 10m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a8f5cf11f5f4adeb84d864ccd0cf6c5 : SUCCESS in 39m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b9ccce3fa381411d9869134d28ba6aa8 : SUCCESS in 1h 08m 10s","accounts_in_message":[],"_revision_number":2},{"id":"8a4ee0593c62feee23772f489ec45b1359d7ad40","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-15 05:14:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0e80d64686b6a0f172d3e99cea93bb5c0dd51a52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 05:14:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"667ead3a91e09430fcc14129af1cfa25586fa6fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 06:52:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c05761afb634311831cd60ee1cef5f0 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/690829685f37494384075378337f06b4 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1bbc57a6a72a405e9c13e7a10517938e : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb3fd67b48364e638f376f7ddf764f40 : SUCCESS in 5m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2108b2bebf3e452080f294ba9a5615db : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d7d11f697e10430a8a57c75b53e7d63a : SUCCESS in 4m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ae45591ca730461f9db59b7f2b828d5b : SUCCESS in 12m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f184ee1be1834bd8aecd72021491e2a1 : SUCCESS in 7m 58s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/25e9516f42224ab8a83029895be6168a : SUCCESS in 8m 35s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/de0a24ace4384385a53792584c91157f : SUCCESS in 6m 15s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/44389e32cec849568302cc4a1f576c75 : SUCCESS in 5m 14s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/85be9218e16145d9b8daa8fcb2c167e0 : SUCCESS in 12m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/718e56df3c924c1c91ff2c753258bfcd : SUCCESS in 8m 00s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e18bd41166a84e3db06bfde05fc53994 : SUCCESS in 9m 47s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4b8f418780df4767a6c59ad7e970ce21 : SUCCESS in 12m 09s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2142bba6bc4a486b92881d6c86dbb2e1 : SUCCESS in 4m 17s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/64b25bd7227844b7b9ce21cec7ae3a87 : SUCCESS in 16m 09s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0adcba36b2734e3db90a9541142aab9b : SUCCESS in 5m 11s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/fbcedf397601442cb12d2cbd0f3425bc : SUCCESS in 22m 28s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/83a5347ce5834ddab0c475c4370585f6 : SUCCESS in 11m 49s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/53bf627f145145f1a9662f59a2e1cab3 : SUCCESS in 5m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9877e0c0fa844ed3acb4ecffa6ba0b23 : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2952fc90f7b94f29a9daafe81c48662d : SUCCESS in 21m 26s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/6ddaada939a94b6bb9b0b8edeb90c0c5 : SUCCESS in 6m 31s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4c4a5e3b59e9481b9928da390baef6a8 : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cb3661d18288433ba6f70e005a2b59cd : SUCCESS in 7m 15s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4c2300ed93d04fec893adccb8970b8b7 : SUCCESS in 12m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e52dbb11409e4b54a08f2fc6f1d6b0d8 : SUCCESS in 44m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2d4620ef432c4e05b6e4de6b250a5d51 : SUCCESS in 1h 33m 40s","accounts_in_message":[],"_revision_number":2},{"id":"98f5324e2bfad8a206e358187aaa35e1d593803b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 06:52:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"394b3a149588a5e8b08dcedaf7d5d554417588aa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 06:56:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e444114d1b3c46a787f0d38702007216 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"32593b8cfff53d8f70013fa5e73b6d565d6ddb6b","revisions":{"141c36dc093640d9c51a043c0dbbfd933e8b66ac":{"kind":"REWORK","_number":1,"created":"2022-02-15 02:47:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/829152/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/829152/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/829152/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/829152/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/829152/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/829152/1"}}},"commit":{"parents":[{"commit":"eb3c3c2b34d655a64736bbf646a1d4c525e7e7d4","subject":"update constraint for python-neutronclient to new release 7.8.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb3c3c2b34d655a64736bbf646a1d4c525e7e7d4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-15 02:47:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-15 02:47:31.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Icbaf6dedd5884458536492acdf3b5931908866dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/141c36dc093640d9c51a043c0dbbfd933e8b66ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/141c36dc093640d9c51a043c0dbbfd933e8b66ac"}]},"branch":"refs/heads/master"},"32593b8cfff53d8f70013fa5e73b6d565d6ddb6b":{"kind":"REWORK","_number":2,"created":"2022-02-15 03:08:37.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/52/829152/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/829152/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/829152/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/829152/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/829152/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/829152/2"}}},"commit":{"parents":[{"commit":"eb3c3c2b34d655a64736bbf646a1d4c525e7e7d4","subject":"update constraint for python-neutronclient to new release 7.8.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb3c3c2b34d655a64736bbf646a1d4c525e7e7d4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-15 02:47:31.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-02-15 03:07:56.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Icbaf6dedd5884458536492acdf3b5931908866dc\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32593b8cfff53d8f70013fa5e73b6d565d6ddb6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32593b8cfff53d8f70013fa5e73b6d565d6ddb6b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
