)]}'
{"id":"openstack%2Frequirements~829552","triplet_id":"openstack%2Frequirements~master~Ic7c0eab3dc335de951eef91f491df0546f129851","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic7c0eab3dc335de951eef91f491df0546f129851","subject":"update constraint for glance_store to new release 3.0.0","status":"MERGED","created":"2022-02-16 14:52:58.000000000","updated":"2022-02-16 18:48:02.000000000","submitted":"2022-02-16 18:43:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829552-new-release","meta_rev_id":"94159507650877a2d82e8c407edb2a8e51294d82","_number":829552,"virtual_id_number":829552,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-16 18:43:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-16 16:26:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-16 14:53:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-16 16:26:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-16 14:53:13.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-16 15:13:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-16 16:08:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-16 16:26:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"7a6d4c1908be70282b0a9c0448ec57fcdcf51794","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-16 14:52:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0bb62e921ed0179bf2a6e58745dc7efde60be34e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-16 14:53:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b4c6a18c15e6f7182fd7bf6a4072afdb4b5660a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 15:13:24.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/567883a5cd6440ffa5a0434571398c80 : FAILURE in 18m 12s","accounts_in_message":[],"_revision_number":1},{"id":"2a03227a3302fda99b3c08e77f899a5bc3d5f285","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 16:08:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b19a267234dc447fbbe6015dba7d35ec : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c9f563da84946619dff85a2a32b9548 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36a435f97d1447598bd8bfe7e5a123c4 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c43d220d828048949baed82ce044d437 : SUCCESS in 4m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9c3509c3a3984e2e9944245af8e8650e : SUCCESS in 5m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ab72ce0024a14f239ff5a653b429e2dd : SUCCESS in 4m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ea8355a123e9430395a3ba7e0041fcc4 : SUCCESS in 8m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/879c254a19ff49a2a3f1492d677f7a5a : SUCCESS in 9m 40s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/87d6c529b0f748f4af1b6d2ab82d01bf : SUCCESS in 5m 20s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e603bad51ca643dbab089eabf8fe0fc1 : SUCCESS in 11m 26s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/33af52c47d4e4af68429521b514915f8 : SUCCESS in 4m 58s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/239d857296e14d639338f63f5cae6389 : SUCCESS in 7m 10s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ee12a790ba0042e684053d205de0387a : SUCCESS in 9m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6e3040730fe94f35864303f50b2af231 : SUCCESS in 8m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1655c13411b8451fa579e3cb4c7855ae : SUCCESS in 9m 06s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/aec3bc7295f4419f8b82cbe7ba446d05 : SUCCESS in 15m 53s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/570c37d828554bacbc99a9adeee04115 : SUCCESS in 5m 40s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6a4eaac79a3b42bb90ad9c98155ac8d8 : SUCCESS in 16m 07s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0cd63a2228534bea83e9f6d3a41791b7 : SUCCESS in 5m 26s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/791ecd31e7b14bddb65ff6491fc0bc4f : SUCCESS in 23m 09s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0dffc639eb8b4bd2ab5e66b6954ff5f2 : SUCCESS in 14m 44s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4598ad3bba054f47add13fced1fa9d6a : SUCCESS in 6m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6163ffa6190944509ff7769f5a3a1022 : SUCCESS in 5m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aa477d7735e948f5affde292b45c24ae : SUCCESS in 19m 36s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b6807db192454c8abe4f9d290187b99e : SUCCESS in 6m 35s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/ba51c597d5dd45d19ec2cdd68c441dde : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e293952f9f8a4828ae0ffadf57e8b7c5 : SUCCESS in 6m 46s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d6a8c2a868f94347ba372e8044a90464 : SUCCESS in 15m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8b768a56cc646fa93a84ed45e90db85 : SUCCESS in 44m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c90e7c7688d4fdfb2b58aed939856fc : SUCCESS in 1h 12m 52s","accounts_in_message":[],"_revision_number":1},{"id":"fa8a11acf2df6eb0652ad6a0f1138ed92eb27653","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-16 16:26:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fa1c389760d04079874ca667514740aec1b4cdc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 16:28:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"22837776f6266fb8881c003adad62979b5aa4958","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 18:43:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4542232404d44037856298e886a4d568 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ac920c6b5484e57a43ed2f827a444d4 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6cf62acf79764acdb060662d92fcc7e4 : SUCCESS in 6m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29b943e284be4ca9854ce5a4d26da39f : SUCCESS in 5m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5b3255aded7640a8bc97fda7f4cb6cf4 : SUCCESS in 6m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e2dfaaa29ca745b8a142c69f3a652dca : SUCCESS in 5m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b4342de0c2674d45a04c269884e85559 : SUCCESS in 12m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3f58192e485344f1bfe0dd6f2b4b9527 : SUCCESS in 10m 18s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0b69ad30f17a4a4b921e1557e5136166 : SUCCESS in 11m 59s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/dfcc98d873d44a579031276af761b177 : SUCCESS in 6m 19s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/6b2931c1883e40338e386d3d7e3706c5 : SUCCESS in 7m 20s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/fc4b2aa6c2024b8086a58460242517e1 : SUCCESS in 11m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b50b3d5106e541379c4d7df2512cf6a0 : SUCCESS in 9m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/164fa28031234de5b7ac6f603d0e7d64 : SUCCESS in 10m 06s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d29dd323dd6f416191da2ddde269e262 : SUCCESS in 12m 55s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/9dbcfb2fca214026ad6dd37655e7f7b5 : SUCCESS in 6m 33s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1e21badc3f624ae584340f2f1f7b75d0 : SUCCESS in 18m 23s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e6c5197f567f46608b61b7844d8f0285 : SUCCESS in 7m 14s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f698026fd84342f2a65420fff1614a48 : SUCCESS in 20m 56s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/95f0b33635aa4812acd4e048236b0fa6 : SUCCESS in 14m 28s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c4279dc118994efc8e38bf11596c9c5e : SUCCESS in 6m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5c4b4d5b6b1e46ea968ca5fceaf09e14 : SUCCESS in 5m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8dffc52b19124344971371169804dae4 : SUCCESS in 19m 36s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b4dbb995754f4d988893fad69369f744 : SUCCESS in 7m 02s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/bd0f79ad14a34b0ab86d3fbf7b2d75c4 : SUCCESS in 6m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f909bcc3427147c894566b12b173a8a3 : SUCCESS in 8m 09s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1b5c4ec8745a44d784561ac28a17c5de : SUCCESS in 13m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59ab07a264fd451ba18c98b2b023a202 : SUCCESS in 44m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d24e2e41bb3427497f5968e70b9fdd0 : SUCCESS in 1h 25m 00s","accounts_in_message":[],"_revision_number":1},{"id":"56536ad7b087841a2bb9cbd5ce7c461bdfac81cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 18:43:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"94159507650877a2d82e8c407edb2a8e51294d82","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 18:48:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0386e763a3b64d9cbe20559dc334037f : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"732a20ec4e9d37f03c34d421cc0c77f95404d69f","revisions":{"732a20ec4e9d37f03c34d421cc0c77f95404d69f":{"kind":"REWORK","_number":1,"created":"2022-02-16 14:52:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/829552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/829552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/829552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/829552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/829552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/829552/1"}}},"commit":{"parents":[{"commit":"0a7e9306d40715c6c737726da71e8670d85d004f","subject":"update constraint for oslo.db to new release 11.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a7e9306d40715c6c737726da71e8670d85d004f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-16 14:52:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-16 14:52:57.000000000","tz":0},"subject":"update constraint for glance_store to new release 3.0.0","message":"update constraint for glance_store to new release 3.0.0\n\nmeta: version: 3.0.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Abhishek Kekane \u003cakekane@redhat.com\u003e\nmeta: release:Commit: Abhishek Kekane \u003cakekane@redhat.com\u003e\nmeta: release:Change-Id: Ib5e65c509bad4c594c53a5ed8038cc8c4b0b1a37\nmeta: release:Code-Review+1: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ic7c0eab3dc335de951eef91f491df0546f129851\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/732a20ec4e9d37f03c34d421cc0c77f95404d69f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/732a20ec4e9d37f03c34d421cc0c77f95404d69f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
