)]}'
{"id":"openstack%2Frequirements~829566","triplet_id":"openstack%2Frequirements~master~I1a992de0e4e96c2e4ac3ce30e649e9944aeaa087","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2022-02-17 14:50:53.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I1a992de0e4e96c2e4ac3ce30e649e9944aeaa087","subject":"update constraint for oslo.versionedobjects to new release 2.6.0","status":"MERGED","created":"2022-02-16 16:16:45.000000000","updated":"2022-02-17 18:19:14.000000000","submitted":"2022-02-17 18:15:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829566-new-release","meta_rev_id":"ded68848f2b0d051db206e9b3c169646ba5e496f","_number":829566,"virtual_id_number":829566,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-17 18:15:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-17 04:02:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-17 04:02:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-16 16:17:11.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-16 16:46:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-16 17:53:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-17 00:31:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-02-17 03:09:45.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ba71f29ad744112c2a492c2003d8ec700e390672","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-16 16:16:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a38006c288e4cae2712d6b4367ffa82f74a1eb48","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-16 16:17:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"567750f56e57fb55f1c55c5c7d90b6ddc1257d0d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 16:46:26.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/95446484bb34476bac0184ca386406f1 : FAILURE in 26m 21s","accounts_in_message":[],"_revision_number":1},{"id":"39e3b2a0866eb92f41e5c64cb11605f597badff6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 17:53:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a22ec895d3884523acf0ee290e053276 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96ae7f7b409943d0a8a920dfecd26e37 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a6cf581e80741429e500e5b29de6c90 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7ab5e6221374d4f972738f1504453ad : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac59fbeb1e394d009ca826aacd2df15d : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aaaeaac97e774ad494634ae9d108d182 : SUCCESS in 4m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e886acd2c4d14a7ebede20e0535d5587 : SUCCESS in 12m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5f1c38ce261d4d5086d26cc04f09cc5b : SUCCESS in 11m 57s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/60a18cd9e729470e958009a4a36d98d6 : SUCCESS in 6m 10s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5886b65325db45f8ac2c8e594e3a7356 : SUCCESS in 10m 10s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/d09233a9b3f940f0840da4e8e4ddd9e9 : SUCCESS in 5m 11s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/569c3ac462fa4c438969f45fa03be47c : SUCCESS in 6m 06s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/39b26a94158b40ce861a048fa9d24f93 : SUCCESS in 13m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/641216180826484e886bd2f1ac014f86 : SUCCESS in 9m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c31c1a0dc84a418781e4de9b131d8aad : SUCCESS in 10m 39s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/25ec312e1f3e462eabf27082a4fd2515 : SUCCESS in 16m 06s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e8aa9b5fe1434f58955bb3dce286ad91 : SUCCESS in 4m 19s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b8670a1072a1407795157b58345e0ef7 : SUCCESS in 16m 03s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/10dac473508f4be2a338c0b71f6e0025 : SUCCESS in 5m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b88365cf44144e789b21c0a5f9bed90e : SUCCESS in 25m 34s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/31212ef7284640e5a17e793766e502db : SUCCESS in 15m 08s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/93146ef8669d43adb7129068cbd7912c : SUCCESS in 5m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c129b7150e414b5cad049a8de7a87729 : SUCCESS in 4m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0028d642a84446df945aee68e4e2cbf9 : SUCCESS in 24m 22s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5cb70f7e55e74e8ba35c1b6468c39e37 : SUCCESS in 8m 15s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2cb7d9a2b663485780865bf8828e9e25 : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f0c025dbbd604f0bab3c0ec7681269f6 : SUCCESS in 7m 42s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/05b06f182d2142c2af566768b44751c4 : TIMED_OUT in 41m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4748c630ac7b43208ddfd8a6ffd22aba : SUCCESS in 48m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3db12d8c819446b84e6c408cda914da : SUCCESS in 1h 34m 04s","accounts_in_message":[],"_revision_number":1},{"id":"eeffbe6a7ff06fb4c71a1dfb6f62db5d7f1973b0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 00:31:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b10a3cf914621a4e1c09e13dc1f8d913899eb1e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 00:45:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/934c15ef97274726a8447830328e380e : FAILURE in 11m 50s","accounts_in_message":[],"_revision_number":1},{"id":"7038dd82d941c467be3c76053c071dc431d19fb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 01:19:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa82efc169694254aa448dde3b635181 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea5893242fc9489f8792a0d56cc45086 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8d91425f29b4abfb77de09f2ad373fd : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ba6bcdcaff14726822a31b37ad08f50 : SUCCESS in 4m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d33dab116dba443a93082733eaad59d2 : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/555ba55f423745aaab1e6c2971720df8 : SUCCESS in 4m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/aaa38af96fb94a4f8494559699bb9b55 : SUCCESS in 8m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3ade8641b96541abb3c6e6f068ce03c8 : SUCCESS in 9m 09s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ab6f73ae35e74aa68ede80b8aa748093 : SUCCESS in 7m 20s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f0bd35707ffd43099bbdd49f14c446bc : SUCCESS in 11m 24s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/581320933d8449ee9402e24fee34c787 : SUCCESS in 5m 44s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2043944439494634992484876d4224db : SUCCESS in 6m 21s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0fb481f6bfca4ff793dd609316115117 : SUCCESS in 10m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d4a7d06790594dbcb9a9ea1269a6fa6d : SUCCESS in 8m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2de5925f54094878aba6361c326fa7a9 : SUCCESS in 9m 29s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/43b5fb46acf54dabaeda06c92f39af18 : SUCCESS in 17m 32s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/691c0d7bf4c0464898034dce007c930d : SUCCESS in 4m 25s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/344401f4868c4442aec34491bdebd7b8 : SUCCESS in 15m 30s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f8de15f85d934ed68f26cf7988890c8e : SUCCESS in 6m 24s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4fc1e079434a432fb0859c0109c45133 : SUCCESS in 20m 44s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0642258e12bd4b0d9e00b0594cde08d1 : SUCCESS in 12m 27s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/7a111f4403724c6787231a924d7eb8eb : SUCCESS in 7m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6d9915c27d014d719c2d19ab3c51b1ad : SUCCESS in 4m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/36cb91c3884f4b0f9146f13c7b570ac1 : SUCCESS in 20m 14s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/bbcd8dbef3c44fce838e774c06f158f1 : SUCCESS in 6m 07s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/eb1a56cca7a24ebea1fb0a383df2d023 : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8d48c96c72f843fd807e480e42713eff : SUCCESS in 6m 35s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a44c0b17453f4916912d7ebf29d89b2c : SUCCESS in 10m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60a44eb48f804b32ba724da4bd2264c5 : SUCCESS in 45m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7f278e4b48fd4626bbfc46c9bd9b58d3 : FAILURE in 24m 25s","accounts_in_message":[],"_revision_number":1},{"id":"804492ac5e76c9f38b9cf00179354e3201e941b9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 01:52:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fb18e3d0308692a7b617a8899e819c2a9a9b9e2b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 02:07:28.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d464deeb43384709a4d14e7f085d7ff2 : FAILURE in 12m 23s","accounts_in_message":[],"_revision_number":1},{"id":"53b6005d6c6f8122ab08c28d12b506c3260227a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 03:07:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02ceb129265b4e59aa6b458ee01a579e : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fac52de9fc64614b1da69207b87cfc9 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1460691412314fc9bf8368c28645a45b : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5b0892d39fd4aa8a21bc7ae9c8ff8ab : SUCCESS in 4m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/280988b4cb684e79ba9f4c8e26b1ba5e : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7057cb603d6b4b378b013c63d7e7624c : SUCCESS in 4m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6d303fbf7ad047b3b58adf8337649d1e : SUCCESS in 7m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6c3dd0ee6b3a4db4b2c3f69aed39c27c : SUCCESS in 7m 38s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b663eb86c88245d194c08cf2a3107033 : SUCCESS in 7m 16s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a6431bfb42eb47f398d1448b7a56f570 : SUCCESS in 9m 29s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/09c45d85e6224ab8aa3266c6c7ad2d50 : SUCCESS in 4m 47s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9465b13f980f421292a7d4a818d7488f : SUCCESS in 5m 29s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5e1234cf8daf4152a2383ac11686af98 : SUCCESS in 9m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5526d6262bfe49eb8926798a98d4ca08 : SUCCESS in 7m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e30a109c50b64335a532bc7e1b5f96bb : SUCCESS in 8m 14s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e422c5736fdb415cbbf3739df4ce783a : SUCCESS in 11m 21s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/3d5f7ef436e54ffb9496da11b269c7e2 : SUCCESS in 4m 23s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4f595d0e7076495d93e83c321e72099c : SUCCESS in 16m 06s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7163128ab28c4541ab7ea80f185cce72 : SUCCESS in 5m 12s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a6403891d5ea4fa9a0709ccc68bacf0e : SUCCESS in 22m 28s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/90c45a989b094b5cbf53b42d6491ae42 : SUCCESS in 11m 33s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/1b8c723b79b44c87b81a941a695a2cc0 : SUCCESS in 5m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/54b776e0d9304c6c96f91ef37563794c : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dad18bcdf879446bb46bb9fb67cf380a : SUCCESS in 19m 26s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/fae098c15b06444e939d9b280c674e42 : SUCCESS in 6m 24s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/bb1449db806e4cad9e5c327fb2136e2d : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e3bf72f510594090baa91555b52ff113 : SUCCESS in 7m 40s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/dc124230bd1b449b8e110cccaec8433c : SUCCESS in 9m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1608001eda8643ecbb9244cebe6ed7fa : SUCCESS in 49m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/17d140897b444b81a948528e5cd78eff : SUCCESS in 1h 12m 48s","accounts_in_message":[],"_revision_number":1},{"id":"67706c6b2dc3beb48d939968013d69929f143905","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 03:09:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e1c86745ab56c720a7ea7c0b3869cf81f056ef7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 04:00:19.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"be0a170187804c19481f5898daf2cb345b270a66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 04:01:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4dfa0b59f61c079ff824c130de66e6807d068002","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 04:02:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"212fab27dbce581bfca34048b05f322215125d26","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 04:50:39.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/56390797c74142d9935329d38e4cc75f : FAILURE in 14m 08s","accounts_in_message":[],"_revision_number":2},{"id":"6ca85515b88e6848459a4edae1de0061c8d4cbb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 05:16:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a656621d8ab046429a4eaeac64226fbf : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e2aa371720e472fbb7695cc8ced3798 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b2217aec58a44debc338210056096eb : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34f782c0d6c74aa1884956c6472c5d34 : SUCCESS in 4m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a11b79c87a6c41b182bc4ac272a38740 : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/af43fd202b084687904ef22af5169eec : SUCCESS in 3m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d7fc248797174c9ea3433fd0a14d9ca0 : SUCCESS in 8m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6d89c9a5091f486c8c2c346a6eb53c36 : SUCCESS in 8m 06s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f03bd2798ed44531b982d70377d537ff : SUCCESS in 5m 51s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/13b83d3d0aea4356832929467da52251 : SUCCESS in 10m 16s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/ad458d2fed5b41c3afd9c994f5d4503e : SUCCESS in 5m 51s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/931caac73f17478c89501aa2d0c26569 : SUCCESS in 6m 00s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/16013a5eef004bc2a25ec823a65d00bd : SUCCESS in 9m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4ea74087e37c47bd8b503a156c0118ad : SUCCESS in 9m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ab7c5d51ca6c4b878c34c6ab93a0a14d : SUCCESS in 11m 13s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/38e57fe596fd4ecbbe5ab84ce59f2b1f : SUCCESS in 13m 38s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/cbba76784577437897eae59f739ac3f7 : SUCCESS in 4m 18s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/15ceaee2f88d46b8ac5c30c157357a0e : SUCCESS in 16m 22s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/917b49f03eb84e048104fbbf7f839e2e : SUCCESS in 5m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e0231ae99a9c40b586191d71ff5323da : SUCCESS in 25m 33s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/17d98a2b040e4f759426fe86868f63c1 : SUCCESS in 11m 57s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/15f06d33e4fd4cbe9e9b107142da59b2 : SUCCESS in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2fd9424aaa1a4e96a8dccb1ec5f2f8c0 : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1c8a8bdc6b8c4b0caab5e512a4296180 : SUCCESS in 28m 03s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/69766929844546cc91fdd0b48b424ac8 : SUCCESS in 5m 46s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c8500a1dfb484e4b8d395c14adbd1906 : SUCCESS in 5m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/63dc5eb93f964b79a072287acc68158d : SUCCESS in 7m 21s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f3573088e0c648efad81c7dd7201bbb6 : TIMED_OUT in 41m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ffeffe0758b4400285a1e5b7de88dc81 : SUCCESS in 1h 03m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0248c912e84341c0a439c032473a9ac6 : SUCCESS in 1h 11m 35s","accounts_in_message":[],"_revision_number":2},{"id":"1499a2cc65f8b2ba2abea361b545e086deb9b857","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 05:19:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8e29624c7837931288872ef209263991167598ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 06:05:10.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9e776e0b17b54612b13da9046cac2374 : FAILURE in 15m 53s","accounts_in_message":[],"_revision_number":2},{"id":"74e7797c094770204bb429accd5d8e122aaf94ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 06:59:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6abed322d7d64661931ba2b31a305c1d : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0f0232a69f74b6ebd5bbad8a4dd90a5 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ff7118e7a234b1797d47fbde5ce0607 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53b787ebf96e4847bbda64e9bb976ee9 : SUCCESS in 4m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bf337c4326b14aeab1b2930f643b5cee : SUCCESS in 5m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/390aeda6d2a2425baf83d03ba7e2e5d8 : SUCCESS in 4m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bb08bac56bda4111bc8b80fbb15b6370 : SUCCESS in 8m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2603463740ca451497578d5809abaa5a : SUCCESS in 10m 33s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/0e02349b55d64f9fbfea99d3db730afc : SUCCESS in 5m 53s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0dd9fff8c8804c20aba0555043e602a8 : SUCCESS in 9m 35s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/819daf3607ef4050bad9039c27dba4d4 : SUCCESS in 4m 59s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/006b1eb29dbe4eb4914cbbc6c83fec92 : SUCCESS in 6m 24s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a634ff4975fe445f985438813ff0c97c : SUCCESS in 9m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7e07765a71a54f38b3807d1ec9a12236 : SUCCESS in 8m 23s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ccffb6aa714c42d08a260455714d14ea : SUCCESS in 8m 42s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/474a6fcd1ed94d2181b0fec1d2f6275a : SUCCESS in 17m 50s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/4e80a3a0ee6045998219e8c164fbd6ff : SUCCESS in 4m 13s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e590c02d8ae64a17b96191b8fbe47f66 : SUCCESS in 19m 31s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c080b0acf1b74859ba487c8b6b8be5b4 : SUCCESS in 5m 37s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/031990dc94f048b69956641bc85bac75 : SUCCESS in 22m 04s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0cc4e56010d74b0ba74b93bbd3484a86 : SUCCESS in 14m 11s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/faf609e829b14752b79f7abd62ba6bdc : SUCCESS in 6m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8cb2f595b7854ccab7b517a43a93605e : SUCCESS in 3m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0402da961e8847fd91b150e338a7ad83 : SUCCESS in 19m 15s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d801f0328f2b4a199b3db65309964e01 : SUCCESS in 6m 45s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9e4ccfcafa144279990d678dc8308277 : SUCCESS in 5m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/29c445af56a7457fbd7bfda28c1ee094 : SUCCESS in 7m 33s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2e18e71c12424942998016ae125e10bc : SUCCESS in 15m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1048765af62449cba43d914fd0dee33a : SUCCESS in 45m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc36539899ef4625b50b90da54cf88e1 : SUCCESS in 1h 15m 29s","accounts_in_message":[],"_revision_number":2},{"id":"273fddb32c0bd38eae000c7fdbf6c70d2c0d0c9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 06:59:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e63ca7f8c0d8efbddaa5279d0ae5a38ac00ab439","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 07:58:00.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c43ba1f8a74c40c099dc80d4349f0c9f : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25e7dfb48f174d408681703f55ea6b67 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2596c60115b741609026a102a7b3bb15 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55593fdbe1e34d7595c1d524be49020b : SUCCESS in 4m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c452877136fc471382675fd67017602f : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b43532a60a364c02bbee010f679f1112 : SUCCESS in 5m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bacbffa951204533abd381434a6f2f4e : SUCCESS in 8m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b5d867a6d2474c439e8610e8c0382e42 : SUCCESS in 8m 45s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ad47d96827824eb3856eec4291317aea : SUCCESS in 8m 49s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/064c9db94d964acab15dad27113b784f : SUCCESS in 4m 58s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8acbdf78a0e94f3aa673dd5d5bb9fab9 : SUCCESS in 5m 14s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/26da755bd2794abc95b705b7ac8a9e49 : SUCCESS in 9m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2c653a23583e48449bda278871cc4b11 : SUCCESS in 8m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2cd88ee74d394afa81d89c32f8767e7a : SUCCESS in 9m 08s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d81f329490024f7582f25b5c14aaf6bb : SUCCESS in 12m 56s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/79f65d408ad942e4a5945f1fd05c91c0 : SUCCESS in 3m 59s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e8434a1a05bb4945a709f9d77206589a : SUCCESS in 15m 56s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/4f0d1eddee8f4a08a7938a54357e0f12 : SUCCESS in 5m 04s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f6e74910276a46e48f893565ad09257e : SUCCESS in 32m 54s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/54aa332b35a7447fb06a5e87227afc57 : SUCCESS in 12m 11s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0a4d8c795cc84141b966a898ed49f4b6 : SUCCESS in 5m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/167c205baa584f47bbcf143ce1a78c26 : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b7a54193f7b9492891a21bc4ca1adfac : SUCCESS in 22m 30s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/40e33456b9fa4a00b6ae012771c98184 : SUCCESS in 6m 26s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/24184051bb8845b789b45b2729ccf343 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1c079ddeaadf42739f73ead5efd50958 : SUCCESS in 6m 53s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/982f863dab40476a99cb485c40276b6b : SUCCESS in 14m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/20ff267c8cec4f79bdd50025ec66f6a2 : SUCCESS in 50m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0695e4be65534fa8b26e0f4903016629 : FAILURE in 57m 05s","accounts_in_message":[],"_revision_number":2},{"id":"cb43967a3ed8fec9c286fd9f59fd86e369a87d31","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 09:00:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"23c050cd2f60193946e1bc4fa8899d8ad3e55715","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 10:09:00.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5a156e6fa8d4420da950173d2e15cd72 : FAILURE in 22m 25s","accounts_in_message":[],"_revision_number":2},{"id":"3fefc79d0136f13eeab7de05c5ffbfae89ffad77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 10:29:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b77fee0241c4ede8cebdc2d5915f7ea : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98047b95024f4118a840d8bf255f3f7f : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58819acd6e7c4ad09dce824504ed965d : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e5703af13b64bc79a005e0382043d13 : SUCCESS in 4m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e17801b69f804571967206720b960206 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/06131e6dcfb44b5a8fc175a80800d063 : SUCCESS in 4m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cbc275ee54fd4be7885c8530bb7c433e : SUCCESS in 11m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b98f86ef41f34963a83265fcfed31575 : SUCCESS in 10m 08s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/55bb10628d43452aace806e9cab30731 : SUCCESS in 6m 40s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/779099511c2e4521be14519a2b2d3f60 : SUCCESS in 10m 26s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/53a721e01fa0402fa6f34eb84001b89b : SUCCESS in 6m 40s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/78b7e92dd6d4451fbdca4db1a5c8a994 : SUCCESS in 6m 57s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/22ed07ea37354314aee3fccf311249ed : SUCCESS in 12m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/167e3b25a18446ac8d7d392d744ee909 : SUCCESS in 10m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8eec3458c7e04e3581112cbdb659f4c6 : SUCCESS in 9m 15s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/13750ae77c004eada976b21bb5b15e15 : SUCCESS in 11m 50s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/0bf021c9c1c74045a0a8f506e691fcec : SUCCESS in 5m 12s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d5dc91e37c9d444e8d0e46a67d01385d : SUCCESS in 16m 21s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e819eab86413440fbc930bbd7610a894 : SUCCESS in 5m 23s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/059587cef6794096bdd742498f90d710 : SUCCESS in 26m 30s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e6075fb3484845b8a15905e7088453d5 : SUCCESS in 12m 30s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a00050ea279c4f98876c581dddd4e2c8 : SUCCESS in 6m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ff447806cdb04fd99b31d4a21c26cf26 : SUCCESS in 4m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1962bad9a06143acb4eee8d30f384cab : SUCCESS in 24m 37s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4fcda882655341da8dc0d43f5b022212 : SUCCESS in 6m 45s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d09feee9dae5471d8255ab5065094443 : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d8672549b87f4bf99f7996ac725e4eb0 : SUCCESS in 8m 02s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/13a37fd0e0694aac812b1bd642a25f13 : SUCCESS in 11m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37eae4e0014345dc90d85998af96a79d : SUCCESS in 49m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/23d8d9d2d3314c21b6f7ead53507b6d3 : SUCCESS in 1h 23m 01s","accounts_in_message":[],"_revision_number":2},{"id":"427de2f16f778b6fbadd7015d671440689b27b69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 10:30:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"251e947e634b11dfd94a5839dd0ed7776c586959","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 11:57:30.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f25de9fa505648899068dc02c418c9a0 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbf88685ed784761b85033a2a3da818c : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/633417d473c545eca843b8eb910927e2 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf7d8ffcd9ee46e2afec5aefdb92d990 : SUCCESS in 4m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c1d3cb55606c45598a76bff1f63eb784 : SUCCESS in 4m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7e56d594dd324331840c639875bf13c0 : SUCCESS in 4m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/70a0a1fab5bb49efa12159e3754aa7af : SUCCESS in 10m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8f87cbabd21141d4a0cd6e969a3e7595 : SUCCESS in 9m 29s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8d81c2fb71754ec787e2ba1b1cef8138 : SUCCESS in 10m 40s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/0add8bd4dfab455f82e274a85f45149c : SUCCESS in 6m 06s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a897816bb1ef499f886f6fe6b7a1ab1a : SUCCESS in 5m 37s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9d1b163ccb0e4764917be58d4a32555d : SUCCESS in 10m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9f1184c19a404b70944e92dae900fa82 : SUCCESS in 9m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/29e6e7a7bfef4e79bb1a9aa4972d2255 : SUCCESS in 9m 17s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/66598e8140b146e5ac094b12ceb24b72 : SUCCESS in 13m 49s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2bc434d1931e4382baaca7a2993917ca : SUCCESS in 4m 53s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8520d5961a9f46eba8b0c5268cbccc3c : SUCCESS in 16m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/01ac992b2e184684879296a59c6bbb01 : SUCCESS in 5m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1a5a40714e13492d9eefbe2433583ece : SUCCESS in 24m 16s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2bbc6714f994468abe9066414d35f5ed : SUCCESS in 12m 38s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f7d6d955a2f84724b514aa7918b485dd : SUCCESS in 6m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b5fe59acba834b3f8bdf1568af50bb67 : SUCCESS in 4m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/733d023517f34a87a51f2dbc1753ce42 : SUCCESS in 19m 59s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e61aa49f7b004f7cb430f8f8c9574275 : SUCCESS in 7m 24s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/fdf6dfc4a08b4e4baffefd70d953ed4f : SUCCESS in 5m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4230b74cbe844c7cbf390d2ef4832bd1 : SUCCESS in 7m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2d897c62a69e443bab360f16387348ab : TIMED_OUT in 41m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/57bd2af05eea470b8f16fc5898c457cc : SUCCESS in 44m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/189930a15c274d019b6752452849c0ff : SUCCESS in 1h 25m 55s","accounts_in_message":[],"_revision_number":2},{"id":"40da9c7f32d1370f7cdcca5377aae38094ffdca2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 14:50:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b2f763260615c68bdfc748b52c8cbdcd174d3721","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 15:56:21.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/74546c0f094a4bf9a8b110d5c12a8e48 : FAILURE in 15m 02s","accounts_in_message":[],"_revision_number":2},{"id":"d900dddacd51f23b3c5337b62bbc2fe6ba2265bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 16:29:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7cb0a06789847d8ba62b18b737d765f : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d792691f6574e88ab8d52e4e7e7588b : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39d67eda31394df48cb43b195773fd85 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d459ff0b0424eff84400c446bf158ca : SUCCESS in 4m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/03fd793329464e858bfce084a641a192 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d3aaf8947209462db81f64a9a3a05698 : SUCCESS in 4m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f4c9b8b542f343c589244f6716373f36 : SUCCESS in 10m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e4e7be4accaf40cfbed0a77674465512 : SUCCESS in 10m 01s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a8628077d8044ac4abec6119f2f8f256 : SUCCESS in 6m 42s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9913137e9f014f9aa2682946a8841676 : SUCCESS in 11m 17s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/6e1608f4b7104a57aaff6df05a4a573d : SUCCESS in 5m 19s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/bc0992269c074a35a8e2ebf4911ce653 : SUCCESS in 6m 32s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9f8b05fd7cb649b7a9003afbac0a302d : SUCCESS in 12m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c243f40e516c4bed952bbb65132bde73 : SUCCESS in 8m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a0bd632c8b3c4755b786f740ddc9e560 : SUCCESS in 8m 46s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d72cda9d34f94de288adc570b6aa31b6 : SUCCESS in 10m 58s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a9cedc436b854f25af68355c0a6190b8 : SUCCESS in 4m 14s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/da3fadb0ac174445a79249775b373e57 : SUCCESS in 16m 57s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/715dd78bf4974ad08bd23d616f1e329b : SUCCESS in 6m 32s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b28ce041a3b74151bcdbb6dab14be92d : SUCCESS in 24m 58s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a9e76cc1a840437c85a9f6373a44462f : SUCCESS in 11m 56s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0e43b2902f1b486ba65846cbe6e24f26 : SUCCESS in 7m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cd6d4370394448bc8b2abd1a2cd18da5 : SUCCESS in 4m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5deeae9fead543539c014a6a39e0e1ad : SUCCESS in 21m 09s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/87883a6678834698a3fe0c31478aa2b7 : SUCCESS in 6m 31s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9ee03fb3a4ea43228c6917c8d7877ada : SUCCESS in 5m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6eb3353cffa44243b0362bf554170da5 : SUCCESS in 9m 24s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6fd877282507473599717ea09b87b82e : SUCCESS in 13m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f28af6ca8584fa1aec8708aee034b8c : SUCCESS in 47m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ee19b7924d564951878c66ead81a1182 : SUCCESS in 1h 34m 13s","accounts_in_message":[],"_revision_number":2},{"id":"1d98f373f07b2597431e6b5548e75cfbc492feaf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 16:31:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3f6d9a6e98c116e6b9d306ce4e7c62042676ef26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 18:15:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfe0a79333ee4666943f4c9852690c9f : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d7a67e4500c49eca07349cd936bef6d : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/990f1e54f53c4a4e9a8e0d339a461c32 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84db619d48024c01a8519463b923dacc : SUCCESS in 5m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4bf55639c2714037baa349ce6eacb0d0 : SUCCESS in 5m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/448a1d5a112c457c815c46d2fd3aa5d2 : SUCCESS in 5m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7b7edd3fe89d4c28b4ef4b9ced2e6981 : SUCCESS in 10m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/38e7784cdf794b66b18a55fd812306aa : SUCCESS in 12m 20s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/43f60f851d96425e857b8998775c3917 : SUCCESS in 11m 32s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/5c489d31b7514209929772e45449f0dc : SUCCESS in 6m 36s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/efe4e79482884f748dd8977a47233ae3 : SUCCESS in 7m 23s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/21bb67fad668487e9283a857da3936e4 : SUCCESS in 12m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f61545fe853f455db349affdb4e75cfe : SUCCESS in 9m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3b5effa4365f4bf3b4c55023adc81b8d : SUCCESS in 13m 11s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/18ae75ec57924a81bbe1e911d61a1957 : SUCCESS in 13m 20s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/54de3f79fae4486895ab282fc5a8506d : SUCCESS in 6m 39s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7b08540ab79c4408afc42f5d51ed7f74 : SUCCESS in 18m 48s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2d7aa23b07c34e7cb139da1a22f05a6c : SUCCESS in 6m 33s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2e09160dc6c9410bb73e1f96d7a6b024 : SUCCESS in 27m 37s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/dbd32887ff394202bd810f95dbea71f2 : SUCCESS in 14m 54s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/6070eb8cf756499c9825a57a84892b39 : SUCCESS in 7m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/614f5083597344f28baa6a064b89b901 : SUCCESS in 5m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d61a91fe2e454775bc4c8e8aba418d3f : SUCCESS in 30m 08s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4fed4647b2a645a5a3746fe305f2d3ff : SUCCESS in 7m 14s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e560665f42b140439016c5b0c50936a0 : SUCCESS in 7m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/584587f7bffb4d50ae934abe3031787c : SUCCESS in 8m 39s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f095fdd3aadd424ab4dc4d04690d6084 : SUCCESS in 13m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1390beda5844af999c3276ae9bc267a : SUCCESS in 51m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7db19f52de2b410cabfec16d818ddcbf : SUCCESS in 1h 36m 42s","accounts_in_message":[],"_revision_number":2},{"id":"51e767b835ac95546c585789b868aa39c4344569","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 18:15:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ded68848f2b0d051db206e9b3c169646ba5e496f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 18:19:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d0dac05ad1e4d688b258ca902e732df : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6608c1c5432524ac82dca10d98eb79361ff9944e","revisions":{"145b080e86f70a9e8b7ae9a2f9a467ed141c851d":{"kind":"REWORK","_number":1,"created":"2022-02-16 16:16:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/66/829566/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/66/829566/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/829566/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/829566/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/829566/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/66/829566/1"}}},"commit":{"parents":[{"commit":"0a7e9306d40715c6c737726da71e8670d85d004f","subject":"update constraint for oslo.db to new release 11.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a7e9306d40715c6c737726da71e8670d85d004f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-16 16:16:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-16 16:16:44.000000000","tz":0},"subject":"update constraint for oslo.versionedobjects to new release 2.6.0","message":"update constraint for oslo.versionedobjects to new release 2.6.0\n\nmeta: version: 2.6.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: Ic32d9fb41107843c3c0f81c96dc86fba24db02eb\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I1a992de0e4e96c2e4ac3ce30e649e9944aeaa087\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/145b080e86f70a9e8b7ae9a2f9a467ed141c851d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/145b080e86f70a9e8b7ae9a2f9a467ed141c851d"}]},"branch":"refs/heads/master"},"6608c1c5432524ac82dca10d98eb79361ff9944e":{"kind":"REWORK","_number":2,"created":"2022-02-17 04:01:50.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/66/829566/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/66/829566/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/829566/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/829566/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/829566/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/66/829566/2"}}},"commit":{"parents":[{"commit":"0e24c72df39927e1dd84d33b3e27434c15a90f81","subject":"Merge \"update constraint for oslo.middleware to new release 4.5.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e24c72df39927e1dd84d33b3e27434c15a90f81"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-16 16:16:44.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-02-17 04:01:40.000000000","tz":-360},"subject":"update constraint for oslo.versionedobjects to new release 2.6.0","message":"update constraint for oslo.versionedobjects to new release 2.6.0\n\nmeta: version: 2.6.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: Ic32d9fb41107843c3c0f81c96dc86fba24db02eb\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I1a992de0e4e96c2e4ac3ce30e649e9944aeaa087\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6608c1c5432524ac82dca10d98eb79361ff9944e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6608c1c5432524ac82dca10d98eb79361ff9944e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
