)]}'
{"id":"openstack%2Frequirements~829567","triplet_id":"openstack%2Frequirements~master~I36b79a556e7df8711c8f443d64aa78e7478819fb","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I36b79a556e7df8711c8f443d64aa78e7478819fb","subject":"update constraint for oslo.privsep to new release 2.7.0","status":"MERGED","created":"2022-02-16 16:18:15.000000000","updated":"2022-02-16 23:09:03.000000000","submitted":"2022-02-16 23:07:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829567-new-release","meta_rev_id":"3019c90c51e30551f48a56b0d8a4edbb53c9a2a3","_number":829567,"virtual_id_number":829567,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-16 23:07:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-16 18:39:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-16 18:49:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-16 18:39:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-16 16:41:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-16 17:30:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-16 18:39:21.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-02-16 18:49:45.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"a8998d25acafd026ee22dd18796a4c706ce9879e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-16 16:18:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7cca23324964ffcbf9853c242f29262778fc17cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 16:41:26.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/478b76d0330c4c479e38a0afcf805e46 : FAILURE in 17m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a36e0925b2f8ce93bdaeae1b3ac73950cb69a362","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 17:30:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62179f03171b4f9784bdef777561cf7a : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/903088e98bdf4c9eba5cb3203a12ff56 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bcf1194d41849db8e4f16346dfac537 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/220be410681544c3a07c37ffe25660e2 : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b647269564ce4bff9e0390a9f0bd5a84 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/92cdf187b5b74b56a49ef8ffe60694b9 : SUCCESS in 3m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5777eb7bb12e45e3ac23e84bf33d79f1 : SUCCESS in 10m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7dad196ab8914063b70edbb9ee6be131 : SUCCESS in 11m 13s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/49ab4e651de44fda823b69a0b384c4d2 : SUCCESS in 6m 25s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/151b789aefd8468fab441f8c7f55b92e : SUCCESS in 12m 13s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/2d901e287480496cab817c4d9aac763d : SUCCESS in 7m 59s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/52921e9bd01a4957a26b475bd81ede1d : SUCCESS in 6m 57s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/18a590949c3047ae8950388dd992b85a : SUCCESS in 11m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3eb62fe3f7224dcf97233d166040a135 : SUCCESS in 9m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9e720dbbf1a24dd1824ec6ae1ca70d73 : SUCCESS in 10m 24s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f3ff2d368f5b4e848fb754884dbbf6e1 : SUCCESS in 13m 01s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2c2139dc98e64712a0107d25979ca5cd : SUCCESS in 4m 57s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/064531b286bb4b22a2445859b540b1c1 : SUCCESS in 16m 02s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e345550a24294c9486d1da6a56f59a88 : SUCCESS in 6m 19s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8848a79828a845bf9f9b51606e289f48 : SUCCESS in 23m 22s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1bcb597fde524cb9b0a0c5dda672a2cd : SUCCESS in 13m 32s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/77bb23d5397d4882b0a0083beb705a61 : SUCCESS in 6m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/43b3e17566874d7f88a8cece7e94312e : SUCCESS in 4m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b8ee76dd4ada4d43a7946f5fbcfded64 : SUCCESS in 18m 27s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/8eb8428a95884595ab0f6e284eeb5494 : SUCCESS in 8m 26s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/acbd3a7c1098442384da358a0066e0a3 : SUCCESS in 6m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/25306b6412774825a0edaa82c6030ce3 : SUCCESS in 9m 53s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c33438bb0c49476e86dfdbd0d9ec45a6 : SUCCESS in 11m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/89d5538cf3594cd6b9e87749019a9647 : SUCCESS in 46m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b9646da8bf845ccb4d02c889040e6f4 : SUCCESS in 1h 06m 36s","accounts_in_message":[],"_revision_number":1},{"id":"4ae16975f6a5a37b18b6f54e8761963785938642","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-16 18:39:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"45764f935f9886a9130c12cb3c97e8cd8331cfa9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 18:42:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5d1136c0e84b3ee2904d4f4dccec26c5d01fff37","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-16 18:49:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3cbc30e1f1fa75c770e2a027ec3981da5832f692","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 23:07:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffeaf2f81eb54b028bb6d5337d1227cf : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea8cc341c3fe446484ea6c64edbb1874 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/256dee38fda047d3ac47253800a22c2a : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66bf37d2be5c438fbbde9d110ef0e8b2 : SUCCESS in 4m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b8e8f56df45f463a8a300ef8e6bb95f6 : SUCCESS in 4m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/12bf1f631af540f49d9ac193a2d8138f : SUCCESS in 4m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b7be5b7c958a41f49c8974d86dd6fb25 : SUCCESS in 8m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/00cf248e91e7454ea33af301a3d27bcc : SUCCESS in 8m 41s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7b0e153d3af34c83b69bb46d7f3b51e6 : SUCCESS in 9m 37s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/5707a89f7c8b49cfa0e4b79843eae0d2 : SUCCESS in 4m 53s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9dc9fc012b3340cd89a7b3afec37dca7 : SUCCESS in 5m 54s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/54326e2283dc4a1e95c80782a19946a2 : SUCCESS in 9m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5545c5eb438a4ede9d2b0f45560bbf97 : SUCCESS in 7m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4684fb60fa7e4a419aeef875c4b63ac2 : SUCCESS in 7m 59s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/aaee8cd4fa1e4e3fb2a84f78dda656a9 : SUCCESS in 14m 23s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a5f52da5841645c581aa70ef189efe65 : SUCCESS in 4m 19s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/fe52eb0ecd0541af91b130c342056316 : SUCCESS in 23m 59s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/706c66e852824256862216fe80f94d44 : SUCCESS in 5m 01s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8f398754ebd54a7da97d58bff64c79d7 : SUCCESS in 24m 13s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c5f03bce154e4f8985626c3330e99e82 : SUCCESS in 11m 58s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/03b8abf513b84bc3961397eb128528e0 : SUCCESS in 5m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/846073ac620644b08c0a58c58c07aca8 : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bf0707a946c54db39b872897353ce5f5 : SUCCESS in 17m 47s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c65c53f5a8e24efd81df6078c894e00d : SUCCESS in 5m 05s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/318a9f67c0de44f2931be5382dc6607b : SUCCESS in 4m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/98e1713ec6254baf917de053982729e5 : SUCCESS in 6m 43s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b8682a66120a45aebc9b2dc480be339f : SUCCESS in 9m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec6d91d2a17d41e4bbaed8787944512e : SUCCESS in 30m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1f5cce823cf54ddc84e7e7a99cf92f24 : SUCCESS in 1h 20m 43s","accounts_in_message":[],"_revision_number":1},{"id":"5f8248131f42af437931955f6a3c57e39ff79878","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 23:07:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3019c90c51e30551f48a56b0d8a4edbb53c9a2a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 23:09:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2304be2c00134230a258f1aec4dd3895 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"52c739f98a5743cb1821865f9819ecc5512c205b","revisions":{"52c739f98a5743cb1821865f9819ecc5512c205b":{"kind":"REWORK","_number":1,"created":"2022-02-16 16:18:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/67/829567/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/829567/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/829567/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/829567/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/829567/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/829567/1"}}},"commit":{"parents":[{"commit":"0a7e9306d40715c6c737726da71e8670d85d004f","subject":"update constraint for oslo.db to new release 11.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a7e9306d40715c6c737726da71e8670d85d004f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-16 16:18:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-16 16:18:14.000000000","tz":0},"subject":"update constraint for oslo.privsep to new release 2.7.0","message":"update constraint for oslo.privsep to new release 2.7.0\n\nmeta: version: 2.7.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: I59b1e5829619fde8bc43291ff42ff47ecb466116\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I36b79a556e7df8711c8f443d64aa78e7478819fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52c739f98a5743cb1821865f9819ecc5512c205b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52c739f98a5743cb1821865f9819ecc5512c205b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
