)]}'
{"id":"openstack%2Frequirements~829568","triplet_id":"openstack%2Frequirements~master~Ibf8efd7a59b45a4dd9ea46b37adc54084ed0c753","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ibf8efd7a59b45a4dd9ea46b37adc54084ed0c753","subject":"update constraint for oslo.upgradecheck to new release 1.5.0","status":"MERGED","created":"2022-02-16 16:18:49.000000000","updated":"2022-02-16 23:10:14.000000000","submitted":"2022-02-16 23:07:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829568-new-release","meta_rev_id":"744a26f5a1fb057c90027349a823b656163fa3ab","_number":829568,"virtual_id_number":829568,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-16 23:07:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-16 18:39:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-16 18:49:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-16 18:39:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-16 17:00:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-16 17:30:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-16 18:39:21.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-02-16 18:49:52.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"edd4f62df95977315557daf583931421f0c1d505","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-16 16:18:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96430926fa15ba9e1407f01614eed67f538c58cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 17:00:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/553a324ee0f8436f981924e651fe8b1b : FAILURE in 17m 42s","accounts_in_message":[],"_revision_number":1},{"id":"34be46363a907bbcfacb63d6f06aea38a8949103","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 17:30:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35b5d976118643f996183ee354b22254 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/232d334460fd4df98e211aaddafe65c4 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8794eb804ced4c4a84c725d0df2bc550 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4967e9c7848246b8a4bd713eb35e23eb : SUCCESS in 6m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e8801f9697054efb97cd96e241fffb21 : SUCCESS in 6m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/207b8d63eab044889102bcffa0c1aa77 : SUCCESS in 5m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c6c60fde5b98429ea78e8e86d9d648f9 : SUCCESS in 11m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1daa355dea0747d1a8ab2a0f13b8fe6f : SUCCESS in 11m 08s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f455a5dfaea54fb891f93f71081a46c5 : SUCCESS in 9m 10s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/531469ef3eed42d793a0964a7f1e9b7e : SUCCESS in 13m 28s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/d2b917a33f6e44b3ac0b86f8dd79ec65 : SUCCESS in 6m 20s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/aeece239aeda443b834b1bd29cca6ede : SUCCESS in 6m 47s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/38332c9fdc6d4572a4b0ed51562674a5 : SUCCESS in 10m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4caa78cfc2644f3986e61fd83838b4b1 : SUCCESS in 9m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/efb1adf3327b46c680964fd6979a95db : SUCCESS in 11m 43s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4e4af60acbbf458488c35a4e20d89f39 : SUCCESS in 11m 12s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/dc46535834ad430e83928fc83c410e97 : SUCCESS in 4m 40s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2780d0b90dcc4d0cbb48be4a654f958f : SUCCESS in 16m 34s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/244c3547e13a4b96ad2addfaf4fc54a6 : SUCCESS in 5m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/06aac273e83e4e10a928d202a7426d4f : SUCCESS in 23m 15s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/fe4b025bd55d4073b4f791056a614b4c : SUCCESS in 17m 45s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/b4dbc7a0affb4c35b7747bbb061c4279 : SUCCESS in 5m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/399aa1ca21f540b99558948ae168a63c : SUCCESS in 3m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/17620ede66e2496eb53efedaf8b4da8b : SUCCESS in 18m 57s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4335a742c10b47d5a7a8ee676e380c7b : SUCCESS in 6m 29s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/dd2afd81e3d44faa93735fa3812ba843 : SUCCESS in 7m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8b6178f9879c428faf0694071c424874 : SUCCESS in 7m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8cbc9625cabd442d8c9f3cdc05d1e056 : SUCCESS in 10m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/316485ae346c4fd3be32a187e1ef4cd4 : SUCCESS in 48m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1b23c3e85d44da9850d051f4f93c5d6 : SUCCESS in 1h 03m 26s","accounts_in_message":[],"_revision_number":1},{"id":"73e08a1c9f538a547f8f3390f09ccc3b84f531e0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-16 18:39:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"990849c1a71cdf5e446d3bf14bae1f2aa9c601dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 18:42:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5eb818fe28f1583375de8291bcca081b10e5295e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-16 18:49:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"65f0bf303a5533a915724e5e8fb9363b362914ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 23:07:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe33e7d737e04aa4a78a59f6f07bd786 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/545b12c7892f468397e53bca261fa0a5 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4467c9615e914334bd17c4ac615f7e79 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a90b1585879245d3bce35b3aaa902083 : SUCCESS in 4m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/18fbbb819d2f40aa9c49b83e2fd6eb0e : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0d973e2a5657404595c8d198b15608e2 : SUCCESS in 5m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9e22c455c53243669d0d11def154eebd : SUCCESS in 8m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9e55a1f1a52c4368958c927dc2c796df : SUCCESS in 9m 43s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b0d89be137e240a3af2d991043950cde : SUCCESS in 9m 33s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/46c216f4cda34d299190668ce2236e12 : SUCCESS in 5m 08s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c80c8febad074dadaef706e34ded8f8e : SUCCESS in 5m 52s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4e7978bd2c3d4e029969d85122628ccc : SUCCESS in 9m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/382139a50b4c437e9969c9959ac4c913 : SUCCESS in 9m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bb1a9e9c409248cab151ebd2fcbcce84 : SUCCESS in 8m 53s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2821dd1c75994b8fb352149d0fc3636d : SUCCESS in 14m 42s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/1568e933ed244b439f6ef722551cecfa : SUCCESS in 4m 21s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/79989c5bf4e34090be51aeaf2e7e99cb : SUCCESS in 16m 28s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d3973ef95a9841288e4739865f015512 : SUCCESS in 5m 45s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3ce4126971f441dc9305acde6f35398b : SUCCESS in 20m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a79a7ddc699f4002b80bbf602777e0a3 : SUCCESS in 11m 48s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/562c157c97664c3ab8843e1e66ac4a89 : SUCCESS in 5m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8d17190eb6454a75afb8e4c789f41d70 : SUCCESS in 4m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/18863c76921241b4a63d9a832b50a37d : SUCCESS in 21m 49s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/49abbc9e0c0c46e1978b254db06f0988 : SUCCESS in 7m 25s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/02d8a66c7c994f0a9358d4a76b42138d : SUCCESS in 4m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/be3db841960b4deb8827ef0f413b3135 : SUCCESS in 7m 25s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/819f1ef2bb514891993cc35410a7e28f : SUCCESS in 12m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/258c75016c5248d6ab8e551246d19a83 : SUCCESS in 31m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/baf1c5c9648c48c3b6bd106064e02eaf : SUCCESS in 1h 16m 01s","accounts_in_message":[],"_revision_number":1},{"id":"e94ed8560a11a267dec43265e48556fd607c5f50","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 23:07:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"744a26f5a1fb057c90027349a823b656163fa3ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 23:10:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af1aa63d7c6a4a2b8a063a3e9f4223f9 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e1a698ebcda120e785a8815570b4ffa0e29de751","revisions":{"e1a698ebcda120e785a8815570b4ffa0e29de751":{"kind":"REWORK","_number":1,"created":"2022-02-16 16:18:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/68/829568/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/68/829568/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/829568/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/829568/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/829568/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/68/829568/1"}}},"commit":{"parents":[{"commit":"0a7e9306d40715c6c737726da71e8670d85d004f","subject":"update constraint for oslo.db to new release 11.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a7e9306d40715c6c737726da71e8670d85d004f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-16 16:18:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-16 16:18:48.000000000","tz":0},"subject":"update constraint for oslo.upgradecheck to new release 1.5.0","message":"update constraint for oslo.upgradecheck to new release 1.5.0\n\nmeta: version: 1.5.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: I8a6f7d9b51d2cc880142b6b041a3df462c4cdc8c\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ibf8efd7a59b45a4dd9ea46b37adc54084ed0c753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e1a698ebcda120e785a8815570b4ffa0e29de751"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e1a698ebcda120e785a8815570b4ffa0e29de751"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
