)]}'
{"id":"openstack%2Frequirements~829574","triplet_id":"openstack%2Frequirements~master~I25080b457a9fe07cb62ac55cf6124687357c5a15","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I25080b457a9fe07cb62ac55cf6124687357c5a15","subject":"update constraint for oslo.serialization to new release 4.3.0","status":"MERGED","created":"2022-02-16 16:27:08.000000000","updated":"2022-02-16 23:14:21.000000000","submitted":"2022-02-16 23:12:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829574-new-release","meta_rev_id":"4a8ec752d594c1b45a16503dd777839422d76673","_number":829574,"virtual_id_number":829574,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-16 23:12:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-16 18:39:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-16 18:49:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-16 18:39:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-16 17:05:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-16 17:51:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-16 18:39:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-02-16 18:49:58.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"dd413909f44d1eccf6fd84c94cea768ffee26902","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-16 16:27:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a65ab5a956cfb5c9d9bc967c151dac19a5ece81","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 17:05:47.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5f5735adea9b4e4f9de98473069689a9 : FAILURE in 21m 06s","accounts_in_message":[],"_revision_number":1},{"id":"bb965445d4a19cc7a69dbe0f41df6b652822b3f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 17:51:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67dcd8d0b77a4f6fb3c7e190a35bd2ec : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9fd461dffdd42e4aba23af795f02f2c : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a29809ef3454145b1a22bf3d7cd7ef4 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ed6df9307004bf691a06df7921af4c3 : SUCCESS in 5m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1c4ee7648ea64514a2329c03ac97d687 : SUCCESS in 7m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1e38866946e54bd0a33839c535a126f8 : SUCCESS in 4m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/65638e4690b24e4ba372722efd7d3e2a : SUCCESS in 10m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7b4e98caa46346adba826c096d84e953 : SUCCESS in 10m 16s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/445470d95fcd41a9a19052a83854ef0c : SUCCESS in 6m 35s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4469d7da95f84ad8a2a21968dd92160d : SUCCESS in 11m 48s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/b9b08de73d884ee7b8a7d0e1e278a6f5 : SUCCESS in 6m 22s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/33dd7303345e426ba0dfa63ff1bc27f3 : SUCCESS in 6m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f15dba01536242e8bff984908770cbb3 : SUCCESS in 13m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1d7100b92b9e44a5bc69ed48c1b0e48e : SUCCESS in 11m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2e7e2e43d3cf473596817f15bc455fe0 : SUCCESS in 8m 57s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/520a3c2dd227464cb9a24caaa382230a : SUCCESS in 13m 44s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/324a59495b4c4bdcac5873704e0c70bb : SUCCESS in 4m 14s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d70d7af4195349bbbd010dd28b09099a : SUCCESS in 18m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5f9956908bd441ea873831ac5ccbb213 : SUCCESS in 6m 08s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c7c3edf30b454c9986f96634a5eb1aac : SUCCESS in 24m 05s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/96ec88fba7e44631bc9f0c99bf610fd7 : SUCCESS in 13m 21s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/5346532ff2644d21b7b48751f8c531fb : SUCCESS in 6m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/87bfbc6f926f41b39e9938575befd965 : SUCCESS in 4m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/42cbe3d0f6834105b5041be39b742ac2 : SUCCESS in 19m 55s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4035c93b620f40d1ab180d058282e69e : SUCCESS in 7m 08s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/115a2338b43d419899e7fee256e4e6ec : SUCCESS in 6m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/83e9c223bd4344b4ae6fa6d4198a31b5 : SUCCESS in 7m 57s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f7986907eed64486b61c50d312dcecb7 : SUCCESS in 15m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c637bd7d86e46f79fa87d874ff29936 : SUCCESS in 48m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35e2a47cac434c718c6d6b6fc3f05406 : SUCCESS in 1h 20m 10s","accounts_in_message":[],"_revision_number":1},{"id":"ed7dc5d2b18eefd8e88b3a9f17ef974fcb7b8b96","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-16 18:39:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"df5f6b85a1c5b4601f8e205a43630a4be867bdd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 18:42:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"67237bcb52dc8e35307470cdaa57f6a180369557","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-16 18:49:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"75c185c7dea0a056c878d117e388787afb911313","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 23:12:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04f982e4d2ee4cb2a54a2f1264469e41 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c60f0d67118c4f5a864cfb9a9052c4f7 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b574212768914de5869df1ecfaa7eb98 : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f11cfe90213b47e28f94b14cc2bca75f : SUCCESS in 3m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e475b442c51647b6b549836c6232c82e : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/caf52beddbbc4c29801dfbf796becc70 : SUCCESS in 4m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/21528ed245dc47f7ac390161e8465a8c : SUCCESS in 7m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/53003f98457e43129b7149dbc7bc4f99 : SUCCESS in 8m 38s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/96743e06fa9845138fb655093bda03ca : SUCCESS in 9m 41s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/6b017829128141a0bed662a24d926017 : SUCCESS in 4m 45s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b6105862410b45a18055326b1e441c99 : SUCCESS in 5m 26s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b82eaff2795a4c50874204b41038c1ca : SUCCESS in 9m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/09ff0209557c42c1b6145fc1762e78c4 : SUCCESS in 7m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/02b18f0e54bc437aa8e3d12e2dbc4c0a : SUCCESS in 7m 22s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6519ad29dfa1450db1d54e34ec7f2a0a : SUCCESS in 10m 43s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ef3686f9cf5844ecb5932bf2d929a529 : SUCCESS in 3m 57s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4fd4d81501d5441084a5d4a653e02954 : SUCCESS in 16m 15s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f622a719d6164d128c41f04d2b45854d : SUCCESS in 5m 07s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/66f7219c6d4d4b59878d873f2c1289fd : SUCCESS in 21m 00s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/eeb0c97079fc4d6699ff33b3ffffcf88 : SUCCESS in 10m 01s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2fb7774f055142b6a9f3219534d1f9d4 : SUCCESS in 6m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9cc3eeee54b44976898eb8c48f250799 : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/84f113d16fdf46e8b6dea49900c80cc7 : SUCCESS in 17m 33s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a77d8702085246acac6295432467b691 : SUCCESS in 6m 55s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b0ddcb3f12d840869831d048f150a254 : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/179d70ac7490449e97056bbe48d13387 : SUCCESS in 6m 33s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/db91c8f29ffa425ba14c5bc59210b914 : SUCCESS in 10m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e85cd25a1511432b9e47deb6f8cc7f31 : SUCCESS in 36m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/93f7c43b99644b3887a6aa896921764c : SUCCESS in 1h 25m 40s","accounts_in_message":[],"_revision_number":1},{"id":"3023a2da62195b791a9520482a3e137b78bd8f12","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 23:12:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4a8ec752d594c1b45a16503dd777839422d76673","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 23:14:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65500fb340cf4ac59a0698d341196a74 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"056f15860628f7986ae63109d097a3411693dd66","revisions":{"056f15860628f7986ae63109d097a3411693dd66":{"kind":"REWORK","_number":1,"created":"2022-02-16 16:27:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/74/829574/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/829574/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/829574/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/829574/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/829574/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/829574/1"}}},"commit":{"parents":[{"commit":"0a7e9306d40715c6c737726da71e8670d85d004f","subject":"update constraint for oslo.db to new release 11.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a7e9306d40715c6c737726da71e8670d85d004f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-16 16:27:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-16 16:27:07.000000000","tz":0},"subject":"update constraint for oslo.serialization to new release 4.3.0","message":"update constraint for oslo.serialization to new release 4.3.0\n\nmeta: version: 4.3.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: Ia5ee30f53c3eff165dc177e782cac88ab6aa3711\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I25080b457a9fe07cb62ac55cf6124687357c5a15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/056f15860628f7986ae63109d097a3411693dd66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/056f15860628f7986ae63109d097a3411693dd66"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
