)]}'
{"id":"openstack%2Frequirements~829581","triplet_id":"openstack%2Frequirements~master~I327f28fbece2cea0d66e15c3e3a34c7c83e335ca","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I327f28fbece2cea0d66e15c3e3a34c7c83e335ca","subject":"update constraint for oslo.utils to new release 4.12.2","status":"MERGED","created":"2022-02-16 17:35:22.000000000","updated":"2022-02-17 18:23:09.000000000","submitted":"2022-02-17 18:20:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829581-new-release","meta_rev_id":"dbbd4592a17a5e8306f72e895448fffc8457867f","_number":829581,"virtual_id_number":829581,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-17 18:20:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-17 00:31:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-16 18:50:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-17 00:31:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-16 17:51:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-16 18:50:06.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-16 19:04:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-17 00:31:32.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c5f95ba6372ce0bd9735a537babe50e2a4007184","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-16 17:35:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4285d8be4bba18c863f42f10a4dd79c87b4c4439","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 17:51:32.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f663fcfba04640818bddf2d0432c87ab : FAILURE in 13m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b3958aedc806f5723a66a3e788dbee9646274f59","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-16 18:50:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5df8ea1e6cedefb33645474a60afb1df100ce69d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 19:04:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b1d958d243c472e8d624bbb3edb2dbf : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f187ea9451744a7b3f4c53f55c5c510 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fbe2de93a504e06932d6162b00606a5 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10691c6a464e43e09002bea7f10b4c8b : SUCCESS in 5m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f9391c501eaf41f48285bedc3f948a5e : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/859136dcb9434cd6a545043f7a438722 : SUCCESS in 4m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2112c041eb55402c93e3370fd155ff95 : SUCCESS in 9m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3478701046944ed59fd67c03001c5f87 : SUCCESS in 9m 04s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/26a35399817f4f9f9ab88f8349b45bf0 : SUCCESS in 5m 29s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6a6327ae38e549a9ab6829c069aa2ab7 : SUCCESS in 9m 59s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/192d1922e12f471da9868b0a087da7b7 : SUCCESS in 5m 11s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/de849606837d44ba8b16b7454f20f458 : SUCCESS in 6m 08s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9ff8a488cfbb44a18ef129f5d2715c1a : SUCCESS in 10m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6166ee0e35cb40a3830d1691b71cab8a : SUCCESS in 9m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b56fad1e1e234763a9fd89a9b3155ac9 : SUCCESS in 8m 39s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d57ba9f84e0445c1b54c25d3fdf5625b : SUCCESS in 12m 27s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2b3981e6069c45748788a5ec371a2b72 : SUCCESS in 4m 32s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/65fc64ffa9c24fe0ad83b46330b647da : SUCCESS in 20m 23s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5844b3764a264a0da5d8352742cffdea : SUCCESS in 5m 50s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/5c5de0af15334b97a931f3fcded3cc5f : SUCCESS in 21m 14s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/bb01e870cb2142c4bb058c883a7b55a6 : SUCCESS in 12m 27s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2d9ec01d53fe470da76dc91b70e2943e : SUCCESS in 5m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/85e168ea6cd841e382ea33fabbad971c : SUCCESS in 3m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4e79e906532e4c7e97f41fb892251041 : SUCCESS in 19m 03s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/eef8d2ac87d54b569e98faf52b9805ff : SUCCESS in 7m 05s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2559d045af1e4de4ae5f0a77de05df1b : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a867b33d5f984edabe5e65fd3ff3a69d : SUCCESS in 6m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8e3cb51177c9478d93134dcb284cb0b0 : SUCCESS in 14m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/332fba9968df42afb18bc5f82e0ea84f : SUCCESS in 45m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e24207d19c24f878fb39b837a3e7a78 : SUCCESS in 1h 26m 25s","accounts_in_message":[],"_revision_number":1},{"id":"1caa2f2a78c81cd48cc873d60ea42037a4d60a81","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 00:31:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"484af7ac304de802783f5d9b59a9979504ecf208","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 00:32:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fe6c5b0bbc69623ebf2cfde1f1a10d139e7bea95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 02:15:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67bf1541c78f4929876810a7b33e8e83 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/118e23f8c50843c0a9b3235dbdba8265 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e425b64b0954ea5b640b35ee5e313eb : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9cf63dcd4b846b0a30a237f2e7a250c : SUCCESS in 5m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1e4873cd074b4d9d81e6856d95dddb97 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bbb92f5dd11f4150a07d9703514f29f1 : SUCCESS in 5m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/46b268c63c044a33ac106972561ecbae : SUCCESS in 9m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f0477787c101455c99dbc524ad78d851 : SUCCESS in 11m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/bff0d51b08b342ca86258285cf97c19f : SUCCESS in 10m 02s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/60bcb34529f5414bb5345c0b76c73642 : SUCCESS in 5m 45s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3800549d5a334cf1a1c5a277871147c0 : SUCCESS in 6m 49s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5b6654c2137c41f9884bce47401b2af6 : SUCCESS in 13m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d90c98c50ad4467cb79f3f12d79b1845 : SUCCESS in 8m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/889fbd6f32494938b0f92b84d757330a : SUCCESS in 10m 49s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e624204bffab40e79dfcfb4ba3d8f4ba : SUCCESS in 12m 57s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/42e04dd1d1ab4ea78ee7223c77f2f3ba : SUCCESS in 5m 03s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d01db1cbea0544c7bd4cc7c76fae9806 : SUCCESS in 28m 30s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5841cc1988144424bc19777745e49bea : SUCCESS in 6m 35s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c84a4f5bfcc345158e9685cc004d2e1a : SUCCESS in 25m 40s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/20ec5e1b24df422e8520209c72206593 : SUCCESS in 12m 19s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c5d5845876dd4ba1b3611b7a8113557e : SUCCESS in 5m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5739b798bd1640788f16d9b8bee8cc5f : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/072b7fcf87744627a391f93a29073281 : SUCCESS in 23m 09s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7a877eefd68b4a06b6fde84918255cf5 : SUCCESS in 8m 29s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/addda1c6e4914f1b8648f732855e083f : SUCCESS in 6m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/97d54f065969404aa2292bbb1d6258e8 : SUCCESS in 8m 11s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/12b1ed6ac9864d33be6eb4a43d1e1856 : TIMED_OUT in 41m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b02e154362f4ad1a9d393779e30b9af : SUCCESS in 1h 05m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41194f3ec00048769f7e6e44ac0d55d3 : SUCCESS in 1h 37m 35s","accounts_in_message":[],"_revision_number":1},{"id":"58b897647794064f6c8f06b6818ec537e49c1311","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 02:19:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1422bd6e39cf70940c77fa66d262f040e6966758","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 04:17:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/121afa5df1f24b8b987d461c52cb624a : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b4ba91665d048248f04903ce71755fe : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8d1e0c7c8444cadbb9bf41b2b97f196 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c257d85db034deab80de2fc510081fb : SUCCESS in 6m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cb78f719f1124ccea8c613ff17fd1a57 : SUCCESS in 6m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7334aebb18004ce5917f153bcee1876e : SUCCESS in 6m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5eccddd16b5c4856b4c14c3a744420bc : SUCCESS in 12m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4a960de2169744deaa33c4557d91ae22 : SUCCESS in 9m 42s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/9ddb976eaccb44e68d608405ac934261 : SUCCESS in 7m 53s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/da3ed5b8cb5c40d8994db8cab4e9cc82 : SUCCESS in 10m 56s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/96e2bd4205274cc48f9ffac61b1fb169 : SUCCESS in 7m 55s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/dd76dea0929f4a8ca4593c2c58fb6839 : SUCCESS in 9m 35s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/01fa130a87804c809f01273f27256e64 : SUCCESS in 12m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/58fe3d2e46da45dcbfef2da82a4f2281 : SUCCESS in 11m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/da97ab74c2434bb1b11275f43012f09b : SUCCESS in 9m 49s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/3a8b1613ef644450a56ebb3ff9ad90b2 : SUCCESS in 20m 27s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/4211ee0dc4fa488db03c958554d02e9a : SUCCESS in 5m 48s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a33fcda2b92048ff8799406b3044cd62 : SUCCESS in 17m 10s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5bfe2c99874047a6aed05daeacc9ec14 : SUCCESS in 8m 22s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/0dcc7b3ef92f46178aaf14b221df5366 : SUCCESS in 20m 13s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5608370f1f5a4686afab3ef98cda6a9c : SUCCESS in 11m 30s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a45c44fbb8bf4c18973939bd0ff466a1 : SUCCESS in 5m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b6351a69c2bf4d68b66041b276eda903 : SUCCESS in 5m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/554a276d61254904a290807a28f15c8e : SUCCESS in 22m 33s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7637e018f15842679ad0bea23f9c0a0c : SUCCESS in 6m 55s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9d78bd1da49b4e7eb2e2d20cd2e13712 : SUCCESS in 6m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/99f7539b8e844026a29b75145a058435 : SUCCESS in 8m 53s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/94b67dacb4a44461a3225bf41e0a69d3 : TIMED_OUT in 41m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e217d329233465193ee0af1e11f1ccb : SUCCESS in 49m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d898e620425a41fb82b64215ed2f1090 : SUCCESS in 1h 22m 39s","accounts_in_message":[],"_revision_number":1},{"id":"df7c3b3765c27e944e226ec8c58236b83167569f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 04:20:30.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/182130668e314bca8c6e40eaba97725d : FAILURE in 14m 59s","accounts_in_message":[],"_revision_number":1},{"id":"679b988080142618d95c7634ebc169a23fbc040a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 04:24:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"80cdb83c480e95d243563e06bc9358503fa91f01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 05:04:27.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d73fb5f3267d4f88a7e235ddb6dfa6ae : FAILURE in 12m 44s","accounts_in_message":[],"_revision_number":1},{"id":"0389fca0e330e135170ee6fd58ae5b9c5b963746","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 05:47:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26250d7178b8420d9995440b29f7035c : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06136032890d4951854f759cc3f87738 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb85ae5a2986413b9e61f24c2f37ed5d : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f3673049f7a4f2f8228a81713dcc5ef : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/40fd391d25e4403ca62086401731511d : SUCCESS in 4m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/162e6636341548ccb9857d424f48e4c6 : SUCCESS in 4m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ad774f67da1c40e99cbba7e2c0b2967d : SUCCESS in 9m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f2f0ebe075234407b728fe560871b4e4 : SUCCESS in 10m 14s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/1d1006f9050c429c8ff4884e1827263b : SUCCESS in 5m 40s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e2390e15ab4e4594a168db1ba7e6297c : SUCCESS in 10m 09s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/3e92e172bd524ed69c407d0f477d5935 : SUCCESS in 5m 00s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b65001154b564b2d9807c393cf3a0d56 : SUCCESS in 5m 49s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/83d9f85050f549d7a0a5894cdf2849a3 : SUCCESS in 8m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cd29b2a971314a34bd91085333026f05 : SUCCESS in 8m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/280c458582a44fba88ae51dd35d76b02 : SUCCESS in 8m 02s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/fb943ced9fcc46a4961f009ab121307c : SUCCESS in 11m 48s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/dac8587cba024b5fa4c204531239a99b : SUCCESS in 4m 12s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f8f495ace9bc4b24934cd71410ab9c60 : SUCCESS in 15m 44s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/797bb50653b34162b7d5372bbeabc228 : SUCCESS in 5m 39s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1808d669099c4f3aa19067f11dd34431 : SUCCESS in 21m 46s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6eb3a70684624481a6de897f7de41f03 : SUCCESS in 12m 35s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/664197a5613d41fdb66e666ce5d3b973 : SUCCESS in 6m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/549da45874504edf8916885f5309ef69 : SUCCESS in 3m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6e5eff3e4bb648feae0b87808cac741b : SUCCESS in 18m 47s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7c11ffe39be4416285eefc568c899402 : SUCCESS in 6m 11s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/eb3f689a2ee74721aa40ac5aa85d68a7 : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/940b946aa1644524a7f4875d55f5192b : SUCCESS in 6m 27s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/178960c4b61946b5a34516bd1662263e : TIMED_OUT in 41m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3741c8e1c7bf4f309f4d37836d217c5f : SUCCESS in 33m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/205509fe63374a88baef054101d02324 : SUCCESS in 1h 19m 59s","accounts_in_message":[],"_revision_number":1},{"id":"21ecc52064e557248b00f8a85e9590db7cab764c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 06:02:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f1d600ca12d0e8b154946986a7caee06a1e097bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 06:49:36.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b089a05db8734cfebedbe865dd12ca3a : FAILURE in 13m 55s","accounts_in_message":[],"_revision_number":1},{"id":"b8342db4ca6ac18e97f78e9e61a0e06a99ef7253","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 07:31:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8b7006100b34d3cbdad9d08251f8a5d : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6414b45d6e7346d3aca77084b0f9b152 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6aab4e3f7cd4a8f8480643ec58c6b4b : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f2101add65554f6c9a9548195b792548 : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/95f4470b3c8e454d8f6f307a70ac9dd9 : SUCCESS in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/36b6a3bd46344fb0b5b9ee9b13788c08 : SUCCESS in 4m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0ac10aa2ff6b49149d0470103c1436bb : SUCCESS in 8m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/89ffe40b9c8b4148a65c0fd29a5bfe5f : SUCCESS in 8m 56s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/1cfb2282ecd541cf8faa657cd292a6d5 : SUCCESS in 5m 57s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/892c87b2d86b493899231c99081ef5c0 : SUCCESS in 10m 24s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/0fe15547996e40c386792b0b5586ff91 : SUCCESS in 5m 00s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/54801f76fe36415a80a1d3e7e9e5ba24 : SUCCESS in 6m 07s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ccaae649443f40c19174d509c1b48a97 : SUCCESS in 11m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8e67ab333ec2472b80bd11052dfa2337 : SUCCESS in 8m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/52d3405fc3404d6fa4142ec09ed051d7 : SUCCESS in 9m 27s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9a21ee7641a542eca5751036cf5c7be9 : SUCCESS in 13m 22s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ec167adf82fc43689bf7a0d1834a4945 : SUCCESS in 4m 30s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0302d4e7b14f488e9ff485c2af4b0323 : SUCCESS in 15m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/832197e497d54b0eb1d6a6dd774cd8ed : SUCCESS in 5m 47s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7fbc5a9057104c368df28eead5375526 : SUCCESS in 25m 43s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/25a63c93d49642df9259ef65adb201d5 : SUCCESS in 13m 40s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e6e56ca3ed704dfbb26daa97d54e1842 : SUCCESS in 5m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/aeafca05f8a74e3882699733187f3904 : SUCCESS in 3m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6229fc95f96541619536e0401bb2fd6a : SUCCESS in 19m 50s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e07788d109634e8795305f5f6c751089 : SUCCESS in 6m 51s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/eca4ff97c8764ce1b59276ebe37ec264 : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/98362006492c45f1b92ed563f9ca3ef2 : SUCCESS in 6m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5b00dc46915045bea4496648e2bb2afb : SUCCESS in 11m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b9c00f3cbd50407f9fd20634647a8663 : SUCCESS in 36m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/572dd50b8f0b4e308364d9f44e15026d : SUCCESS in 1h 22m 30s","accounts_in_message":[],"_revision_number":1},{"id":"7bd8729e36b57ebd7a84ec0eac30e31622f8fd2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 07:32:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c9ddf655b8bb54671befd1d321eb1a85807bb4cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 09:34:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef327a7d8da94d73a0d9955955a95c21 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0d1580005394f43adb6665a54506e59 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07c4b00b23194d599c0b6cc069ffd4b4 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7e5f07f11f94945a8348ef33d4c9f4b : SUCCESS in 4m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fd35bddfa7be45cf970419c04b8872f0 : SUCCESS in 4m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dc10b1e1da7c40d797aa2c3930ee7798 : SUCCESS in 3m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2e7e9671f1284473916044136e6a5cb3 : SUCCESS in 8m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/73742a09059146f0b01c64dbead4de8d : SUCCESS in 8m 46s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d6285bd3ddf4420ab8e6a675c13cef68 : SUCCESS in 10m 18s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/46fe1eb379744ba6a22d5556310e7788 : SUCCESS in 6m 56s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f25829fb8bbb4654accb8a550eccb9ad : SUCCESS in 5m 59s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6e290ea9ad294ceab4377f19af1e00c2 : SUCCESS in 10m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/64495be9f2534b338913275ebb9999ea : SUCCESS in 8m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e9a26f32ebbf4caca65c77428bcbb055 : SUCCESS in 8m 54s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b94b317a94cf4a9fa57feff9cd8ec815 : SUCCESS in 11m 28s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/68641170963944c88a0b19ea00193ba3 : SUCCESS in 4m 52s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e902a463f4c44f718ebb4ba3355bfac0 : SUCCESS in 24m 36s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/dae4ac2d413e4b27a4536a50ffc5bed8 : SUCCESS in 5m 45s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c6b3d66263a5490e8b30be846f78e9ba : SUCCESS in 27m 58s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9e1d66bd83c445b1b9a604acc462f2d4 : SUCCESS in 11m 33s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/425dd74d6d4241668101a5c806387a4b : SUCCESS in 5m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d34b19a3c3094b959699f56eabba405b : SUCCESS in 4m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5c696ce47e7c40dd8966f323c47bcd65 : SUCCESS in 20m 34s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/731f3d1cbb6b4e9cba475b35247c3418 : SUCCESS in 7m 22s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2071b0afe79c4e9eb327add6ebb0d0c9 : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7fc774112f63493b97c6b9aba519ce84 : SUCCESS in 7m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5a6c43b6aacc42a69208ac3fe586d068 : SUCCESS in 12m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/191aa5482f2e407dbfe394b2fc22cb92 : SUCCESS in 53m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8900ff04d2894f84a9c6abd9da3d3e18 : FAILURE in 1h 17m 33s","accounts_in_message":[],"_revision_number":1},{"id":"7a5922adfc7b292fff6e0fb096e58e2fbf0f0fb6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 14:50:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8c6135f6aea77ae13d9651385ffbdddb23d91829","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 16:13:00.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a871a8f5e90940bf8929d5084e4db4b4 : FAILURE in 15m 01s","accounts_in_message":[],"_revision_number":1},{"id":"5319fe6672543101beee8d6c468131a567c2d352","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 16:48:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f119c7565794070a10131fc026de351 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/214a34ef5671445e97a481723f3b5086 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3e3f41800f14568a8b16f94decc924d : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d23e4ee05511452a9785b6492d1cf4a2 : SUCCESS in 5m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/783b8cc2b5254fd2a67dba321ba5fa60 : SUCCESS in 5m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0d3cd5508f9e4f0a99e9f23bed0995da : SUCCESS in 5m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5ac594ee8eea480f8a7c36d010d9e7f2 : SUCCESS in 9m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/208c4e5169a44bb68805d0788b7caae7 : SUCCESS in 9m 08s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/431c8ceb0e334b0ba2d37e3ec7623f50 : SUCCESS in 6m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4e3f8c7952e74830940d4f4001df4a62 : SUCCESS in 10m 24s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/1f0848c6bc5643c4a4c7393bf97ab826 : SUCCESS in 5m 27s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0d27392ee8ad41dbb99c0feee062ef01 : SUCCESS in 5m 56s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3d04038d6fbc49edaaff3b71746da750 : SUCCESS in 11m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c1e10c85891e4101bcfd8485b858903f : SUCCESS in 8m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/22ad0a423eb240ea91d160fa0bdb2ef6 : SUCCESS in 11m 12s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/692d06bc7b5b4df3b0d595f39da5237d : SUCCESS in 13m 19s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c7888eadbe3a4b5585a3ce2e912b609a : SUCCESS in 4m 48s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a1b9bc13be434b31bebb180b9a5962d8 : SUCCESS in 17m 30s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9f8b037a47e84c73b7117ce5fd1e2f14 : SUCCESS in 6m 22s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/0ea4cab0d5d644699bb27458f354a175 : SUCCESS in 26m 50s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/190f2b901f394eb48e3dfd7cfb3661c9 : SUCCESS in 12m 53s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/3d504e1d02f2474f8ba6b16aff411623 : SUCCESS in 5m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8d2c7912a4374c0188897872dc729a1b : SUCCESS in 4m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/48462813074649b581a055718d45750d : SUCCESS in 20m 33s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/827ecc77300f4d6a8b3a41e7e31e6348 : SUCCESS in 7m 48s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/210e8291cfc24d7dbf90dde274e5e05b : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5519032073ab4a9cbbedf8258c69158b : SUCCESS in 7m 38s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c09766b8a92c49fb9b3518e90ddd63d7 : SUCCESS in 15m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/08590d3e14234599930bed8d356df709 : SUCCESS in 55m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d9b02c460e97459fb19e9f0d50af5d96 : SUCCESS in 1h 48m 40s","accounts_in_message":[],"_revision_number":1},{"id":"6d05d748ea46732c5d6c8ef6e38f1c037fba46c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 16:50:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b12868db32a55c08271b43e4ea087895cd7feb26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 18:20:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e05bf8befed34fb6b7f8be3fa8c45454 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa3465da1d6340b28ea5a77cd82118b9 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cad2dd4a69ea4c3f932f4e6b66439e0a : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/444482cf3fa44724b5d1a99dcbc53f4e : SUCCESS in 6m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/44f9980d60dd46abb8696cf00625f21f : SUCCESS in 5m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e432a6be9e3445f9b4add0f95b9be468 : SUCCESS in 5m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/01ba34a851df4e8abcda67d6c9dc384c : SUCCESS in 10m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cf4edb8b79b445a78e33e7e321485389 : SUCCESS in 11m 30s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4c59c15632d9496ca2c35f669155de20 : SUCCESS in 11m 44s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/daef2e48159e4079bf4ef91eab254068 : SUCCESS in 7m 22s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/86b17fa3473d4609a46ac31fd8bb9921 : SUCCESS in 7m 14s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/734807373aaf4982a9d8e2e317e64d20 : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/22958ab4653440d3a131cbc88f0201e3 : SUCCESS in 8m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c5372322d7f14116a6aee371321b9cf6 : SUCCESS in 9m 49s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/38bd353893da4af8a6283d176f546355 : SUCCESS in 12m 30s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/9ea12fe034c24c32a54a919b76ac0f45 : SUCCESS in 7m 05s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/83d4c3f1ad8545d28d2ef843be49bb88 : SUCCESS in 20m 23s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ecb8c4f109b1472ea52c804f62eed907 : SUCCESS in 7m 20s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/758fbbf8509544319e19657a94f4981b : SUCCESS in 22m 33s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/509c9ae7e7b64c9c98bd2392d2838aff : SUCCESS in 13m 40s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/45a07886832947a5a6b66c70f7f3fae4 : SUCCESS in 7m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fa2c915ed71f4fc9a064169677bede82 : SUCCESS in 5m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6f72b04eb9f1414a9669c7e07cf37fdb : SUCCESS in 20m 42s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/fe07d3999bec4b6599d061499dfda920 : SUCCESS in 7m 32s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/24e9aa79ebab4bb09c1d4b1e041ef124 : SUCCESS in 6m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a40837e95c10491189b1baccdaba787e : SUCCESS in 7m 40s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c3d770cd02fd43d897442df1febdeb69 : SUCCESS in 17m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0037614dab83443ab3ade0e4f1e6898a : SUCCESS in 46m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7c0b10e8f0d4443cb87311867145afcc : SUCCESS in 1h 29m 29s","accounts_in_message":[],"_revision_number":1},{"id":"341efd9de515629ea0f020388ac2e79f1d389b86","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 18:20:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dbbd4592a17a5e8306f72e895448fffc8457867f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 18:23:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30f2472ff33f4258913b4b4b23c6b33b : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0664319a9d31604a1c8c04531556094303de81ae","revisions":{"0664319a9d31604a1c8c04531556094303de81ae":{"kind":"REWORK","_number":1,"created":"2022-02-16 17:35:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/81/829581/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/81/829581/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/829581/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/829581/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/829581/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/81/829581/1"}}},"commit":{"parents":[{"commit":"0a7e9306d40715c6c737726da71e8670d85d004f","subject":"update constraint for oslo.db to new release 11.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a7e9306d40715c6c737726da71e8670d85d004f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-16 17:35:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-16 17:35:19.000000000","tz":0},"subject":"update constraint for oslo.utils to new release 4.12.2","message":"update constraint for oslo.utils to new release 4.12.2\n\nmeta: version: 4.12.2\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: If4720eeca0c0cc66d485a55aad14b64cd4152408\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I327f28fbece2cea0d66e15c3e3a34c7c83e335ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0664319a9d31604a1c8c04531556094303de81ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0664319a9d31604a1c8c04531556094303de81ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
