)]}'
{"id":"openstack%2Frequirements~829582","triplet_id":"openstack%2Frequirements~master~I59e7e905e6607d9003edee1b7ab2d42c0a973788","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I59e7e905e6607d9003edee1b7ab2d42c0a973788","subject":"update constraint for oslo.cache to new release 2.10.1","status":"MERGED","created":"2022-02-16 17:36:40.000000000","updated":"2022-02-17 02:51:38.000000000","submitted":"2022-02-17 02:47:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829582-new-release","meta_rev_id":"611e0d37e1370032cf252aa795df3f5d41ecf450","_number":829582,"virtual_id_number":829582,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-17 02:47:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-17 00:31:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-16 18:50:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-17 00:31:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-16 17:50:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-16 18:50:14.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-16 19:21:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-17 00:31:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"493f8b5a642140c3585e3380c15e9c322e228fe0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-16 17:36:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50afd5e6b2127a82edb21c19fdd9630e0436fa71","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 17:50:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/dd99e1cbf7494ab98121cd0cf1b79ceb : FAILURE in 11m 53s","accounts_in_message":[],"_revision_number":1},{"id":"bd31ac247d90bf97819fd925fe94d391bd9ef6e3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-16 18:50:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c2ee835bc8829c9c806cf4b2aed8767994bcbf8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 19:21:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcf65b26ec5a47f6aa9f3166cfc4f1d9 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/351a9ca0686d49cfbc535cfbfb3faec5 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91f4352166d0488c87cfd1deb0033ac1 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f379e67a5f141c4bb8f0ff55e5ea904 : SUCCESS in 5m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/223f82a19dba4e1b939b9beb3fbdd575 : SUCCESS in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0136a97d20c0456896994908986134e1 : SUCCESS in 3m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/abe71dec63914822b1747238c4eda771 : SUCCESS in 8m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c55dd7ef962b4961a7dc042a264657d7 : SUCCESS in 8m 49s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/7eea4641b2fc4a9198def15f7d87bf30 : SUCCESS in 5m 22s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8a77055646ef47ffadaad156957c672e : SUCCESS in 9m 39s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/5feadfe7e75b4ea8808c223de035456d : SUCCESS in 5m 02s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/209cdfb8f6394faf861162e5e5d72db0 : SUCCESS in 6m 06s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/df0b9baab0f0417686780bf6a57e35fc : SUCCESS in 9m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e8a0d201c59d4854b27a9b763f6abb03 : SUCCESS in 9m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d364265be06d44008a40157a28f167ba : SUCCESS in 9m 06s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b52ad1280be4494195682c7a9eede3e2 : SUCCESS in 13m 15s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2efe70c6f0664d5c8924a55cee61fa23 : SUCCESS in 5m 20s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/73a8bc0027f040cfa0d51492e7d3e6c9 : SUCCESS in 17m 23s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/bcf1754ef8e6409583c403029e33dc91 : SUCCESS in 5m 43s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/275a5ea1d81949f7859f5a49d79b613a : SUCCESS in 26m 25s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/28e710e2072c4eaba3e317bde4f2d722 : SUCCESS in 12m 19s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/31e99f52cd2a4ff2b930a1f3c076824f : SUCCESS in 5m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/86e2334ff2e548f6aedd4696a46909bf : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8a8e4256483946ad9ca50d646c45c44c : SUCCESS in 21m 03s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a6e1967d467647c58daf709669c1beef : SUCCESS in 6m 39s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/cdd1bca6a08743f6b405bdb560c8b3c0 : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a36ad8cd87ac466a929833d227e18ab1 : SUCCESS in 6m 14s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/11be93d389bb480bab8664d454962827 : SUCCESS in 12m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a24f9a901fd44e8fa1c6bcb49d374d20 : SUCCESS in 44m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/78cf8fbbd979449e85c66cbb293a02a2 : SUCCESS in 1h 38m 04s","accounts_in_message":[],"_revision_number":1},{"id":"c3a4040cfe5f48306967a58f869d61455acdb4ec","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 00:31:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0b011cd30ba00d172bbe39d31315ff6818abbd8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 00:32:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5077dc0e7f1ff0d5312ce228d8570d16c493ba03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 02:47:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88ce67be75434001897005237d94fbcd : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cff472a8fb924db197ffbb06da7ce32f : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52000724dd9c4e53a459219045586db6 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd3f2ee33ef14926a2faccbc8aece7dd : SUCCESS in 3m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/403e1137ba0f4189bdf91cbd55d1602c : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2a0edaab8ee1481ab27f9586ba4f0ab6 : SUCCESS in 3m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cca77bced80944f6826d96b5fee575a9 : SUCCESS in 9m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d96044f31e734f0d98d4cc039366d823 : SUCCESS in 10m 10s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c5623c92d51b4e08bca63b62ceed9238 : SUCCESS in 10m 33s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/80ee85dbb7b74d11a1de55e6e79aadd1 : SUCCESS in 4m 50s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e9d7acfe6265418abaab12174b16039b : SUCCESS in 5m 25s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7b0852ae3147480ea0231edf7d7b1d3d : SUCCESS in 12m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/92cf3a48641544609ac1eeed3b213ed9 : SUCCESS in 8m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/52d4a9489fc44e81be01872ba12260c4 : SUCCESS in 8m 32s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/36b1e02ae41c4b51ab7d73656c306f96 : SUCCESS in 12m 35s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b611406e93b74a66b5d7858c00b75eb9 : SUCCESS in 4m 15s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/fa062e6400e541c78df6aa9333289ab2 : SUCCESS in 24m 38s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/75b05a8386e24470b488908dbd1f1aed : SUCCESS in 6m 01s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b8bf3443ba1141a4be749e1fac875f12 : SUCCESS in 26m 36s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/bbd8a62579e04b3ba006ce33c7250362 : SUCCESS in 12m 29s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/1357f45fe8f044518cc19b4d4c4e191a : SUCCESS in 5m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0ee47aac6e044a90a998b253b5430c86 : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5130351a8e274339b9c0a873e6803812 : SUCCESS in 18m 10s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/fe8eb4a0ad834c438cd83a9ec3088cc2 : SUCCESS in 5m 44s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f7c017a85c0d47c489453139c51e11a8 : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1d1d69982c01487da9ce4dc41cee0c4f : SUCCESS in 6m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/878aa7814a21403c9549114c0ba4f5e0 : SUCCESS in 11m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/323cdedb81e24f339b4c2567731c5dba : SUCCESS in 49m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1837c6d64ede4ed4918788e9c817f99b : SUCCESS in 1h 27m 24s","accounts_in_message":[],"_revision_number":1},{"id":"00231d3b6292b13a5cb577f6485f4ecfcd19226d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 02:47:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"611e0d37e1370032cf252aa795df3f5d41ecf450","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 02:51:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a02e396f73a749f5a79a4bdbed0bda6c : SUCCESS in 2m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"270dac4adec79e5f23caa14ff8085b9aa43308b3","revisions":{"270dac4adec79e5f23caa14ff8085b9aa43308b3":{"kind":"REWORK","_number":1,"created":"2022-02-16 17:36:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/829582/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/829582/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/829582/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/829582/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/829582/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/829582/1"}}},"commit":{"parents":[{"commit":"0a7e9306d40715c6c737726da71e8670d85d004f","subject":"update constraint for oslo.db to new release 11.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a7e9306d40715c6c737726da71e8670d85d004f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-16 17:36:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-16 17:36:39.000000000","tz":0},"subject":"update constraint for oslo.cache to new release 2.10.1","message":"update constraint for oslo.cache to new release 2.10.1\n\nmeta: version: 2.10.1\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: I8ac6fa0c2a05dc651337c9a039bedf4328266f6b\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I59e7e905e6607d9003edee1b7ab2d42c0a973788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/270dac4adec79e5f23caa14ff8085b9aa43308b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/270dac4adec79e5f23caa14ff8085b9aa43308b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
