)]}'
{"id":"openstack%2Frequirements~829583","triplet_id":"openstack%2Frequirements~master~I79a65e00d3cf29e9471d84521936f47fa1b90e01","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I79a65e00d3cf29e9471d84521936f47fa1b90e01","subject":"update constraint for oslo.middleware to new release 4.5.1","status":"MERGED","created":"2022-02-16 17:37:59.000000000","updated":"2022-02-17 04:02:07.000000000","submitted":"2022-02-17 04:00:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829583-new-release","meta_rev_id":"41ecc73a90c29ea68198b2d50c4aaed3f621a03a","_number":829583,"virtual_id_number":829583,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-17 04:00:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-17 02:18:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-16 18:50:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-17 02:18:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-16 17:54:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-16 18:50:21.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-16 19:01:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-17 00:31:43.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-02-17 02:18:52.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"8004a5c0f00698c3beef76abe053265efacd4db5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-16 17:37:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8049408d0d6321f9ee9dcea046d8caf6a6b22a0f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 17:54:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b395e1138458447f8407e34a24b6398f : FAILURE in 12m 15s","accounts_in_message":[],"_revision_number":1},{"id":"3dc56f614bf3dc8cfa5610b14a672a12c38bca33","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-16 18:50:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b952927e08c500644aa03a88564ab390a73cb37b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 19:01:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af97af68016f43c4acd035608eff83b5 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc0ba1f6040c43038a85e33bba0853ec : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a6757e317ce482fa1ba6722f51e8592 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a4f94cb88b84b95a47a0c40db261cd1 : SUCCESS in 5m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c05b1704c8424180a978de238699f9f5 : SUCCESS in 4m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/00d1ed2189d7436e8b2368b2ef36ab0c : SUCCESS in 4m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/16c7f90564dc40b492f56e343e20de12 : SUCCESS in 9m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/82aaec8c0eec4ea888c5ff3dc60f974d : SUCCESS in 10m 15s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/71b1b0d165df4ae697c51066ac9bae4f : SUCCESS in 5m 36s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/86be06d2eb7349859296a85b7733b9b8 : SUCCESS in 10m 00s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/1d3f1696d5684a8e976d80fd89879f60 : SUCCESS in 5m 03s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2c8fd5defe2445a5a5f426be417625e5 : SUCCESS in 6m 23s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/cccc4fffc1dd4fe8afad6d7221bae4aa : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b6240d2dadb942a18892bc74b5d4a82a : SUCCESS in 7m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7e90ee2c3d67484895e14bbc67bd547d : SUCCESS in 9m 39s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5b355cb321e342bebda218b27d7c33b5 : SUCCESS in 11m 32s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a6ae36199825445e904abdf20282c497 : SUCCESS in 4m 58s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b68201f6972c447c876dde82e1f964de : SUCCESS in 16m 01s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c4bc883932614a908fa0e571a0749ff7 : SUCCESS in 6m 08s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/52cf1a8762f1497b8a0f6f66cd4c7710 : SUCCESS in 29m 34s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7ad471f4a9b641c193fef2253747aef9 : SUCCESS in 12m 55s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ffcc329d4e9044d49e84579f443fb3e9 : SUCCESS in 7m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5a743f6d59664dce9d3da495e7a91fbb : SUCCESS in 5m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9459abe211c44be7b9c2338281a1c7b8 : SUCCESS in 21m 08s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/ddaaee589efd45379a2951b943b6ced1 : SUCCESS in 6m 55s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d2284199794a4da181da4f32aa6e41a4 : SUCCESS in 6m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f1f126d7f2a1491ebc95eacd6669b192 : SUCCESS in 8m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5a921affca9647838b3415511b4155f4 : TIMED_OUT in 41m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/135b94fdd494462c8753bc8968fa0ff3 : SUCCESS in 44m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/820ddb41a7da43d3ba4b4b1705069669 : SUCCESS in 1h 16m 45s","accounts_in_message":[],"_revision_number":1},{"id":"dde9f2d8540e7c44976ff1348670cf9ba4370639","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 00:31:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"43441ba876193dea99526e4873b6c186a3d06870","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 00:45:35.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3167d05ede744d89bed1947a23910409 : FAILURE in 11m 49s","accounts_in_message":[],"_revision_number":1},{"id":"4918341c618b879fe3a896e768002ba1536f18a7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 01:52:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1ed77247e40a5d7b997b450a315f0b112719f18c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 02:07:29.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4afe7ceefe2642ef866fc3f61a4d3a82 : FAILURE in 12m 25s","accounts_in_message":[],"_revision_number":1},{"id":"5ff2850492f3fd052a8ede577fa81851f63bdbed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 02:07:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa5f1404f911460dbe5b722215ad98bf : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5b5da6e689443e8aa9b88314ce83fe3 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fd89e3a64b847e194ccffa0d1ed8d7f : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed079bac30cf4e23ad790696f40289fa : SUCCESS in 5m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4b53befc889d4f49992b578ee5d3390a : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5bfdbee616b84457b4ab294a4843225b : SUCCESS in 3m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/702e4cc35547429d89e3727d3cbc5a5d : SUCCESS in 9m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/65a8415a3a6846bc899c1ff3d7f161ab : SUCCESS in 10m 40s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/741d68cdb47a4bfda75f4fe383e6223f : SUCCESS in 5m 37s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/dbb41be905984013998546275acebbfc : SUCCESS in 12m 37s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/c1b42c3ad2b841c1afe98d66275c812a : SUCCESS in 5m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/02744dcfa2924023968d0c1b16f4cdf3 : SUCCESS in 7m 54s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/177fe8c752f842ac8404c5523afe5340 : SUCCESS in 11m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/085bda7f05f94963b41af9c53cd6c488 : SUCCESS in 10m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c4c864377fc844889f1dfd732f093fc7 : SUCCESS in 7m 17s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1d632b55ba0740aeb786f561595ca9a4 : SUCCESS in 11m 33s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/326609b89e3948db9a9f42ebb4ddf4fd : SUCCESS in 5m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9a3daf58ad50448f9866087297c0b98f : SUCCESS in 16m 54s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5d54b4f458774d1597ca0d048d5b0476 : SUCCESS in 5m 18s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c9895fba58a54b08aa747ac40e8bd9eb : SUCCESS in 25m 08s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4205bb7b3fd24dbf80450e0b8f84702e : SUCCESS in 12m 53s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/95c670f06c244ecea1e3a382c2516cce : SUCCESS in 5m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/65b82686332641f4bf370f0ee7157d56 : SUCCESS in 4m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8806841025124d51a6cd86b1bb357662 : SUCCESS in 17m 55s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1743ccc18688417180fc1c04ec8b95f1 : SUCCESS in 8m 11s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7a98cf9c44be4236ae9f519b803a684f : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c055ab9389db4d4b8b611f7cce3e473c : SUCCESS in 6m 14s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ef85bb80c92c4af9a141c6d394179865 : SUCCESS in 11m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4571bd97976f4ca99943868a781cda79 : SUCCESS in 56m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/897020926fe646c5bc219b724dba19cb : SUCCESS in 1h 29m 39s","accounts_in_message":[],"_revision_number":1},{"id":"b9348c95618d9714af35d33b7a11ddee5aa46d32","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 02:18:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a7db4b70d98d7d801dc6d3b5853368f1b4720bf6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 02:20:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7f09c583dac666f222a2d6e50a21b3bd538057c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 04:00:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3af4969bbc4a4c85b0a124bcf29a77ff : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bab581dba99049f7928c8936d6a8a26a : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85d9cbbc4a464b298f8dfa0921ac3fe6 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0bfbf1e35a8d464e96dbbea1567a330d : SUCCESS in 5m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e400079cf97b4d0dbbd4d83f2e0c4763 : SUCCESS in 6m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/598afbb0ba2645a49ca3430761fcd1b7 : SUCCESS in 6m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f5c6fddaf2b04217b97c84be91d2595f : SUCCESS in 10m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1448eb9428db4ad1aec7ca771c3cd4b6 : SUCCESS in 12m 19s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2b7d673e6d3e4eb6b98247875b5655e9 : SUCCESS in 12m 37s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/11aafe38cba04f908b929ee29381d866 : SUCCESS in 6m 17s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/385bf4638cc34dc1a66d2337df1c48cf : SUCCESS in 7m 55s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ca393b3020c847e7a3aa66f464a50a43 : SUCCESS in 12m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9be1fde55e834e98b0d1a5a13de16a1a : SUCCESS in 10m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c9c1701d660349bd982987d32a987668 : SUCCESS in 9m 37s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/52b9b858a3204a47812661c66311d314 : SUCCESS in 15m 39s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a0059a6ae4044d18ab63ccebb68d9e27 : SUCCESS in 4m 58s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/af2518bf54bb439a9e32cbac2ae8f8d4 : SUCCESS in 20m 03s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7ede28b643c24e36b0df4cc363dd963e : SUCCESS in 8m 06s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d0f27fd34ce844b3a11b38fb84f43e41 : SUCCESS in 23m 59s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d78432c97b474619bf1178c4afdb074c : SUCCESS in 15m 04s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/6f1ea9de81584c218dc655933e936cc2 : SUCCESS in 7m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/281e4c33330f42b095134cfaeb9ece8c : SUCCESS in 6m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fd795f9b9c1542298ece1ef4e105f33e : SUCCESS in 21m 01s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f34e3ef80cca472eb22d9fcc9d78b372 : SUCCESS in 8m 35s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5dd31d7902624cccb036a720d44a5ca8 : SUCCESS in 6m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7645d120663e45f6a8fd52b6c09a7852 : SUCCESS in 8m 24s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/fd5fa5d218104ea0bf0c7423d9b04d98 : SUCCESS in 14m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38833dac6d814b9e9cccbe21411065aa : SUCCESS in 52m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ebdbefe941f14f74a578cafd49e35126 : SUCCESS in 1h 33m 48s","accounts_in_message":[],"_revision_number":1},{"id":"85c5cf872893eae8ee0b22a344213fbff95b79ae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 04:00:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"41ecc73a90c29ea68198b2d50c4aaed3f621a03a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 04:02:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2daff04c288e452085758df7561af992 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f2bbb378b3f9f6aaa6047c44557add23a704fde","revisions":{"4f2bbb378b3f9f6aaa6047c44557add23a704fde":{"kind":"REWORK","_number":1,"created":"2022-02-16 17:37:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/83/829583/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/829583/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/829583/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/829583/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/829583/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/829583/1"}}},"commit":{"parents":[{"commit":"0a7e9306d40715c6c737726da71e8670d85d004f","subject":"update constraint for oslo.db to new release 11.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a7e9306d40715c6c737726da71e8670d85d004f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-16 17:37:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-16 17:37:58.000000000","tz":0},"subject":"update constraint for oslo.middleware to new release 4.5.1","message":"update constraint for oslo.middleware to new release 4.5.1\n\nmeta: version: 4.5.1\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: I8863c46e91a8640e8ae070418a2e611b369f5aec\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I79a65e00d3cf29e9471d84521936f47fa1b90e01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f2bbb378b3f9f6aaa6047c44557add23a704fde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f2bbb378b3f9f6aaa6047c44557add23a704fde"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
