)]}'
{"id":"openstack%2Frequirements~829585","triplet_id":"openstack%2Frequirements~master~I4ad4dd01a783bb7090b0dff3187f1606afb6df2a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4ad4dd01a783bb7090b0dff3187f1606afb6df2a","subject":"update constraint for ceilometermiddleware to new release 2.4.1","status":"MERGED","created":"2022-02-16 17:40:21.000000000","updated":"2022-02-17 03:27:07.000000000","submitted":"2022-02-17 03:25:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829585-new-release","meta_rev_id":"585808ff9b0e044a0a8420f12c32633a873d7e2a","_number":829585,"virtual_id_number":829585,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-17 03:25:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-17 01:51:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-16 18:50:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-17 01:51:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-16 18:39:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-16 18:50:32.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-16 19:30:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-17 00:31:45.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-02-17 01:51:54.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"0cc3740dfb34853148f5097b009de781e0256693","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-16 17:40:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95623045b6af7f560d67929931e07f4b059f5cd0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 18:39:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e459d922972842499657cc384261be85 : FAILURE in 14m 32s","accounts_in_message":[],"_revision_number":1},{"id":"8a8841e5a42206ed645f66821890bc692479ada5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-16 18:50:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c93a90f8932b912ab9552252f0adcb07bae64740","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 19:30:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/661aef1c8cb34a95809b25d417e84f54 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a0328b45537496280cfdfe2e4f50f9d : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f9b7e6297c0480b9c701d89d4ea0ead : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f09e8a3dc29e4b7a998c7da9df3481ba : SUCCESS in 4m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/67c13d387be04905849b5955bd0fe7bb : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a0f07b5fb0544fa78f89452971fa05b9 : SUCCESS in 3m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b18afe2c888a4dbd87c48e07e2d7cf7e : FAILURE in 4m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ce1f1ec38a74459d83198516dd6066cd : SUCCESS in 10m 27s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ef3322765caa4915a837b33553d92a66 : SUCCESS in 5m 53s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/235a6156d9da46708ff54739531797fb : SUCCESS in 10m 18s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/918581eb4ae74d04b45db9fd36540307 : SUCCESS in 5m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3d471e75480043b99d01e95d25d969db : SUCCESS in 6m 04s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/24e73156eb33408f8c2577459f32f519 : SUCCESS in 10m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5a466c154b98400f83e39da78a3ba070 : SUCCESS in 8m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a80bbb83f6154c22ae5d6a03fab97e0a : SUCCESS in 9m 14s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/3c375fd86e544e26886608a309c37979 : SUCCESS in 11m 02s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/217e38a1d5b8413c9a75d2d698c9c5fa : SUCCESS in 4m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/933ee52aa6f245c0aed1d532404a73e6 : SUCCESS in 15m 12s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/87166794b8f647fcbc98676f1c53ef9d : SUCCESS in 4m 59s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b66b02f28f3342f492648470ab824fa7 : SUCCESS in 24m 16s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0fcaf9f54f8f4793bb7f52b35f51d619 : SUCCESS in 13m 28s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/fd49a908260d4303b36a7a6bd5281b20 : SUCCESS in 6m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/11f8e92ea680485a980cb0026fb42724 : SUCCESS in 3m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/764d58a9de2340e09560abe8c93366fe : SUCCESS in 22m 17s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/db1194dff6ca482b916295aa7a6c221e : SUCCESS in 6m 04s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7a264abc56fd4b2c8424385ea9213cdc : SUCCESS in 6m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f5d2566e013d4d789db4db29cc7e1e5b : SUCCESS in 7m 09s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3e00526b1045428bafd30079eef0292e : SUCCESS in 9m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1361bebe76a7402d8a76bde98939d1cc : SUCCESS in 46m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3b64a4d63522481996e40838804cedc1 : SUCCESS in 1h 39m 01s","accounts_in_message":[],"_revision_number":1},{"id":"217963a3eead197b9bc700ffa90bd41a28f35d88","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 00:31:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8906f7017b4ee4ef032b787d9cbc78781898adca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 00:46:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/511ac183387e4978b9fc6b9837233a2b : FAILURE in 8m 53s","accounts_in_message":[],"_revision_number":1},{"id":"def9898dc14fd11baa0590f626f346e20c7a50cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 01:50:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5707b9ade5f0455dab5889055986a5b2 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e3692987d444abe9aaba70bf168a523 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9f526a6ce7d48818b519926ce5b7638 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef5c6799fbe94213bde1158a32fcf554 : SUCCESS in 5m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8cfe1d6fe149438286bcc8245c01ee30 : SUCCESS in 5m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a9cb9ab4851d4945aca55595f1ba780a : SUCCESS in 4m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4388f33b62894b27bd4b172f89065396 : SUCCESS in 8m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/06524551ad2a4e20b742b6d71664eebc : SUCCESS in 11m 55s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/588d379a328046c1bc92a893690f1ad9 : SUCCESS in 5m 55s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ae0d4b721e5c41fc805a8bde0a9e5d0d : SUCCESS in 9m 28s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/75f4678c53c94db7a6148322e5774caa : SUCCESS in 6m 41s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d4e287e132a24e55ae56a795d961b299 : SUCCESS in 6m 19s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6f1936430aaa4cd1bb22efacd40a036c : SUCCESS in 9m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/049962917e8a4aeca1d6d67d74020625 : SUCCESS in 8m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cf81dafacebc4e78a9285dbc50ab6913 : SUCCESS in 8m 57s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e72776b7d1be42e4af7eb09ea0f947a0 : SUCCESS in 11m 58s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/0f37d6a6b52d4af89135cd1eba97e309 : SUCCESS in 4m 15s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/15207ccc40f04aad9c452480026f0dee : SUCCESS in 16m 14s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2977c7e10fbb451a899626594fe51d99 : SUCCESS in 5m 35s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d8ad31d636c643c5b83417b10df0f9c2 : SUCCESS in 23m 37s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/92eb5684317f48cba8f05ddcf1f2175e : SUCCESS in 12m 17s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/74de0d6167934fd1b7e93ae69dbc385d : SUCCESS in 5m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/75c8bbbb57014216a04a5baefa451a92 : SUCCESS in 3m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e0ed70222b9b43aa9c182fac8b15925e : SUCCESS in 22m 58s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d0bb95c56b3f4c7893f17505cf129e35 : SUCCESS in 6m 20s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/53d75ad8b477473c85be6f0a56714748 : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/719b90a63fc7478b9bb356177270c3ae : SUCCESS in 7m 48s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0dc309ff403045888c6506f38e998987 : SUCCESS in 11m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6edf919b4974032ace0275e24b52aaf : SUCCESS in 48m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e8da33215a8c4e4d854c9af172722510 : SUCCESS in 1h 16m 35s","accounts_in_message":[],"_revision_number":1},{"id":"271576ca6cecf3e717008402bcec4c8d375b2c0b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 01:51:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"106a5db4d4802ae83a0e5e4bc1cd548af3a0ade7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 01:52:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"61a395986c1c225b7e1b2e570d7008f85afd7c37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 03:25:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92f4ad6a75a94b25babe38796ddbd2d5 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f535919749c489a8902f25404484ce5 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a359bc1e35184c0ba15f288a5d2dd2e5 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bdfd8279fdbf4fdeb8d03ef9df5f625c : SUCCESS in 4m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/87ef30304715454ba92c41eba25dc712 : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2eb3f9f2f46f40af8b4a7efecadf19f9 : SUCCESS in 3m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b81bed6e09ca4163bad4a307de92f37c : SUCCESS in 8m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/22e4bdf02df04129a2201007587b563c : SUCCESS in 9m 58s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/af36582767e8472c8adc95f3ac679769 : SUCCESS in 9m 34s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/4e61c78c10a349dc9b5dd0dbf09a0862 : SUCCESS in 5m 01s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e7f14b24c57a4656a6a6edf5d89c8546 : SUCCESS in 5m 49s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/cb234a89f0b54ce697032da977e140d7 : SUCCESS in 10m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1e4f90c735774ad99e9e4708073313d2 : SUCCESS in 8m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/849531b08074438fb86529d3e5c3fc39 : SUCCESS in 8m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1921728201df4a2b924bbfc1f0427b04 : SUCCESS in 11m 36s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/4cecdba5a2d44dec9e2b43b631682a1f : SUCCESS in 4m 20s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7851c77a498c4b97bdeb99b619da7ae1 : SUCCESS in 24m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/29fb266d12fb416eb9ca2b3dbf63f7f7 : SUCCESS in 5m 22s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/846b3e03222a4bd6baf85ddce5f4675b : SUCCESS in 27m 01s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b65f770a005048ecba3e56791bfc2f7b : SUCCESS in 11m 20s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e0051d68d5644bb5a1f67ae7fcd2fc56 : SUCCESS in 6m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1474dd938b5f4ecca94779a54605af27 : SUCCESS in 4m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8a4a0da9b9d14f5cb90c85ee39ef0b3f : SUCCESS in 19m 50s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/22180c6782da4acea1db11abcc017dd9 : SUCCESS in 7m 18s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/335a63365fe84a308aa2390893f32069 : SUCCESS in 6m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f75fb12c13624f76b67e691414e01acd : SUCCESS in 6m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/34bdd099416647f9a74be83c5f090fa5 : SUCCESS in 10m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e89b2a6cae3d4dddb488096bbc392543 : SUCCESS in 45m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/48fa6ed4c5244192976d011b5035ea72 : SUCCESS in 1h 30m 26s","accounts_in_message":[],"_revision_number":1},{"id":"5df96e7b6191694607b663ec0783ba44c5011f92","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 03:25:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"585808ff9b0e044a0a8420f12c32633a873d7e2a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 03:27:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2198bf09dea34c5a831243ef656cb8e4 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1a5372f81273a39957f9592c99218779620680c3","revisions":{"1a5372f81273a39957f9592c99218779620680c3":{"kind":"REWORK","_number":1,"created":"2022-02-16 17:40:21.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/85/829585/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/829585/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/829585/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/829585/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/829585/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/829585/1"}}},"commit":{"parents":[{"commit":"0a7e9306d40715c6c737726da71e8670d85d004f","subject":"update constraint for oslo.db to new release 11.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a7e9306d40715c6c737726da71e8670d85d004f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-16 17:40:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-16 17:40:20.000000000","tz":0},"subject":"update constraint for ceilometermiddleware to new release 2.4.1","message":"update constraint for ceilometermiddleware to new release 2.4.1\n\nmeta: version: 2.4.1\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: I70d380bb3ee7ad8720869566579c6a51b1b22474\nmeta: release:Code-Review+1: Matthias Runge \u003cmrunge@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I4ad4dd01a783bb7090b0dff3187f1606afb6df2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a5372f81273a39957f9592c99218779620680c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a5372f81273a39957f9592c99218779620680c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
