)]}'
{"id":"openstack%2Frequirements~829624","triplet_id":"openstack%2Frequirements~master~I31e55c9cc5b8f8cc741d8f2e357db73d85ed3001","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I31e55c9cc5b8f8cc741d8f2e357db73d85ed3001","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2022-02-17 02:41:33.000000000","updated":"2022-02-17 09:01:13.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8b6c350e657c43f6b3853eae8abbcb4c0df8c636","_number":829624,"virtual_id_number":829624,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-02-17 04:14:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-17 04:14:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"db4153bc9539c8d569156c12025f2f36956390d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-17 02:41:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0f0c2d35ea06bff79d7657cb6127d14bc377a2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 04:14:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/093b6df749a8453ebc7d863cc93c1441 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f75b368378540e696873b29f68eb38b : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88399fe21e2a4f3491f1ca7ee33493ef : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4dd7c913335d426a8686873431f7c8d6 : SUCCESS in 6m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9b178a87f8404590a1c7227a187feb49 : FAILURE in 6m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8e82f0ce96fb4ffd8f83d16fabd4c52d : SUCCESS in 5m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/de2ca0f044984a3f8256828057af082b : FAILURE in 6m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8618078fa51d4b0788f0357288574309 : FAILURE in 4m 43s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/edaab7a1af964a9aa2b56698115088b7 : SUCCESS in 7m 05s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ab45961b5c10423fb3638734e334a2da : SUCCESS in 12m 21s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/36b39475b8e34a77996079ddb7ce3202 : FAILURE in 5m 49s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f537d48c2a8e48c9a769e40d5ab57daa : SUCCESS in 7m 23s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ce43b45ebef048868f06ce0d943841ad : SUCCESS in 12m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5ce0ea1365704f6dbe01dd522dc4206c : FAILURE in 8m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ee78f8aecc4242e59e7aaf82987a82ae : SUCCESS in 9m 27s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4ad9f5d1899d4c89a90ef909ffddc380 : SUCCESS in 12m 43s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e74a86bdf9e54f7b997c4c475438ce37 : SUCCESS in 6m 11s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2663a7ef79d14c42b97d554f7bf453d2 : SUCCESS in 17m 36s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/55cfef92a58045a4927faf4fd2e2e212 : FAILURE in 5m 43s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4abd3183664e4082b8d87c3b56eb825c : SUCCESS in 28m 27s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/94ac64c8ad914498a89a3c3b448a4099 : FAILURE in 12m 36s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/02fa92f9840e4b5299b43dbcf3d088fb : SUCCESS in 7m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0b81b71cb9b946c4ba67e9601b737d6c : SUCCESS in 5m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/23cecc5661584e488fd0fbc70fc3cf95 : SUCCESS in 26m 13s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d72b2358a50e4ca395c361d2e72c2ed2 : SUCCESS in 7m 24s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9f24f9841e1c4d75a758e1579d615142 : FAILURE in 6m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/69de9d9ebde94f8fa0a4e4645cade8c4 : SUCCESS in 7m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a14b2003acac41efa824a0d8c26694a2 : SUCCESS in 13m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07404f93cf404dc2b533f2d410aab868 : SUCCESS in 45m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1eb60f15c6124682abc9acdcac4d54e2 : FAILURE in 1h 17m 59s","accounts_in_message":[],"_revision_number":1},{"id":"44132b91c383491d18357f7dad7a3a4c4fb65c4c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 04:32:09.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5d7bba9f8fb842a1bfea394b9d06868b : FAILURE in 9m 42s","accounts_in_message":[],"_revision_number":1},{"id":"8b6c350e657c43f6b3853eae8abbcb4c0df8c636","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 09:01:13.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3a253e5a53386c118432cbe33ff9d8d2528c5188","revisions":{"3a253e5a53386c118432cbe33ff9d8d2528c5188":{"kind":"REWORK","_number":1,"created":"2022-02-17 02:41:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/24/829624/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/24/829624/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/829624/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/829624/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/829624/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/24/829624/1"}}},"commit":{"parents":[{"commit":"3dbf17b3b2334f39a8f9f5bb499a703396ec264d","subject":"Merge \"update constraint for oslo.serialization to new release 4.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3dbf17b3b2334f39a8f9f5bb499a703396ec264d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-17 02:41:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-17 02:41:32.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I31e55c9cc5b8f8cc741d8f2e357db73d85ed3001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3a253e5a53386c118432cbe33ff9d8d2528c5188"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3a253e5a53386c118432cbe33ff9d8d2528c5188"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
