)]}'
{"id":"openstack%2Frequirements~829669","triplet_id":"openstack%2Frequirements~master~I9896c9643590c5ed3fea73cccf7f5b14171e17a5","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I9896c9643590c5ed3fea73cccf7f5b14171e17a5","subject":"update constraint for os-win to new release 5.6.0","status":"MERGED","created":"2022-02-17 10:33:40.000000000","updated":"2022-02-18 05:55:55.000000000","submitted":"2022-02-18 05:54:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829669-new-release","meta_rev_id":"16c9b300266b1f1d142b13c8ecebadc558263322","_number":829669,"virtual_id_number":829669,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-18 05:54:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-17 14:51:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-17 14:51:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-17 11:03:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-17 12:10:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-17 14:51:06.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"a6efebbeb88983e5bb9f73fa287cb6959c45c8df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-17 10:33:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1730e6ff686b2587ff435ff5efb7d8091fde7e0b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 11:03:13.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/948d460d8186462187c0d1ca16fc8498 : FAILURE in 15m 52s","accounts_in_message":[],"_revision_number":1},{"id":"5d16b6c3307ba84a3f2b0389d67e6cbf5c7196ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 12:10:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41d2e2c41cae4f0a89890cb7e3f2fe9b : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b55836c0ef9943b8abf59fd9fb5f56b1 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f632d6ed57564eb1911837e0797e6285 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2756caf2a6264d2a86904cd5fa7f5d91 : SUCCESS in 5m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e4d68adabd4445eca51d6408eb113b76 : SUCCESS in 5m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5789bc38cc55432995f0e8ff0e65cdc6 : SUCCESS in 5m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b521862b57944365b09e98de159984bf : SUCCESS in 13m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f0385e5d5cac40a782453b7a0fe8b0c9 : SUCCESS in 12m 17s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/1a0af2ff4ed1499e87d3e28d157abe9a : SUCCESS in 7m 34s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/203c3c37ca404ede9acc5ea282f5480c : SUCCESS in 11m 50s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a0d5034fc91640b0b74b509391624e35 : SUCCESS in 5m 19s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9db2a8649c524c3fa1ba5c2c85a58fad : SUCCESS in 6m 47s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0afff2ac8e2c43f8b607cea838c6d97f : SUCCESS in 10m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7b9363bf645040cca4bc89f71053cc86 : SUCCESS in 8m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5700b00993864a47be0514870308b085 : SUCCESS in 9m 19s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/dee8e27dbc6c411e957fb28d89c73987 : SUCCESS in 10m 18s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/771bcb3be47c4a0a859833ad765eebc6 : SUCCESS in 3m 55s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9600adf74b2842e0817474cd59beb219 : SUCCESS in 17m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/335a44623aad40978976d8db8bbbd8b9 : SUCCESS in 5m 23s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4ec1ce130d83487ba065722dce92871e : SUCCESS in 36m 01s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3e49213c91e746a2b3eb822b0f17f86c : SUCCESS in 16m 05s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0eb0eed896df4a44bb20ce32309568d3 : SUCCESS in 5m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ecc00b62aa9044a68b0ba63e5407c86c : SUCCESS in 4m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0768c05caecf411fb67c899124e564f4 : SUCCESS in 25m 00s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/11ab1a1d458c4982b53a4a4e63b319d7 : SUCCESS in 6m 17s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3dad8179c8db41299bd0c7231ef8152e : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0430ff4c388942c8a048982e6f249077 : SUCCESS in 7m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a9031fdda0394007a6fb3df45ab52449 : SUCCESS in 11m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d38edc6d94184815904ab256197b9123 : SUCCESS in 1h 03m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8892aa6829d4fcda3f8a73659eee6ec : SUCCESS in 1h 29m 19s","accounts_in_message":[],"_revision_number":1},{"id":"7f02ea604c2f2088db4868537367b13bd53625f0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 14:51:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"450bc89256f6e5d52d94507e251ea69d79929149","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 14:52:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"637f131db312e4e23ac7df268fd3f9c14e74b4b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 16:21:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f91a52abedcc4bfb957079341bd701e1 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/467d451f0b5b4f22ac03a3fdff93c4fd : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fdac093e043459d87f768c1b1957227 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0cb7a3bd3d04f5690e6766806344296 : SUCCESS in 6m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c266a3cb23846d19838298701a36d85 : SUCCESS in 5m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f92849c89f664b7bba7d27c3c59ecb55 : SUCCESS in 5m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3ae5ba8ea19a482ea2b80c0f60f8cfda : SUCCESS in 10m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ff2f72140fc842cd8fc3545531d08158 : SUCCESS in 15m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/72ae1269554a42fe87c5ab90bfe5d37c : SUCCESS in 10m 25s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/dbe6e0e90ca34ea18e7a94fc34d99664 : SUCCESS in 5m 34s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/406bf56053de475f99db90c3d33a2edc : SUCCESS in 5m 59s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8dc568388abd44b0b7ccaa16e0994320 : SUCCESS in 12m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6da3edc402c84a3ca6b2ceae325c81c1 : SUCCESS in 9m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cf42dc83d3544626aa655900591d9e3f : SUCCESS in 9m 59s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/33c2d961eb9b4f95a4acc4d1e85dfd45 : SUCCESS in 12m 41s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ad0686f248b14044a066b372107905f6 : SUCCESS in 4m 40s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d8672e5bbf554fbcbf14722c6e7f3782 : SUCCESS in 18m 34s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/69d189d663854c3ca99624ce75f1c561 : SUCCESS in 6m 00s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e1fd6fcee46c428ba5aa1f0c0a51b9b3 : SUCCESS in 25m 13s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/27c369e89ef943b084c3702fec291181 : SUCCESS in 12m 54s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/52b0552a9ed742939305c154aa0a50cb : SUCCESS in 6m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8505622e1ff747b28f68be854736cfc8 : SUCCESS in 4m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f4422dbc2c62403abd6f1384aa4de47b : SUCCESS in 21m 00s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/55cf5b679c0b4a628caa639c437b63a0 : SUCCESS in 6m 37s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6ff9133c213f4e0890d5714350b76629 : SUCCESS in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f5aef66a515c4986a3f63d5c4dd169bb : SUCCESS in 7m 01s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/227c6f8e0fb54243adb6a7d7f136f6a9 : TIMED_OUT in 41m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/09fa321afd374ec0b5c3ddfa5942bfb9 : SUCCESS in 54m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/09c8ff2425b64b56a09494714a90ea2c : SUCCESS in 1h 25m 43s","accounts_in_message":[],"_revision_number":1},{"id":"6425f2ebb9cc9807509f095e26b1ff0aa5368795","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 17:32:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"06b8783f089538b631f4fd82746e97bc30993f5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 18:30:33.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9c9598af42c1441793ebcf0e3f52ad42 : FAILURE in 14m 13s","accounts_in_message":[],"_revision_number":1},{"id":"a539d4477fd51e9998e5228a7389ece7ed1a1b3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 18:56:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0008694c081f4dfba753c1226dd78c11 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/938dd320c5dd4461a636bc855d392f89 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40cfa8edb2f74bc1b5261677e9509812 : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71433ec7e1904118936cf299d93f3dd5 : SUCCESS in 5m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2d64ef6d3e0642b5a9905e9ea3ba98aa : SUCCESS in 5m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/065ee85d667549a2b160afa26c5c5d88 : SUCCESS in 5m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c5c0c632f31a4bc9a351b6df58e772b9 : SUCCESS in 10m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0b972e889784445e8653ac3315ee42d3 : SUCCESS in 10m 37s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a3a889a8be52467c8c5f57072f29398e : SUCCESS in 6m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/72d2415b9e8542a4b176f7577728c26b : SUCCESS in 10m 57s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a442f69b452c455796fa29ddf427fd90 : SUCCESS in 6m 47s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f39fe663699941ee9e7aab5fb129498f : SUCCESS in 7m 01s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/97f8571a92af4fb6a254ab2818cad1ad : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e479a17dd3ae41c388cf87cd64cba642 : SUCCESS in 8m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/dc832caacf30430fbe2aa812aed50141 : SUCCESS in 9m 19s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8871e1b0d55f47d4ae3e95f1481e9781 : SUCCESS in 12m 17s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e8e9b967f0734c618779d393bffa134f : SUCCESS in 5m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0fe204c6ac4f415ba35398df00eaccef : SUCCESS in 17m 18s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6642b016e8aa44cb9113d4b896b47029 : SUCCESS in 6m 54s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e4887c6795c24760bb3867afea5fdca5 : SUCCESS in 26m 10s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/62382b6ee6bd475f850c3e016ed834b8 : SUCCESS in 13m 40s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/8165b5daf95b4d72a523b99a76a8cff5 : SUCCESS in 6m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/18832d8aa65140f2bfcee32145d20af2 : SUCCESS in 5m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f6dabaec37c3492d8992b2ac4470905d : SUCCESS in 18m 46s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4e2890f566df4f2988539f5cf397af9b : SUCCESS in 7m 59s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/93f30f0ee740497caedf812ab1680539 : SUCCESS in 6m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/929544c056f14241aa59141e7e4e2ac4 : SUCCESS in 8m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/87e95623ecb741ab9d10029df9069a82 : TIMED_OUT in 41m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e269ba51227462381a55ec4855db469 : SUCCESS in 47m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5c33ef83f8ec4e68a98530db528b82e1 : SUCCESS in 1h 18m 58s","accounts_in_message":[],"_revision_number":1},{"id":"30a8366e13b1c5ecb718e559e8cc585d191efd71","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 19:55:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f9796fef4331c0f0be2d3551268be34430a93899","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 20:20:36.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/087314a6112f4b399634720e4c8d80d7 : FAILURE in 16m 19s","accounts_in_message":[],"_revision_number":1},{"id":"cf57c9c4110a3807c42d919a268a524907d00653","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 21:34:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b64ba95760484c6f9ad3984e310b8765 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15ea26421365409e8cc801363525f84d : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2604db79cb4a46dcaf8ad5a1a44888d0 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79bc3368d6a447909d670e56771de854 : SUCCESS in 5m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0f379dbd12d04b3fa83139afe26e20c0 : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f8f132f20f364b518b3c86fec58eabac : SUCCESS in 4m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3776c05506594e4c9e331f02818e4f0f : SUCCESS in 10m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e0d4c171c1984f35ae86bb5b927a67c0 : SUCCESS in 9m 22s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/3ea06a10648446cfa91e65ab10301532 : SUCCESS in 6m 31s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e3f795dcb77c4463a8a2e2971829d169 : SUCCESS in 9m 56s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/4305f96db6b6445b87421a59ebe0443c : SUCCESS in 5m 14s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f722f681c21e4620b56cb078a7f1c94b : SUCCESS in 5m 57s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/54b099fc97d3408e97acb6fd56c4bfe4 : SUCCESS in 11m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f68944c497fe49d490ef1dcaeda849b0 : SUCCESS in 9m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c0da8a91ef7c429fbb015e3c4df84be3 : SUCCESS in 10m 29s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/822fd746442a45f28d9b5036ae1de4a6 : SUCCESS in 12m 53s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e3b48b2750544771af74339e46b072cb : SUCCESS in 4m 12s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e10a289a134341bcac3bbddbb2b83b40 : SUCCESS in 18m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/da0ab8959c9c43e0990406d57ea6860d : SUCCESS in 5m 35s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9e9a7f6da4654b2c85a68ef3bf6a2f34 : SUCCESS in 31m 28s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1e94377758c84deeb40f6207b0fec9f0 : SUCCESS in 11m 37s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/5d93692f262a4de0ae285db5bcd45459 : SUCCESS in 5m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1468c65144b7453e8ca9fc37b61d28b6 : SUCCESS in 3m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3864b95684b14347bd468dee6586beee : SUCCESS in 23m 00s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a5bf57371a104c83a4c2e67a177a776f : SUCCESS in 5m 53s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7215efae1a434d55a9e49d34e713e04f : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c6f841911ef14a9dbb49009dc91219e7 : SUCCESS in 6m 41s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5f72f3d23ef046e380b183746e9d2bed : SUCCESS in 13m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dde0cb913ed54739a6b06bf684874b05 : SUCCESS in 53m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f5e264158bea49b385700c48743f4366 : SUCCESS in 1h 33m 04s","accounts_in_message":[],"_revision_number":1},{"id":"a608314eb7d243c8d977562c4cfc1bb35d491009","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 21:35:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6e7978b974e3e4ca15699ddee4428ea8f13bd302","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 23:01:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dc95f73ff0444bcb7caca30f5d32e3a : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47ef876f6d7b40808a94c79497322ec8 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e27a4684f4c045518619d7ae7079708a : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a56ae4d4873748a1891851322255b056 : SUCCESS in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b4ea724f7c21459ea32f647cb9909a34 : SUCCESS in 5m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0c7e9f3e750f40cbab2c5871505cba95 : SUCCESS in 4m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cff5c49419554df8961bbd083829fccc : SUCCESS in 10m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a4742b385d9c40548cdf2b0e8ef48e3b : SUCCESS in 8m 48s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3e5ebb146c4241e5867620fe0ba387e9 : SUCCESS in 10m 03s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/d5c0c7709a164e7895dd76deed7fe0c8 : SUCCESS in 6m 19s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/20717214129f46c3aab760c35fdd1c78 : SUCCESS in 6m 04s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c14ea0cb171a4922aab16c0bb6385dad : SUCCESS in 10m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4ba2ca6a05574fed82a35df35079824f : SUCCESS in 7m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/84f7c30da2ee406cadf7f75b012c568a : SUCCESS in 8m 46s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5cec31d4c550452a9890c6d0f75cf6c8 : SUCCESS in 11m 15s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/bc764707f1b745dcb03fa9f8f1a77fd2 : SUCCESS in 4m 48s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/96b1a1df8b134ea8b99b78611cadb1c5 : SUCCESS in 18m 00s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a18ec29ec0954fdabf1e066225c38ae6 : SUCCESS in 5m 37s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c734371de11940ff9c3c6cc677da9223 : SUCCESS in 25m 12s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7eca0eec14774d07b8697ad030f0d4a7 : SUCCESS in 12m 07s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/8e6176fe09a84866b0064f9372f4763e : SUCCESS in 5m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ad62db3fc8e54139bcfc7605b370038a : SUCCESS in 3m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4c1f571b865341a5a279453b528ce9f2 : SUCCESS in 20m 18s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1bd3b1d5a1ae443aa4747e591fcd5af3 : SUCCESS in 6m 07s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2b290bd2cd054e1b920c8f1b857faf8f : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/290cf427eb504d79b6cdd443035ab486 : SUCCESS in 7m 01s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0e59859e63834427aca34a75657aeb23 : TIMED_OUT in 41m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5cf4cb59aee1495b868260b3a221f426 : SUCCESS in 48m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae75caa49061413f8df1cf4e25f5c5dd : SUCCESS in 1h 21m 19s","accounts_in_message":[],"_revision_number":1},{"id":"3a670638f3171f7d63ab13c99b172967ff54b79a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 23:11:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c99a2bab872d04caaf03b91f570ab3b46c2f37b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 23:27:45.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1abb07f1b8934228b0130a4821bf0dca : FAILURE in 13m 42s","accounts_in_message":[],"_revision_number":1},{"id":"9fdbfc527359f1590be598d45dbd2021ecd13844","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 01:02:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1379ffcafca84aebb03e605845228ea1 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/418fadeb424446e59e693e6af0b80849 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/474f3fcf5fde43a3bdd68e4abebd7047 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81cf8a12f79240aea482d9ee64e258b8 : SUCCESS in 6m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e1fd1ce3bbb748f98c175efceceb26e5 : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/064a585e398945b8a0eea92bbfd216d2 : SUCCESS in 3m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/07d8ee562a494781bba9bfb848b65b4c : SUCCESS in 8m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e2772897e7514d2bae6d282f2c2d7de0 : SUCCESS in 7m 35s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/eccf42a4ee354dd08936b8e2eb21938c : SUCCESS in 5m 24s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/230f71a797e84601a96d823e8f38bd2f : SUCCESS in 9m 47s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/af08c90045c14aba9cb8ccea5580e909 : SUCCESS in 5m 17s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/53204c049f6c41d89080630281763b00 : SUCCESS in 7m 51s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/39ee28d89b04429394fe298eb55ac2d8 : SUCCESS in 9m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2a95e41ed5064450a9af8e057d219ca5 : SUCCESS in 9m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/12f6ac6fda764cabb5f89e58305892d2 : SUCCESS in 6m 41s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/fa667a5310f949a1ae6f1c7d40bbc7e8 : SUCCESS in 12m 38s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/193aecd85f3541d283b358ccaa42667a : SUCCESS in 4m 55s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8b1bd6c030b940e7855bc14c0bacde54 : SUCCESS in 18m 57s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ee9ea949c9804fb1ac0cd90d59f9b74b : SUCCESS in 5m 24s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/50378fe1732b4238bf467ebdb3ea58a7 : SUCCESS in 23m 12s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e34a5a82b6e54ebf96dd25ceb4b77144 : SUCCESS in 13m 05s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ccb8359ac19e45af8e76f351cdf7ec71 : SUCCESS in 7m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b99554e8fda144e896f05024f35f3d15 : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e589f8d33bb541549c7a0e95bcdcbf84 : SUCCESS in 25m 11s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d1064d633e53484bb38a198b151601cd : SUCCESS in 6m 06s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/78c31bfa909e46dba5a43c6544103edc : SUCCESS in 5m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/842b98dad3b44384b4116ebd92ab51b0 : SUCCESS in 6m 47s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/dc3af0cb58194eb491dcd1505f655865 : SUCCESS in 12m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5fce0cbecbef471584e0a2ebe9a6aa14 : SUCCESS in 48m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dda2f0eef1d140e5827439c50263a4bb : SUCCESS in 1h 45m 09s","accounts_in_message":[],"_revision_number":1},{"id":"9a48da92ff4c6829c4de3aac453fd8fadec17bf2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 01:03:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b04a60141bb9d751a12fa24fc149468bb43f9ee2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 02:38:59.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec6d3e8aa6264d91a20353cbacbdaacf : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/decedfa4e590409985019182e9fab92b : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aede7991d7604390aaa16a29933cded1 : SUCCESS in 8m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad358cb4ee1747be943d0a3a72ed86df : SUCCESS in 7m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/759d6028a03c4d8bbf8693012e418930 : SUCCESS in 6m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f7cb02f056c04af6a2da83aaafe7e2dc : SUCCESS in 4m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a4b91cef811548b691ee1610918e072f : SUCCESS in 14m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cf3f916e5a094ac9aaaf96f23776ec03 : SUCCESS in 8m 36s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6f265dbdade94e16976ef0cfa7a5c41d : SUCCESS in 9m 12s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/850b832715ea4a1fb0eaf8dcf9732ff2 : SUCCESS in 6m 55s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0c01d9f347d6449ab22eea45d1317734 : SUCCESS in 6m 15s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6d7b0327d6cb4079852d7dd9cdc4f729 : SUCCESS in 10m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fc96223a345c4583982eca05865b8602 : SUCCESS in 8m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d8b443229e1a4e74bc4a20a5845df080 : SUCCESS in 9m 22s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ee64232b0791425690e4ec932c923da6 : SUCCESS in 11m 14s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/50c8988f699c46a29fb22fc8c8561543 : SUCCESS in 4m 40s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/79fd7a86e54341828eff0deca4684c3b : SUCCESS in 15m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2a583482dd324db6a95f49b3fa46003e : SUCCESS in 5m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6f3b6fcf6c9a40d29d7f85e5edc6c9b1 : SUCCESS in 21m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a2297165f4a04931a877b1536e061895 : SUCCESS in 15m 16s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/63e03c345e7449f499c50f576f693bde : SUCCESS in 6m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b0f86b6232244cb2b4e5c1873d4d2be6 : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/89e622d26c044fd095917046fde9bb09 : SUCCESS in 21m 48s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a7e37b58b8544dabb37de45ec8d56c27 : SUCCESS in 7m 06s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/288c9790738e4f88862a497215d4be3a : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/36206f02d41e44e6917f08cd7273f789 : SUCCESS in 6m 54s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d208444b85694fa69a48462c2357ce54 : TIMED_OUT in 41m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d9497f2b4df14e00a9fb2a258fd816e4 : SUCCESS in 34m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/19a3c92c060b42eea30c882e8fc3d1e8 : SUCCESS in 1h 30m 51s","accounts_in_message":[],"_revision_number":1},{"id":"25db1c29103368d7f1fcf171ea50f0b964aa06ce","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-18 03:01:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e99ea5eb482b9e42fd8778687a2124f33d80839c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 03:22:11.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/286a95adc0ae49de806e4561dcfc0140 : FAILURE in 17m 39s","accounts_in_message":[],"_revision_number":1},{"id":"5dbb24cc4d67d6fa4ef4288df53f067b020d30fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 04:21:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13084f3f2b9442808ef07fc9521a653d : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c6824c6ab2f46f9ad0ae9d5bf90e677 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6371b19972d64d408f00d9e8f6ad42b0 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f31746305a447bab65bbbc0ce4b211e : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4f3163c596d24e3496ccc679b262410a : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0552e25a79d64e91bb222ede3b9e6655 : SUCCESS in 4m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a38132bdcaa443d0905a3154a1b03153 : SUCCESS in 8m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/235b341c8d35465c9b940105d1402013 : SUCCESS in 8m 50s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/6c67a655c9c44d87bc64d266e5cbe25c : SUCCESS in 7m 15s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7de7ed1945a74dbe8e501c36bd66b749 : SUCCESS in 8m 28s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/4a8a988f5da242b895c661273102bb84 : SUCCESS in 5m 13s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d5e78bd79fc544539a02e0557978f536 : SUCCESS in 7m 48s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/118a1543db124b67acf0200149d61927 : SUCCESS in 9m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5686f18e0b864750a0081cb2e45be538 : SUCCESS in 8m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e51f842924ea498e831d6189b74934c4 : SUCCESS in 7m 59s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/642e0552f9b348c98360ff074d2b2e8c : SUCCESS in 15m 01s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/cfe8fe9d61154113852ccab2777e2696 : SUCCESS in 4m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/32e2c3140abe40cd978e95045a2bd588 : SUCCESS in 20m 05s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1c1a3e95d03746cab5a7d5c96af1d232 : SUCCESS in 5m 55s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e224c7da03be49f2886e959c563b481e : SUCCESS in 21m 40s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8753112276a344cdb2690862ae6ea420 : SUCCESS in 14m 21s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/fb79474f9a174b1096f72d7d7dd0e8bf : SUCCESS in 5m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/29297fb6f24748cc9d7d9cba889f7cd6 : SUCCESS in 3m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3537778b99d24860b8e6596c6c370121 : SUCCESS in 18m 37s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/49dd9caaf62a46c4b0877c609ad28731 : SUCCESS in 7m 04s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1f2f61a37525443e979e0b0cab27eb42 : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/35896e17812545549d7af87f6050fc8f : SUCCESS in 6m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/78fa6c52b61c419fb7e64cde46cf7d7b : SUCCESS in 12m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc0f7914a8b64e8cb6102b0ab08300bb : SUCCESS in 39m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/90aeb447533a49bf88e2004931d75e81 : SUCCESS in 1h 18m 02s","accounts_in_message":[],"_revision_number":1},{"id":"fdb849644d54e89c0bbed613ff9521f4e849799e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 04:22:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1bb442eaf5e02cfe57c6ece626523b677f584829","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 05:54:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf76e7e3901a4bd0ba41fb6320495b20 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33ea0dbbdafa4c7fb560e908edaca896 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/787c54ce775c4d2f835009879c02d209 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eefbcc170ebc4ec0be79525e8fb40515 : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/79509271bc1041ae970cd89aba03fe74 : SUCCESS in 4m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7c8001aa23ea48b4841d07155a00d891 : SUCCESS in 3m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cc2504907cfe45879e0bb7f08ce4cf1c : SUCCESS in 10m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8027fdb2b8864d48b852e682fe576e41 : SUCCESS in 9m 27s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/bebebe419d234244b41c5775dca1d252 : SUCCESS in 11m 11s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/592015dfea0843f991ef724f584fc3a0 : SUCCESS in 4m 51s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c367dbfdd133447da7c6197ede91140a : SUCCESS in 6m 16s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3650dcc19b8b4f6caefdc1bd5167b20d : SUCCESS in 11m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7d460bd598d84872bf402a9943d83152 : SUCCESS in 9m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f33d03af072548999f69273e13a9a37e : SUCCESS in 9m 06s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a5d0f7755579473ca1758eb7e7a38494 : SUCCESS in 14m 41s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/486fa659e48f48d493f0a16fad268b9c : SUCCESS in 4m 21s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9ad4b741156f436d89188f7831279113 : SUCCESS in 15m 52s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e4c3ca07850f495b99f3e5488a72de59 : SUCCESS in 5m 06s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/628924d0cd5e4378a36d91e8f639da1d : SUCCESS in 24m 28s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/cda4cbad7c7946a18397c4cda573a5b6 : SUCCESS in 13m 11s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/332158314701428cae0c87078ff8b1ef : SUCCESS in 5m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/aa8a074f796b4961a833402978c2cee5 : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/34838f541a6e4031857b869365e34ea9 : SUCCESS in 19m 58s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/ab8f4df49d4942dd80567afb4537608b : SUCCESS in 5m 58s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/dbdc47c39fc1435a9f2bbc87ba8259b0 : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9afc6e87fa044e8aa3cbff82ad707a62 : SUCCESS in 6m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6fa71cd6958249a78fbd3e763069e67a : SUCCESS in 12m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba3a3af388824727b1ca866795718d4e : SUCCESS in 52m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bcadb7ebb27d47c1a6a575e531748fcf : SUCCESS in 1h 28m 17s","accounts_in_message":[],"_revision_number":1},{"id":"876af8a9e74f5cd500e9fdfc19b9cd3b75d80fbc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 05:54:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"16c9b300266b1f1d142b13c8ecebadc558263322","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 05:55:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b94cce348bcb4550bf05e7004925a8f9 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"090552276de514767d199b44a2d2f10cb082d95a","revisions":{"090552276de514767d199b44a2d2f10cb082d95a":{"kind":"REWORK","_number":1,"created":"2022-02-17 10:33:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/69/829669/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/69/829669/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/829669/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/829669/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/829669/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/69/829669/1"}}},"commit":{"parents":[{"commit":"0e24c72df39927e1dd84d33b3e27434c15a90f81","subject":"Merge \"update constraint for oslo.middleware to new release 4.5.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e24c72df39927e1dd84d33b3e27434c15a90f81"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-17 10:33:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-17 10:33:39.000000000","tz":0},"subject":"update constraint for os-win to new release 5.6.0","message":"update constraint for os-win to new release 5.6.0\n\nmeta: version: 5.6.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: I72bb9c7ff04be6bbf74c80636b1b08881561fb12\nmeta: release:Code-Review+1: Lucian Petrut \u003clpetrut@cloudbasesolutions.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I9896c9643590c5ed3fea73cccf7f5b14171e17a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/090552276de514767d199b44a2d2f10cb082d95a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/090552276de514767d199b44a2d2f10cb082d95a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
