)]}'
{"id":"openstack%2Frequirements~829670","triplet_id":"openstack%2Frequirements~master~If0e53f7603f5601e0931b4c2200988c2f0f51e87","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If0e53f7603f5601e0931b4c2200988c2f0f51e87","subject":"update constraint for neutron-lib to new release 2.20.0","status":"MERGED","created":"2022-02-17 10:36:29.000000000","updated":"2022-02-17 19:53:15.000000000","submitted":"2022-02-17 19:51:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829670-new-release","meta_rev_id":"49d7107a34a7eaec8391ea0a2e60acf593a3d9a8","_number":829670,"virtual_id_number":829670,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-17 19:51:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-17 17:32:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2022-02-17 16:40:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-17 17:32:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"}],"CC":[{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-17 11:47:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-17 12:04:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-17 12:57:29.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2022-02-17 14:50:46.000000000","updated_by":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"reviewer":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"state":"CC"},{"updated":"2022-02-17 14:51:16.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-02-17 16:40:24.000000000","updated_by":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"reviewer":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"state":"REVIEWER"},{"updated":"2022-02-17 17:32:26.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"14837db61d67502bd919c6ab098b6a06febcb556","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-17 10:36:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7aaf0c81e727e7a7083b11a4e4aebcde82c0301f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 11:47:24.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/498e15abfced4967a5d38e8ed903f6bc : FAILURE in 17m 32s","accounts_in_message":[],"_revision_number":1},{"id":"fe9f133d85d66e76bfa49f4673f936fee1ced8b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 12:04:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/164989ebd13f44fda71adb44335474ca : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7344c30a398249b4a7bcd94b7476e931 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4510fee03cad4221b8efdadd960e994c : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/445a1958db7e4a0c902dc92eafd7aeb7 : SUCCESS in 4m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2f54f47939d146d0910e3ffe1aecfd6a : SUCCESS in 4m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/98bfabd160de43d3aa243771d335156a : SUCCESS in 4m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1c1538dc15ae4d49b636aa72e29884ef : SUCCESS in 8m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dfd879b4a372403083de694c74ef8bee : SUCCESS in 10m 50s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/2c11bb629938423cae97e83d51fa013f : SUCCESS in 6m 24s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/39e5e288ac544a39ac4d4967ea23c2e4 : SUCCESS in 12m 41s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/bb833abeda65467b91518df9eae775a9 : SUCCESS in 5m 46s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/948a9367577d45a4ad7e1e4c95a64873 : SUCCESS in 8m 08s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/acca768de86243a9b3247c3c6a5f65ac : SUCCESS in 9m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7e20a0ae1e8e496db1e37ff1cce809c5 : SUCCESS in 9m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/709180ffd5a3422594078e0b0ef915a0 : SUCCESS in 9m 03s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/fee3324ef69a4479a33971b03242368b : SUCCESS in 14m 00s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/93fcbb716292474b9bacd8cb6afeb7ec : SUCCESS in 4m 42s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6869de1906c5412e83a072b0ccd7d868 : SUCCESS in 17m 08s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/390733e1c1c743fcbafe19aca1c22425 : SUCCESS in 5m 44s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/93eaf3de305f4a7aaa6b225aede2a7d6 : SUCCESS in 24m 46s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f68c1cd71ee94b14b96b6f9ce6b76a8d : SUCCESS in 12m 26s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d2a8f6970a6e404299149315b55c40ca : SUCCESS in 5m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/274360b51ede4a9791f496a97edd7966 : SUCCESS in 4m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ce1d75578fe1401f83a3305fc57e3c4d : SUCCESS in 20m 37s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/fb953f24c5d040118e1cd550f8789547 : SUCCESS in 7m 21s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e80aea389d9d4f8f95ab507ba6af13bd : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b9d12e905b004c388d424472cedb94b2 : SUCCESS in 7m 57s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b84dc352beb940f7ae2fe01c75fe4296 : TIMED_OUT in 41m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/290944b0c54b49378de68bb543bd3f64 : SUCCESS in 45m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9802eb0ff1c9497aa8ceca2592d34786 : SUCCESS in 1h 14m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f30c0e9353c6b5e15ba3ccfff46a2bb423d5b132","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-17 12:57:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba23a0a54e4a4b39312cd69f47f5ebf74a751c50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 14:16:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06388365878b44afb9e391c212465af8 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f03a912259d74999b1a0fd64041675b4 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3c455f24c294f2384d2d5a57a44940c : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ed9ea8defc54aa3b5f1a3b9f323ece8 : SUCCESS in 5m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3b504792a68c4ae984975011b8a03594 : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4a772f80447d4754a90cb4d4ae1cd4a9 : SUCCESS in 4m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ca8d4ff82ce1414fb61151d721cac3f5 : SUCCESS in 7m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/38e4f6b253bc4ec59dfc5f959138ddd4 : SUCCESS in 9m 12s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b81e94a1b0f744ea9b7e6b0fcc5ef2b0 : SUCCESS in 6m 10s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a4804088f0a54f32ae249befcc78a422 : SUCCESS in 10m 47s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/14830202d9f64c65a653cd90ecb3a8af : SUCCESS in 5m 56s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0ae0ad0b8ae845b189d63b2620b2b279 : SUCCESS in 5m 22s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0f3dccf2175f4a1d988dbe67a188aa6a : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7391184604eb46688bf0d4d841520c20 : SUCCESS in 8m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4b6ec0fb1187491aa1dc5fb8f7874147 : SUCCESS in 9m 16s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9c297dd8bfb1433ea1c5e366b3def5cc : SUCCESS in 12m 54s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/75ee17ba83da4ad09845a5c6f625da35 : SUCCESS in 5m 16s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/fe7670df451d4d58a297f45ec8b8325d : SUCCESS in 17m 20s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1d605e59c3e949ab9ac248de2f486690 : SUCCESS in 5m 55s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/67678a6b336841a18e460ed4c35c2d3a : SUCCESS in 21m 42s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0c6078a9b55f4a79ae81d69312a81696 : SUCCESS in 12m 08s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e8b34d16aff54fa6bc3434a8227ba21b : SUCCESS in 6m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cc0e6c7155ab4d5697202a735a9aabcc : SUCCESS in 4m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7f82f5237a9b494fbde6d3f87122454c : SUCCESS in 18m 31s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d3c9a79a90924c8aa2d5f1f61a024548 : SUCCESS in 7m 01s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7b37d8ee1baf4df5acbdec9001b637cf : SUCCESS in 5m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8cd263b06c614d96af71fdbb262751f9 : SUCCESS in 7m 27s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c8408dc0baf94a459edc709d558181c6 : TIMED_OUT in 41m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52f1dd01ba324cc69c193c679cff3d8d : SUCCESS in 44m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f1b5a788803b4f919cc9f9001c504c00 : SUCCESS in 1h 15m 48s","accounts_in_message":[],"_revision_number":1},{"id":"111ff7b2100dbfd8fb0dc21eacb4d5d4aec9db17","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 14:27:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6376761cc637446aafd3a49778c5d0a2 : FAILURE in 15m 40s","accounts_in_message":[],"_revision_number":1},{"id":"9ea820de2c0c1b7504212d27bd31aee5a7e21213","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2022-02-17 14:50:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0a91737f608d4d084d17f729aff4fc63daecd0e3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 14:51:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"661e25d907623866429109b1720ab49170c3a870","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 15:36:38.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a249a8a2dc614dfebf90488c00bae8e3 : FAILURE in 17m 58s","accounts_in_message":[],"_revision_number":1},{"id":"dc1ae54ebc58c108e39bf4631d9ad47fa161a487","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 16:21:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66e866a76ac441ab80314c1ddc29853e : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0983cdcd247446aafeb3e482dffa74c : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8a4984739ef454d85b8b0d2cd11b8f0 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa93a1290708493aa53e8f0307163b44 : SUCCESS in 6m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4cb8facb1d1b405489ecfe706da0b269 : SUCCESS in 5m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/66f238282a1f43d69c4262e61a4ac9b5 : SUCCESS in 4m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/afd72997d7d14f8c9114ae4a6fb74d99 : SUCCESS in 12m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/83cb6f8c21c3461aa6a0caf0a7063047 : SUCCESS in 11m 01s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/1ccb10639e1249ea941cf53d89f1e2b4 : SUCCESS in 6m 59s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b56ea4eca7024c668324755d9cbe2f21 : SUCCESS in 10m 34s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/c92cd33cebab48e8a308f8fb5fbc414e : SUCCESS in 5m 35s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8646d307ea5a4351b0a30af6b27c53fa : SUCCESS in 7m 28s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/2edb0c03aa274b76b30fdd680240caf5 : SUCCESS in 11m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/75a9e96439b44fefb6701d34c66a1526 : SUCCESS in 14m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/76c52aa432df4ed3be002c27d406ba1a : SUCCESS in 15m 13s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/3245244c00304bdc92ba11d46d2ab0f9 : SUCCESS in 12m 22s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b41120852a1e44539e3a6e0791a2f0f0 : SUCCESS in 4m 56s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9b5fd517370d45a6af4c9e0b9840f07a : SUCCESS in 17m 33s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5643709c272e45c38ad55e2f14812784 : SUCCESS in 6m 43s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d5027f1c71e748aea6752760dc95f6ae : SUCCESS in 21m 12s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/abeeafd5839e40b599f8f73091701ab4 : SUCCESS in 12m 30s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a9e2d0bf306048b6b1a943b74a2b1463 : SUCCESS in 7m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dae48fe71487401492e267823433631f : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1fe49efffd3845ffbea48703bb6977ae : SUCCESS in 20m 39s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f41e6c2e19c3444e8a6cc9d94563a2cf : SUCCESS in 7m 06s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/80bc969249d04988be5bf5b23b179a23 : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4fa449d323bc40b1b82fb7668b4ca5f9 : SUCCESS in 7m 32s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1f6fa2a605cc49a8a502f7c234f2cce4 : SUCCESS in 20m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16bf5fe152e74c2483f72147ac347a6b : SUCCESS in 50m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb83adb06ab44e2e82746e0e1fcc831f : SUCCESS in 1h 25m 49s","accounts_in_message":[],"_revision_number":1},{"id":"6c2ab503f7e700549b7c8e2c27840067747ce7cf","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-02-17 16:40:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dadb6035ab3d8f26e2614fb48f2d249938cb9326","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 17:32:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7c4da0fcd5693b832f8f1ce43c661c013e63e696","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 17:33:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5887889fba086c8bedadf6c73b33d7d9751ebb4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 19:51:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46c3397278514f7f84732969d0631c44 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10f382f5925c466ea12b6d000516551f : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa8a6d204b8d4bc79460f8aa541e580d : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16cb1d3f9f1b4a07930bba39d457b79e : SUCCESS in 5m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6a402857aeb347809898f40c61dfebc3 : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/756f1740d9474d1caa2449cc42031d12 : SUCCESS in 4m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/94093f35b87345bc85da4a3d93b14010 : SUCCESS in 8m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a159fdc65ce140abafe16f7cd267290a : SUCCESS in 9m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8a6afe9adfb54f07b506710a45a39441 : SUCCESS in 11m 05s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/b5163a4d58544c1383ab193178f337df : SUCCESS in 5m 14s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/6e450b5b66a84d8bab6d373906625e13 : SUCCESS in 6m 08s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/05ccff4124c44e98be4e43288c7dd816 : SUCCESS in 10m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/18169a0116644994ad3973c3b78eec1f : SUCCESS in 8m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7100d2ca941f4fb7a65fb3add817dc0f : SUCCESS in 10m 43s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/38b4f286fe444db9aa377475035f41ca : SUCCESS in 14m 01s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/5650ce2f064a43ac92cd78a64861479a : SUCCESS in 5m 14s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c4ab8f1b9cff408ab3dfee7c393e9039 : SUCCESS in 17m 08s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/72e57d067ad64cc3852e6b5977f044bb : SUCCESS in 6m 07s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1bca5f391f504c73a584bad045c0c50a : SUCCESS in 21m 17s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/609517456d574db7b89455abdcf8da96 : SUCCESS in 12m 44s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/5a12a37428474241b7047fd36c839a16 : SUCCESS in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8f175b8baf3e4f37b2eabb9057e95738 : SUCCESS in 4m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ffc25113b2444f3faa511bccbd43edbf : SUCCESS in 20m 28s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d264f8fc77294fa49e40e01f1df35b8f : SUCCESS in 6m 33s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7fa5ecdcd39a4a2680c1eba2c12de9a0 : SUCCESS in 6m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/93b16fb99c4e48b1930c238ec6752d2f : SUCCESS in 7m 09s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d5cb6be7cd184dc199e4ebffeaef7e74 : SUCCESS in 11m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17bf05461b07478aadea40e1f3b3e8ae : SUCCESS in 44m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97420fe5b50743d2a15dc22cccd1a6f9 : SUCCESS in 1h 31m 18s","accounts_in_message":[],"_revision_number":1},{"id":"562b323f08c76a3f19ab4de567b72a4088ef48df","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 19:51:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"49d7107a34a7eaec8391ea0a2e60acf593a3d9a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 19:53:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7b3c9702cf0438190556a0d3e680f8f : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"53714de67567beeb0bb21cafa2007ff8a2868e5a","revisions":{"53714de67567beeb0bb21cafa2007ff8a2868e5a":{"kind":"REWORK","_number":1,"created":"2022-02-17 10:36:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/70/829670/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/829670/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/829670/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/829670/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/829670/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/829670/1"}}},"commit":{"parents":[{"commit":"0e24c72df39927e1dd84d33b3e27434c15a90f81","subject":"Merge \"update constraint for oslo.middleware to new release 4.5.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e24c72df39927e1dd84d33b3e27434c15a90f81"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-17 10:36:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-17 10:36:28.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 2.20.0","message":"update constraint for neutron-lib to new release 2.20.0\n\nmeta: version: 2.20.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: Ib79171d06b30d8d5fca6103fcad225342d5a1daf\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Frode Nordahl \u003cfrode.nordahl@canonical.com\u003e\nmeta: release:Code-Review+1: Dmitrii Shcherbakov \u003cdmitrii.shcherbakov@canonical.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: If0e53f7603f5601e0931b4c2200988c2f0f51e87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/53714de67567beeb0bb21cafa2007ff8a2868e5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/53714de67567beeb0bb21cafa2007ff8a2868e5a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
