)]}'
{"id":"openstack%2Frequirements~829671","triplet_id":"openstack%2Frequirements~master~If25c7a6383695229af02001c8a21b1777b44d77e","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If25c7a6383695229af02001c8a21b1777b44d77e","subject":"update constraint for ovsdbapp to new release 1.15.1","status":"MERGED","created":"2022-02-17 10:39:25.000000000","updated":"2022-02-17 17:02:59.000000000","submitted":"2022-02-17 17:01:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829671-new-release","meta_rev_id":"3c17f427fdd5c5e9b27af32efba22f1b6fae09f2","_number":829671,"virtual_id_number":829671,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-17 17:01:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-17 14:51:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-17 14:51:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-17 11:49:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-17 12:57:19.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2022-02-17 14:51:06.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"6e71083efdf56cc3457eff4058eb7cd1c71cc579","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-17 10:39:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"792d63b8a14dd93f62f0118f35efa48b0f8221a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 11:49:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9d50da39bb84df2b5c771acc093bec2 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/236143f9f287431c8d9553faed2cafcf : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a5155e920a041ccaa620d19eec86c2f : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74ae4cfd35044dd8bf120577966075ad : SUCCESS in 4m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dd0b670e169e49469839f640d96401b5 : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0b6b14d1169f485f866fe4060e9ae08a : SUCCESS in 3m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d9f9a6a51fb3442991d430df22568898 : SUCCESS in 13m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0078a92b06d442baaf7e6cd9ec7000fd : SUCCESS in 9m 18s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/781d1877f6464a6b909781c515ad084f : SUCCESS in 6m 05s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/950bfc06e0f34dd9bc135920baeb5111 : SUCCESS in 9m 47s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/c73939502f604fd6bb3d06e2a79c5ddb : SUCCESS in 5m 14s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/972d584607104e028e3416163645e15b : SUCCESS in 6m 28s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/54eabccaf0ad4ecc86f7ba401cfec091 : SUCCESS in 11m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fdb4a8d544a14988a07a3fe48f7d642a : SUCCESS in 9m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8bb622de8d8344b394d7c9268d5a00e3 : SUCCESS in 10m 21s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b542a66bba114ed994331572aa2ffe39 : SUCCESS in 10m 57s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d960dd94c17e491098f82ed3e1e852a9 : SUCCESS in 4m 29s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/262d36df032341689e9dad5bd2169596 : SUCCESS in 17m 25s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9b96d0bfb8b64d6b87172cbe0fcec773 : SUCCESS in 6m 10s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/86c04224171640b1a43158ada3d49b2b : SUCCESS in 30m 59s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5ca49c3d9c30458b90e7335bf6d4e649 : SUCCESS in 13m 17s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/54c6605e77b947ca88833617c47ccdef : SUCCESS in 5m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a1f5a436b4624c8881e51cf2c48d2d42 : SUCCESS in 4m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9498b489a8744d6cb77aaaf83e65657a : SUCCESS in 20m 43s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b19d63c16f5c4c03bcc5220f27752d00 : SUCCESS in 6m 29s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a9da483b68a04f6abe8cafedd48e01a3 : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1a6f5bfeda574869b26c0e5611dd849b : SUCCESS in 7m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4e23cd3719044ed6820cc718a4a2953f : SUCCESS in 11m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6343cec79e0947d48cf22d42e9d8545e : SUCCESS in 44m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f7f71b3fa7f84362a7a2b08ca004e0a5 : FAILURE in 1h 06m 28s","accounts_in_message":[],"_revision_number":1},{"id":"78ca901ca44421c62393964895e12902dbf5cedf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 12:15:31.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/eec912f98c514d5e821007a35988aa49 : FAILURE in 15m 14s","accounts_in_message":[],"_revision_number":1},{"id":"5be043e9f7ae2d88caf5c7b423eabcd689714f95","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-17 12:57:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b1be19df3e2817c2f9566d7e410f8c0bd5c959cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 14:18:54.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6ae95f758a8f447d895657779e7b977e : FAILURE in 17m 18s","accounts_in_message":[],"_revision_number":1},{"id":"608a05b8ce615fc65e92ded1c5c9ded139b9feaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 14:20:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f00ee8fe47a4deaa56b1dff3ec94814 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb321f0f1ef94b39b2fb7434d83d01f9 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a92eb08b792494bb8661d824d03dec2 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b9e3f3d70f64019869b2d08c75550c8 : SUCCESS in 6m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/900fe176cc424ada9603df86828f2771 : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/24cf7ddb9c0b41ad8ccae3ccf43af70e : SUCCESS in 3m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0ac0ef0623ff4fe5b2966a100862e6de : SUCCESS in 7m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bdeb39c035b7477d85a65ed08c7a306b : SUCCESS in 7m 44s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ab6da3855718455198c49cf0476e0aa1 : SUCCESS in 6m 28s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ed3af8d127f4483f9b78896744c2e1f9 : SUCCESS in 10m 14s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/7dcfe784573245a3be4ec8fa3573ad83 : SUCCESS in 5m 52s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7b04942b95d34e39874b1951935e87f5 : SUCCESS in 5m 43s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/bdc7e16dfd044489a7491132fe74a086 : SUCCESS in 12m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b375f372dc2d4cd9ad5aea74d2a801b8 : SUCCESS in 8m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/98cc0ac79bb943c48196b7be52ef6429 : SUCCESS in 8m 40s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/fdf7d0488aa348a085971933f093d58b : SUCCESS in 11m 19s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/acb07285ebca469a990bfd526d075cb3 : SUCCESS in 4m 33s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d8fb34c4089740968a2ba6308be468bb : SUCCESS in 15m 20s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/38a8a1c5102344b190255996b83895d2 : SUCCESS in 6m 42s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d83554b4ec124658bb448d6a516d0938 : SUCCESS in 24m 34s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8d55a406ecf84460bf26340d854de9a5 : SUCCESS in 10m 10s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a860228ad88e43d8b44de731040d5a2b : SUCCESS in 6m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0e77812fe43a432485f813cac029d88b : SUCCESS in 4m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6f9a53f183094916a1abad9a9512f1c7 : SUCCESS in 18m 51s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2f8f6c7def084b12aaf74ac644f2c377 : SUCCESS in 5m 51s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d9ef53668a1c43fcac79f3f4c1e3afa4 : SUCCESS in 6m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d704892f80e2435891a6bc45c1029ddc : SUCCESS in 5m 53s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ee26dbef13804dd480bc4ed8e03658f5 : SUCCESS in 13m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/743c5961ffba431dbfa5c5d07cf85f74 : SUCCESS in 36m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2450f6f6214041218e963882ad700929 : SUCCESS in 1h 20m 16s","accounts_in_message":[],"_revision_number":1},{"id":"63f4ffde83fec665156f8686878758a2cec52803","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 14:51:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7ae720c7de6a69bb69f0d68d9be666738af8e281","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 14:52:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bdcad63bed4978fad182571a364ddc6e5b8a334d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 17:01:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/854c9c2533414df2995ee50ec99edfa1 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90ffb954186749caaf4df983d85975f7 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4033928225a464da4c5b0374e254099 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b1e9fb1a6a24618acc09e7fd463a062 : SUCCESS in 6m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/01277dd988924d36b9dc5905a70e305a : SUCCESS in 4m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/82511ad8d40942169a9684f80ec1cc09 : SUCCESS in 5m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6719106b77d0422c9c2973cdc7f3e507 : SUCCESS in 10m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/30183ac49e6640e9a5f95938653d36c3 : SUCCESS in 10m 41s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/09412e17c66f4df29432b276810502c4 : SUCCESS in 10m 25s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/3b25ac517e1a4d31a15a0da564c3e2a2 : SUCCESS in 5m 50s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/06613847a5ea46e98d592d34e04590d0 : SUCCESS in 7m 21s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/dd1dbdc4bbaa4b339e5245f6449f426e : SUCCESS in 10m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a9a149a1c118473ba77983a48e8b6bbc : SUCCESS in 9m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/55fb813a416546f0ba75d1e10f0d53c6 : SUCCESS in 9m 38s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d70fcd1759c043a4ae8f9554f395c115 : SUCCESS in 15m 42s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/cfccade5b67240b190362f79ffb1b2bb : SUCCESS in 5m 05s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/72a1a146878e4ce6af525c4f61571ee9 : SUCCESS in 19m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c1f9d26fa825430cb8168e76eb9e5ced : SUCCESS in 6m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c48d7c90ba76425cbc7de20f0be249de : SUCCESS in 28m 40s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/95e9f0f4058443ff9aa9ddcd89862c9c : SUCCESS in 13m 51s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/cef7bf668074498dab72ea88ab018e87 : SUCCESS in 7m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9e7f599591084fb39722ccc91f7453dc : SUCCESS in 5m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/387956fdfeb847e2b691675679f50538 : SUCCESS in 19m 38s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f8ae253a696f4d0f8533c0907c6712f2 : SUCCESS in 7m 33s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e3887f73806847c9aea5298d91d08af1 : SUCCESS in 6m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5b52aeb870d24c439ff8481f22075080 : SUCCESS in 7m 48s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/dd655ca8c90e404eacf1842d36d2771f : SUCCESS in 14m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13cec5277f234ac59a71baa40e8e46d5 : SUCCESS in 45m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4842c592412a4cb1b8e55caf1efa9559 : SUCCESS in 1h 21m 00s","accounts_in_message":[],"_revision_number":1},{"id":"b2c738efe6f398de2220e38a7986c30dc83e33bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 17:01:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3c17f427fdd5c5e9b27af32efba22f1b6fae09f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 17:02:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/272ed52be1b4430c802cb0c5885dd938 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ce0273be1d8948812e14e3e34b656757aa6dab52","revisions":{"ce0273be1d8948812e14e3e34b656757aa6dab52":{"kind":"REWORK","_number":1,"created":"2022-02-17 10:39:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/71/829671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/829671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/829671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/829671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/829671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/829671/1"}}},"commit":{"parents":[{"commit":"0e24c72df39927e1dd84d33b3e27434c15a90f81","subject":"Merge \"update constraint for oslo.middleware to new release 4.5.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e24c72df39927e1dd84d33b3e27434c15a90f81"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-17 10:39:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-17 10:39:21.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 1.15.1","message":"update constraint for ovsdbapp to new release 1.15.1\n\nmeta: version: 1.15.1\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: If5f6162a76893917efb58ddd2c604832cef3e1f6\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: If25c7a6383695229af02001c8a21b1777b44d77e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ce0273be1d8948812e14e3e34b656757aa6dab52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ce0273be1d8948812e14e3e34b656757aa6dab52"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
