)]}'
{"id":"openstack%2Frequirements~829673","triplet_id":"openstack%2Frequirements~master~I674ae332bf9efbec669ec43ec9ab61d4124be57a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I674ae332bf9efbec669ec43ec9ab61d4124be57a","subject":"update constraint for os-ken to new release 2.3.1","status":"MERGED","created":"2022-02-17 10:41:45.000000000","updated":"2022-02-17 19:56:24.000000000","submitted":"2022-02-17 19:54:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829673-new-release","meta_rev_id":"4384eb238e1cbc6ce9657674f8982b78cd504688","_number":829673,"virtual_id_number":829673,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-17 19:54:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-17 17:32:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-17 17:32:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-17 12:03:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-17 12:57:11.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2022-02-17 14:51:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-02-17 17:32:26.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"f406c48f9d681b26b1470355033d869ad779cc26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-17 10:41:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76d586f275aacd263b46bda9e353f0b5abeb0ca6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 12:03:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1c45adb80e84746b3b930e603450fe1 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cabcd534b35b4d2793d1e1627f28553e : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c964ef367fe4cf6b04554bc27807708 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c8ae1134a284ba1aa69b622de5cde57 : SUCCESS in 4m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/12d18e83fcc04b33bef92f3e40902147 : SUCCESS in 5m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d30c81db5c6a42e7b0a70933b38247d4 : SUCCESS in 5m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fb80337d966148ddb07434dbcc424b4e : SUCCESS in 10m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ea1fb9e6a155475c9a750fc1222b1143 : SUCCESS in 8m 54s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a18899cd4359498b864fc2de947b81ea : SUCCESS in 6m 55s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6e80fffc3a794839b7a8847859f953ad : SUCCESS in 10m 34s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/eff727fab6534454bd1862b65aec932b : SUCCESS in 6m 28s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/cecf209038a844d1b3c345cb7bc13eef : SUCCESS in 6m 22s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7f08a66e382a4854823b8723038f1b92 : SUCCESS in 11m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bc4235b0aa3d427bab0c77ba3fbc8639 : SUCCESS in 10m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/45294e34c75d45b69ab8b3479417f3f5 : SUCCESS in 9m 21s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/93740a360d7e4e36a30552d1dbe42c65 : SUCCESS in 15m 37s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/9c19620fe4ce4930ba7347017a40f100 : SUCCESS in 4m 12s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6cfda8b303f64787bd7f7a9fed428d4b : SUCCESS in 18m 01s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b59a79f220c14feeb786274380b3dc24 : SUCCESS in 6m 03s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/eb44f80e5c4b4d10a630672fee217de3 : SUCCESS in 21m 22s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/01aad00a885c429294f5f3196bc31e75 : SUCCESS in 12m 03s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/387ec0c30a914e02b407c0f7b399487a : SUCCESS in 5m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3db0d6ac8c934018aca7b362d1c37929 : SUCCESS in 3m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cf32de3774014666845b417c0d1796df : SUCCESS in 22m 09s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2646a558a12d445b962d167ba0119919 : SUCCESS in 7m 03s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4bb36cd5af824c83abc3567508d611e0 : SUCCESS in 5m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/18fe65b6913d4d88b8486c70c30792ed : SUCCESS in 6m 14s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4f968451536845ff978b05e40519152a : TIMED_OUT in 41m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2bd7aa2c87e947c7b391e1d8b086efa7 : SUCCESS in 48m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4a3ad872b1074b2ea9f9cf3753cbaa43 : SUCCESS in 1h 16m 34s","accounts_in_message":[],"_revision_number":1},{"id":"9f8e7b61c1cf59c41f61c31fcc256d07913ae530","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 12:35:01.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/37ff74103d474cd5bf76837b9f947f79 : FAILURE in 14m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a7793122eefc6e3fd14fdf3cd855c3b75ed0d1b9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-17 12:57:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0855cdb4b6899ec342c0314a8fe0b56bae1de152","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 13:48:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/99acf95c50e944c09173c62aa3ec0e05 : FAILURE in 15m 41s","accounts_in_message":[],"_revision_number":1},{"id":"3617da07a9f409d05ba5f528ecfc063bd008b717","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 14:21:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a50adccd8ac3401b8ab31068922f5809 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e178959c58f4c2cb8225ebae5e46087 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bca690d53d084532b2e0c80570e818b8 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ece25c2b7b6b4057903e47faeb602276 : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/63226c33af254644aac7e7b0583cd2d2 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c91d2c67f29f4fd98e3659a302e478fe : SUCCESS in 4m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/204c6157d3db41e89c7422d6849cd5f6 : SUCCESS in 9m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7d2dd76841794242bb77c14bcf2778bd : SUCCESS in 8m 17s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/c524df3953ee4bc1a87fb4a0799a9bb3 : SUCCESS in 6m 00s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5b28b6ab415a407fb44e6f1aa57ec489 : SUCCESS in 9m 42s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/fbd2ba6299bd4929a0b6031598124c97 : SUCCESS in 4m 48s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4cd1b36c4f9446f8812225101bb3afc7 : SUCCESS in 5m 29s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/635c82da91874aee996b8c0d4e2be7d2 : SUCCESS in 8m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9c20311e13554afa8b51b330871345ea : SUCCESS in 9m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/099855ef972644f8b33a82437bfe0723 : SUCCESS in 7m 51s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/da2e70a2afaf4b81aa0ce4bbc68c5ac4 : SUCCESS in 11m 51s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/5c4a019d158844c5a8820e6eae45ad4c : SUCCESS in 4m 15s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/49880f83ae8147969cfb5c729883c4fd : SUCCESS in 18m 40s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c11ce7b2fca84c65b42e52913029328d : SUCCESS in 5m 21s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/5bd0b12d86554bc2ac8c1a392bcd07bc : SUCCESS in 21m 26s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/99746fa3fd7f4c2d9ffb71114c31b962 : SUCCESS in 11m 15s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/01a0f5ea10ac4bbc9c0ea388f92e2fc1 : SUCCESS in 6m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b28f50d1439e40ca8cbc4a4ade770833 : SUCCESS in 4m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8edde81ec54d45f08babe5ecdd55973f : SUCCESS in 18m 40s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/18bc8c2c657d47fe8ec5bad7f92d8952 : SUCCESS in 6m 44s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5b703a92485348e0b6c031f267738c36 : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5e84be3fa14141b488ac5f97b4b14256 : SUCCESS in 7m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4aa67fbfc991493083de761cd594ae22 : TIMED_OUT in 41m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cda13ab66ca84784ae00d0f49d557bd0 : SUCCESS in 39m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d0cabead43b943b9ad57dc0dd06362f8 : SUCCESS in 1h 21m 28s","accounts_in_message":[],"_revision_number":1},{"id":"8ff818e759e7a4df42594b04cc389aadea561b4f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 14:51:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bb3bfb7f2e86e7e8a1b20165697d8d0b1ab9ef95","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 16:17:21.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d119087ed43846d387c7872ad3990794 : FAILURE in 15m 27s","accounts_in_message":[],"_revision_number":1},{"id":"92610d325183f28f57e03adfffe91c727ad96210","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 16:21:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c80c4ef0dff4f27bc4f5b070e953831 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d77a7f3d80464946b016fbcfd1e42952 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bace520adfbb4917a0e04d3f3102adc4 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8945b5791474442d885657e10f8a7123 : SUCCESS in 5m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e7322d0e263b4cb2870998f708fcfdda : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/757ab4e2500e46b59bd74db08ac717eb : SUCCESS in 4m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2c3e7795aaca4238a3385e8f78b81d27 : SUCCESS in 9m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3eebdbb8e8e64354bb4b3fd6b0d2c3a2 : SUCCESS in 9m 51s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a23423bc8589487e9db1791f36f5d30b : SUCCESS in 7m 17s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/01ca6e9c6a164f678dc62e413caa45b8 : SUCCESS in 9m 45s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/3344001d9e674bc4946d845db2d2b3b8 : SUCCESS in 6m 14s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/31bb2b5b8a454dd6b56fd0dfceab9233 : SUCCESS in 6m 10s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b738922441b141a589f4565dc4db8b08 : SUCCESS in 11m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e1e4da6b4e594534b096c24319a43b2d : SUCCESS in 8m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f694d05732a7443993f36b3d04a8d26d : SUCCESS in 8m 39s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/50b45ce97d164b9e9bfbe79e9d4ae29f : SUCCESS in 11m 15s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c29defb371ed417bba12ec684e8ff3a1 : SUCCESS in 5m 27s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/14819650afc9439c83929adc50b6da9e : SUCCESS in 25m 25s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c66d14a7b1b74663a83b0a14b9d5fcec : SUCCESS in 6m 22s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/15f96fd4c69641cb8eb054c1246f4b2b : SUCCESS in 24m 55s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8a7083f7c05148afbfe70acc600afb09 : SUCCESS in 14m 01s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ca306daae5bd46958c7cc758a5f1cbfe : SUCCESS in 5m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/608bd9a6ed684baca70acc22dbdeca98 : SUCCESS in 5m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6778fcb8cb12440da546260a15fd93f7 : SUCCESS in 19m 38s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/aba29b77388e44aebdaa0cafce00d764 : SUCCESS in 6m 34s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e9dfca7558144bb8830da5d27ff26f89 : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4da6be2f35404b9ba881c90ae59be466 : SUCCESS in 6m 57s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ea8f7690098e4c76a3b11ea9071a42de : SUCCESS in 12m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59323ac81f26486d83a4f5614968fa45 : SUCCESS in 46m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da9c54f1856e445cb000077c6348bb5f : SUCCESS in 1h 24m 14s","accounts_in_message":[],"_revision_number":1},{"id":"ff0b26e06ae1f7d103acd2b4e4c05fd4e74e72d9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 17:32:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f3d41912d0012f0ad88d165d6de4fc6c84f3f5ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 17:34:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9ae729b48e82f2c1b87f49253287614b5209a592","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 19:54:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44948aadb66a4d15b26110ece43627c8 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c93fec527ec4b619a048935735c7b50 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b3cd9c2a3de46c19bd808b354851f47 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d95a541f8f8495db953ddaf20764536 : SUCCESS in 5m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/21ac43ee0de144d49254cfb27a1e623f : SUCCESS in 5m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5da0c5784e5b488fbf31afa639e9eb7f : SUCCESS in 4m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9476f1a79c9844c182533967cfb12c48 : SUCCESS in 9m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f8ba981ded284cd189262fbbe8ad4cc3 : SUCCESS in 9m 38s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5a31bf07aaf7437fb5f0e8d1a69c095b : SUCCESS in 13m 28s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/01e314c413b747ba8ecdd2c62cf50ef9 : SUCCESS in 5m 29s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c2619282b60e47d992de222ef386cefc : SUCCESS in 7m 11s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e08cdad1b82047e6934708acd6718629 : SUCCESS in 11m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/55488dfd438346299c14b9fd15579cc5 : SUCCESS in 10m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/861087babb9e419cb4489e90ab9f9f37 : SUCCESS in 10m 48s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a5ed7cbc291b4806ab3022dcaa04576c : SUCCESS in 14m 55s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/fd80aa7571f74700be64f44a1a08d1db : SUCCESS in 6m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9457e67219a34b6f85faf83a558c0503 : SUCCESS in 16m 38s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c4a50f6e4abf4dc392aec51c9fc0fa0c : SUCCESS in 7m 15s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/cf7df2b4eab54340ae56b83f5ee14a15 : SUCCESS in 27m 52s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f11eacb948a54599a07d7a4f1cba08de : SUCCESS in 13m 29s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/590e85c21aac47dd89caf9e1794e4933 : SUCCESS in 7m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/670a3f34196446669d7c6057056c03e0 : SUCCESS in 5m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7ab2cd785fc24eb2a21159908a98cfc6 : SUCCESS in 20m 15s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/14ed0eae403649d4aa3b246a24c8ef78 : SUCCESS in 7m 28s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9ed6668fd3dd4af2854d20de6792333c : SUCCESS in 6m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c34d1b4a76cc45db893da86d9e2a4d04 : SUCCESS in 8m 02s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cd7c8c2c8695448d838cea10362e79cc : SUCCESS in 18m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b31250080a724befa5561795f3c20a10 : SUCCESS in 47m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/13de36d7d38f45b9bc93ce878aaa5166 : SUCCESS in 1h 31m 26s","accounts_in_message":[],"_revision_number":1},{"id":"20f373c7cff06b030c1eb29369fc11a18bd341a6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 19:54:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4384eb238e1cbc6ce9657674f8982b78cd504688","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 19:56:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/565aa9179c0a4bdca606b41351622002 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"982ab445504d62ac783085aa05c1688ebc0252c7","revisions":{"982ab445504d62ac783085aa05c1688ebc0252c7":{"kind":"REWORK","_number":1,"created":"2022-02-17 10:41:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/829673/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/829673/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/829673/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/829673/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/829673/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/829673/1"}}},"commit":{"parents":[{"commit":"0e24c72df39927e1dd84d33b3e27434c15a90f81","subject":"Merge \"update constraint for oslo.middleware to new release 4.5.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e24c72df39927e1dd84d33b3e27434c15a90f81"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-17 10:41:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-17 10:41:44.000000000","tz":0},"subject":"update constraint for os-ken to new release 2.3.1","message":"update constraint for os-ken to new release 2.3.1\n\nmeta: version: 2.3.1\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: I108d5ed49222b48ceef37400df75b2effb97573e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I674ae332bf9efbec669ec43ec9ab61d4124be57a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/982ab445504d62ac783085aa05c1688ebc0252c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/982ab445504d62ac783085aa05c1688ebc0252c7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
