)]}'
{"id":"openstack%2Frequirements~829674","triplet_id":"openstack%2Frequirements~master~I1d2a48aa9339d965376f820b7f00f9ac9db46ef4","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1d2a48aa9339d965376f820b7f00f9ac9db46ef4","subject":"update constraint for octavia-lib to new release 2.5.0","status":"MERGED","created":"2022-02-17 10:42:17.000000000","updated":"2022-02-17 21:35:48.000000000","submitted":"2022-02-17 21:33:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829674-new-release","meta_rev_id":"5a4f6453e63145818d99692c7cc00742201b7f71","_number":829674,"virtual_id_number":829674,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-17 21:33:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-17 14:51:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-17 12:56:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-17 14:51:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-17 12:06:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-17 12:56:52.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-17 14:51:07.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"680183111bc3487488728eb6de495354d07cf061","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-17 10:42:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be54b32e167663e9a65a594df11689daa4a8d05d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 12:06:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eeee62f67d504e44a93a51ff183711f1 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e448945e65449e08f46a6bc0e7fa5b0 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6faf30e43cc4ad09d9b76e52db52386 : SUCCESS in 7m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/500b5367ea5e4467b84ac1bdb6406b40 : SUCCESS in 4m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/debb6ccd886b46289301ae005967dbd2 : SUCCESS in 5m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cc656d1bb88d4fa195a7fd0617b3ff07 : SUCCESS in 3m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bc051f319bfc475a91240dc402f78809 : SUCCESS in 9m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6c0f32cadb3847a6956a3375f173bdf1 : SUCCESS in 11m 45s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/3f88a6fb21424f3bbe9c6667e0dccbcc : SUCCESS in 5m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d87db62b374a4b44a27edae104f8da0d : SUCCESS in 10m 23s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/d5399ab86eb44550b30b4d6aba8d3434 : SUCCESS in 6m 38s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/6bf09e1fbf194234928d0bce7f86fa9d : SUCCESS in 6m 17s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c3eabec94b754a358f5365bea6cebecf : SUCCESS in 11m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/42a198f1167e4ce9a92de9d81c313380 : SUCCESS in 7m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ae062bc02f5a45ff83da1f349d7c2c99 : SUCCESS in 12m 22s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8cc9aa28ed404bd28bea57ced21ba4fa : SUCCESS in 9m 58s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/f1b58200a3f74bfaa595a1ea6218147f : SUCCESS in 4m 52s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/26eb2fe88ff04330b3569e99de587cea : SUCCESS in 17m 56s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b5d1736ddd7540a381dd38be4c6e202d : SUCCESS in 5m 47s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/218f7957118a48f2bec69cc74b5201cc : SUCCESS in 27m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/627096f8c5e2443387f512b71d072f12 : SUCCESS in 12m 40s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0eea801e4e4747b0bb314a4569f9c85d : SUCCESS in 5m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d541a2b259a14598858eff24941b5c5a : SUCCESS in 4m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1b342864700e41f6a236588cb9005fcf : SUCCESS in 17m 27s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/df9e4baccd4b4e00a49c8db059866c4a : SUCCESS in 5m 29s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/fcd5338db5de46b981b294b280147789 : SUCCESS in 6m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/88a78e1e10ab4a2d8e90623861d73826 : SUCCESS in 6m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a03729876bbc48ff966b0cc652f363e4 : SUCCESS in 11m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a36f4568fec4ee292ed273222b13c1d : SUCCESS in 44m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a15103ef030e48da9d961e0235f5e5f3 : SUCCESS in 1h 21m 08s","accounts_in_message":[],"_revision_number":1},{"id":"d2b9a5af181d643ac0ae63414d18bb02ef6eea8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 12:54:40.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/361894a80ea6440285df2bcf04a8c0cd : FAILURE in 15m 58s","accounts_in_message":[],"_revision_number":1},{"id":"873f3cf77d6f63acbf5332402470ee8465ee5fd2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-17 12:56:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cfadc821188c1750ce8d3a6b3ff29422688b6f99","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 14:51:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0fe10e6722086af97ca29eec6f796bd46c3b8842","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 14:52:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"18c17ef3454a51cdf00f7920eb5b7b13b9cae150","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 17:05:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26ba07d4362347a183cf0f32c4f014aa : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2b2bd4677614a42ab6d19c94ac2e4ca : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd82ab79921746d1919f361d37fbd4b8 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1874de9be7d4abcba294a3c5ddfc1de : SUCCESS in 4m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/626fc31f3003477dbef91e9668a1465a : SUCCESS in 4m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/169b6423dd1545f088150474059e48cb : SUCCESS in 4m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/05fa28e5930745f089ddedc03b61479a : SUCCESS in 10m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ba1bc848c5364e7ea976da9a0e079429 : SUCCESS in 9m 32s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d2482ef1e20b4386a96ccce83762585f : SUCCESS in 10m 10s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/6518029b3b224d3ba4f8c3f50ead9f4f : SUCCESS in 6m 12s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/6f9b5a305ce544ceace3f7c6c68b50f1 : SUCCESS in 6m 29s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7210fce06dc6454d9af699ab437fabab : SUCCESS in 11m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eefb87f292fc4bc3be3eedb642536b56 : SUCCESS in 8m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c62089e11d5e41b8a9c20f3dfafc7ce5 : SUCCESS in 8m 52s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/aa5d7c640cda43789e3293b66bc7278e : SUCCESS in 12m 21s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/3b931fb83df040449d8dd78edd4b185a : SUCCESS in 5m 01s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2764950fa4994270a7c0504e2ce2d104 : SUCCESS in 17m 17s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a575c87dd6c2468084434a7fee50d792 : SUCCESS in 5m 53s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/013ee889c359401aafd91b40b90f1e83 : SUCCESS in 22m 42s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/02581c194e514f83bef784f99848bab0 : SUCCESS in 12m 58s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a31321ab1c5f4f029d2751010b3d0337 : SUCCESS in 6m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ba28df62180e4183abbe52ea2978e034 : SUCCESS in 5m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1e79954cff7243b4b0a00db528792772 : SUCCESS in 20m 26s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/de6252bb399e449b84c69bd97442c544 : SUCCESS in 6m 31s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/59fe45dce5e64baba61501da6a0110a3 : SUCCESS in 10m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e49f08299b3640d697165f2d2a27d623 : SUCCESS in 7m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ad17bfe0f4bc47d18bf906964c502fad : TIMED_OUT in 41m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5081bb88a94463484e014ce84fde49c : SUCCESS in 43m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3ae9886b1f0740429ac5d9d9db71fcaa : SUCCESS in 1h 21m 00s","accounts_in_message":[],"_revision_number":1},{"id":"3a52280c01105dfbf114a367617a997a287d39f8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-17 17:32:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f2846d316f72d21cae6c2c12ce4d2ccea77b8386","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 18:31:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8731af7c28544c7e865d586c3107a3d6 : FAILURE in 15m 03s","accounts_in_message":[],"_revision_number":1},{"id":"b43863322b5fa41770b06117fe302efece3d6c63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 19:00:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddeea036143741b5928c8cecb9eea883 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b80b30be703476794f3efa2a706e5ad : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1e97d11cd434be5b670cc9e675360bd : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b1c392b21eb43ef9b98a688af109357 : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3e6303f9e9a0465c8c125e50735318e2 : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e2ab9008e076483ca0f5378a9a12429c : SUCCESS in 4m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e5522dbca2bf4deb9f0a53b0533f148e : SUCCESS in 10m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/196457b712cb4d14a848e97cf9a080c7 : SUCCESS in 8m 35s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/5b9be0484e4f4d789eab0fb0e1a89812 : SUCCESS in 5m 32s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/13ce41d7caf24f4d8459b6bb7ccb3197 : SUCCESS in 15m 07s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/839a110d23784fb9986cad4d2e27033c : SUCCESS in 6m 35s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5e051fa6a7b34c1a8955d1c8ecdaf7ad : SUCCESS in 7m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c8590faf2286459bb688a67e871becb7 : SUCCESS in 10m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1db76872587c447bb656bdef3eb02de0 : SUCCESS in 11m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/95db811217a24fb3bfe4b0ae1bd1b2ca : SUCCESS in 9m 45s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ec89ec408e1446a2ad3e15ced4842edb : SUCCESS in 13m 18s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c08a404433d645a4857e6a82dc1eada4 : SUCCESS in 5m 59s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/fa594706a59f41ebad86b88f4c19e667 : SUCCESS in 16m 29s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/21eac58f4d104ba3be5b88c3fa7ba275 : SUCCESS in 7m 33s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/76a4e9a5d12246fbae8f96a993c86a1e : SUCCESS in 24m 18s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/98cc114b1dde4339bc78c35b7a575518 : SUCCESS in 12m 09s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/962b0f4f97794cff958b12e53ae3b4e5 : SUCCESS in 5m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/73a16e280b3a4a6d9139e886c033d795 : SUCCESS in 4m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/90247dd7269d4f71837b10419499ff94 : SUCCESS in 19m 27s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5b7909a68bf543deac1586b7ee5fd67c : SUCCESS in 6m 03s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7f861d58935642e8bea128a502d7c334 : SUCCESS in 6m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2aade66e4316439baa31e7c3c94d0ba5 : SUCCESS in 6m 23s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2298de9de8b24dc7a2045fea4c1ac99e : SUCCESS in 12m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4f30c355eeb4e6eb2b442445ef5fa4a : SUCCESS in 39m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8824aded5b024811b790f59faaaad162 : SUCCESS in 1h 24m 11s","accounts_in_message":[],"_revision_number":1},{"id":"8d1599d0bc5c7041c2e7bfe0bc40ed90b5e20009","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 19:02:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8d7deaf347226d216174362a583a6e504357fc83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 21:33:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a17857ca86743f5995b239e5eac6196 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e506bf0da32f4ddcbe621579c5867870 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c4f9c4776c9420692e9199cbed0beef : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7eb712997ad94dfb8103e4d621770bb2 : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ea1832ce6f944279b3425762da717597 : SUCCESS in 5m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d96b82063db843939cbeae7ae0390ec6 : SUCCESS in 4m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e7b2aeec420742d18090f6c891382f7c : SUCCESS in 8m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3548cb27e4204044b132f89f13fc2f70 : SUCCESS in 7m 15s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/025a8ef561574feb90c5724e0f31b726 : SUCCESS in 11m 02s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/81fd530140b245519ec988afa7b3bf0d : SUCCESS in 5m 38s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/898647f200be456cb554563509f64eaa : SUCCESS in 6m 10s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/47a80487275b46949f54021187c36492 : SUCCESS in 9m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a286182ed3d943bc847157bd63991060 : SUCCESS in 7m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/11af7705a57f4e71a652906833769f47 : SUCCESS in 7m 47s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/359b350dd6534a5695c9c58805c54a51 : SUCCESS in 12m 45s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/4d43b36933e9405c9731782cad9bcbc4 : SUCCESS in 4m 28s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b1cd8dabe0fe4ca0bc8b193b28cd42b9 : SUCCESS in 20m 56s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6d600cfdc43c48cabef98854a54b2496 : SUCCESS in 5m 03s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/720534b68d084ccc8cbab357583f4e2c : SUCCESS in 20m 43s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c8705432625b457193feeef196c411af : SUCCESS in 13m 06s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e4f14b7c3f604c28818cc612ea80d8ff : SUCCESS in 5m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d35d6098652d4ef08a909d3562274585 : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/50dc91eca9dc4c6aa2d6421a2ec43f21 : SUCCESS in 21m 45s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4d90a76d9ab6426dbf73300d49f6810c : SUCCESS in 6m 06s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f3047043633241d2be61126895845792 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9e02d8b29d47498ea02cfde3ce6160a9 : SUCCESS in 6m 16s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9cc5a5e4185e4fe8b153ef5dd72a01f2 : SUCCESS in 9m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68e0222865cc46c6b669553e19717a06 : SUCCESS in 32m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/410af8df94fb42b9b4559b12d45a020d : SUCCESS in 1h 11m 26s","accounts_in_message":[],"_revision_number":1},{"id":"cf1d16834d8ae1a8c73612f23a7e7bed188368ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 21:33:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5a4f6453e63145818d99692c7cc00742201b7f71","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 21:35:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e668271201da4052921ee3365da9c679 : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"263c14bd2407d1d665b4d200977058c0c4289bcc","revisions":{"263c14bd2407d1d665b4d200977058c0c4289bcc":{"kind":"REWORK","_number":1,"created":"2022-02-17 10:42:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/74/829674/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/829674/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/829674/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/829674/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/829674/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/829674/1"}}},"commit":{"parents":[{"commit":"0e24c72df39927e1dd84d33b3e27434c15a90f81","subject":"Merge \"update constraint for oslo.middleware to new release 4.5.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e24c72df39927e1dd84d33b3e27434c15a90f81"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-17 10:42:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-17 10:42:13.000000000","tz":0},"subject":"update constraint for octavia-lib to new release 2.5.0","message":"update constraint for octavia-lib to new release 2.5.0\n\nmeta: version: 2.5.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: Iad647566a86cffa4f49bbb6eed64c1bcbc856f32\nmeta: release:Code-Review+1: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I1d2a48aa9339d965376f820b7f00f9ac9db46ef4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/263c14bd2407d1d665b4d200977058c0c4289bcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/263c14bd2407d1d665b4d200977058c0c4289bcc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
