)]}'
{"id":"openstack%2Frequirements~829932","triplet_id":"openstack%2Frequirements~master~Ifc088f59d20f81451909c220054774c485f3ffab","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ifc088f59d20f81451909c220054774c485f3ffab","subject":"update constraint for os_vif to new release 2.7.1","status":"MERGED","created":"2022-02-18 15:46:18.000000000","updated":"2022-02-18 20:51:01.000000000","submitted":"2022-02-18 20:51:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829932-new-release","meta_rev_id":"6c583ac0f79151aadca45a7a085f6c4130f0d958","_number":829932,"virtual_id_number":829932,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-18 20:51:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-18 19:21:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-18 19:21:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-18 16:24:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-18 17:16:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-18 19:21:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"bac4417c4c5852106d982462f2f03c2aedbb2efa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-18 15:46:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2df8412eb9df9f0d4aa3b5ddcf2c77daccc74038","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 16:24:33.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a70f06dcdd514277a80508d944ab25ab : FAILURE in 15m 55s","accounts_in_message":[],"_revision_number":1},{"id":"4216e7751528bc2ee7c11fbd94246185bff51a51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 17:16:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b02a2124e504556a943cd06926110da : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88cd9370f22c474fb784e0429096ff09 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/071de9c653cf41acafc3576dff0bfbd4 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7c3a4972d2f47f0a8464ff64d12481b : SUCCESS in 4m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/619d8126b8bc47029c21f86b0e487305 : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/804d4ce4e5d748a09adb74f12ed982b1 : SUCCESS in 4m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/086dc28072134a8db567f1474ad24faa : SUCCESS in 11m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/eb855bd2764d4796bbd0e77d5b4dc286 : SUCCESS in 12m 15s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f17f88855dfd4b36958971e21fcbe5a4 : SUCCESS in 8m 13s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e063f1cf543b4015a2d4007d7dda2f50 : SUCCESS in 11m 17s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/007e0b731436494a97438d49e467a366 : SUCCESS in 7m 08s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/22903a9de5fc415ca614fa452612f0d1 : SUCCESS in 7m 23s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/354052a1ba4449ee9464e9ee69db6cf0 : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5a79057080f0450e85153b0e9793be84 : SUCCESS in 9m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cc51edc0c21f4ce2ba869c365809e300 : SUCCESS in 9m 30s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5fa09f945f124ab6907f94f219488197 : SUCCESS in 11m 15s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/23c73a8dbfb948b1a46e840a9ae44f45 : SUCCESS in 4m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/23faab3019174364aff09d56e05fbeaf : SUCCESS in 16m 34s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d1d4deb8a1134c0881b2572dddff55c7 : SUCCESS in 5m 06s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/14082c949e674f0c89360aa231f45a37 : SUCCESS in 24m 27s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/afd73ae04ad94a45b1f137f3ad892a47 : SUCCESS in 12m 01s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/5539132d28d84458b9b54437174eac6d : SUCCESS in 5m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bbd4092047bb4bef9df2aa3462227de4 : SUCCESS in 4m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/28c7bf30aba64dc78af763fe06db9f3b : SUCCESS in 19m 43s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/cb986c774eb84c70b2bfc9fe5952c3f2 : SUCCESS in 6m 47s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d045d6b5c0674870b3891aa50b9f2db3 : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ea855e63f8a54e8ea75f23226cb03dfd : SUCCESS in 6m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f6d54f28a7034418bab6c89f8db8c45d : SUCCESS in 10m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2c31ae889d746c48292bc017e0ee2c3 : SUCCESS in 37m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cf249c9605bf4078a7e82371c8242422 : SUCCESS in 1h 24m 55s","accounts_in_message":[],"_revision_number":1},{"id":"5b65b2d75b3b54e345f6c093766be0ef981f11bc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-18 19:21:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cf490b446da7c00142b8aa767f541de3aec91295","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 19:23:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f0c169c47583241bdabb679946b5e6e6b5009391","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 20:51:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/918477c3a1d14b48acb30342ee58f4a2 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71bb92e06a814c059dd6196837245ceb : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29bd8466eb4144c2b0ea25bd89a19368 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6dade8ee2534b48ae80e85107d4bd26 : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ca0c6957623d41c48746bed52803d96e : SUCCESS in 4m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/78862b374cd0415ea788749bbab98d14 : SUCCESS in 5m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0aa013a091854c9bbcf3c6e89bc84515 : SUCCESS in 11m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6b797534abd642ff811bf14f5f573950 : SUCCESS in 11m 51s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/52a77de33ebb4e0aa329ee73d72619c3 : SUCCESS in 9m 59s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/ecf1afc8ed734412a5c3984fd731bf6f : SUCCESS in 5m 12s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7a81d9a6cf8642f59d4c0f7115267aae : SUCCESS in 7m 15s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0160c92f9f404ed787703c729cedbc49 : SUCCESS in 9m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ce6effb71a7947229a8464a6881ade6c : SUCCESS in 8m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e9faa0f3e842490d93c7f1b4d9460b6f : SUCCESS in 8m 45s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/67def341bfa74f428c82bce6d7fcfe21 : SUCCESS in 13m 27s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/badae16f35bb45e58aa1990faf584406 : SUCCESS in 4m 52s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/907d58a2d110492e8e6ce20925a6cf62 : SUCCESS in 25m 28s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9cd231c6b101448a9ed5a49f116a6525 : SUCCESS in 5m 58s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e0084309396247caa2a03281df866d16 : SUCCESS in 21m 39s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/763ea4a55fab4d538849d237bea19e53 : SUCCESS in 13m 10s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/14dac01f84b946b69bc3f621029fe222 : SUCCESS in 7m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/02fac680c1884f3496fd60590ab21def : SUCCESS in 5m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c11c943dc4ef435bb496a61a279823a2 : SUCCESS in 19m 37s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b48ce891068b4dbc98640e6569121afa : SUCCESS in 6m 32s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/177d16928652483d8c84bbbfcd6f4edd : SUCCESS in 6m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c470c01e770144cb8dcc8c5ac700083e : SUCCESS in 10m 01s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cde170704afe4e7b9c46f36ef325a4c1 : SUCCESS in 12m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca7270158d7747e88c9b24340c5913d7 : SUCCESS in 47m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ca745de6152a4d95a51fe2c809a24c9d : SUCCESS in 1h 18m 17s","accounts_in_message":[],"_revision_number":1},{"id":"6c583ac0f79151aadca45a7a085f6c4130f0d958","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 20:51:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b0b71a2885ab83a6f54ac285a25ffea30d13168f","revisions":{"b0b71a2885ab83a6f54ac285a25ffea30d13168f":{"kind":"REWORK","_number":1,"created":"2022-02-18 15:46:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/32/829932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/829932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/829932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/829932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/829932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/829932/1"}}},"commit":{"parents":[{"commit":"189eaafac6c0707762dd258a09f96bb2fb3cfd68","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/189eaafac6c0707762dd258a09f96bb2fb3cfd68"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-18 15:46:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-18 15:46:17.000000000","tz":0},"subject":"update constraint for os_vif to new release 2.7.1","message":"update constraint for os_vif to new release 2.7.1\n\nmeta: version: 2.7.1\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: Idf3e547adf20fd3e22b8ecf0d015a5b0a37afb6c\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ifc088f59d20f81451909c220054774c485f3ffab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0b71a2885ab83a6f54ac285a25ffea30d13168f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0b71a2885ab83a6f54ac285a25ffea30d13168f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
