)]}'
{"id":"openstack%2Frequirements~829935","triplet_id":"openstack%2Frequirements~master~I327d984e208fed09436e984cf8300db7ed369715","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I327d984e208fed09436e984cf8300db7ed369715","subject":"update constraint for osc-lib to new release 2.5.0","status":"MERGED","created":"2022-02-18 15:53:47.000000000","updated":"2022-02-18 20:51:04.000000000","submitted":"2022-02-18 20:51:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829935-new-release","meta_rev_id":"dc8af977e8bcdbfbfb8866991baebd2fa147e944","_number":829935,"virtual_id_number":829935,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-18 20:51:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-18 19:21:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-18 19:21:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-18 16:55:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-18 17:27:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-18 19:21:47.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"d57cc870d7f0dcb51db7e35390f199c9ab1829c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-18 15:53:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8428d6165e7d25bcd9f4035ca854a2238558585","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 16:55:14.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/72695782f7c240998f57dae339081c49 : FAILURE in 17m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e234dae0214facd7337173ec3598cd96deb3258d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 17:27:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3132a6395a5438fa79a9dba82bfd73a : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ecd3d21677c409386087eafb7489809 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f4c5554d1cb4ceab55ceb77f66cea8c : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4523192abef4c87818c3e2ebeb86ff0 : SUCCESS in 4m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2b953121dd5c45a687cc2f5f051f710a : SUCCESS in 6m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4ba57a03052f4bd589f1479ec75bbef6 : SUCCESS in 3m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f57ae79eee6e48ab82be1776f1fb714b : SUCCESS in 12m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d0eccbf2fbbf4002b00adc1ae3af33e1 : SUCCESS in 11m 44s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/7f128211d1e049949934fee5dbd249a8 : SUCCESS in 6m 40s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9ed1686361a1464d913aeffba4f39aaa : SUCCESS in 13m 02s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/de28f56b8b224e428956f4879a1a5186 : SUCCESS in 6m 11s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/98b76ec413e642ddaf0af85b813e3b4c : SUCCESS in 7m 34s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a10493299f1641f9b6c21dad3651ad70 : SUCCESS in 9m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d020916ab0224835801d5614cc123a97 : SUCCESS in 10m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/67965509e63c4f759c7deedef1ebe0fa : SUCCESS in 9m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0c6a2bd05b244c23a1ecf3c6c8edfcb9 : SUCCESS in 15m 42s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/bc72d32224ad4761bc117fdaac2ddbb1 : SUCCESS in 6m 16s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/eae4e3ead3da4dceaed4d054153918d4 : SUCCESS in 17m 05s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2b88750988b24b3fbe86ef8e23c1414f : SUCCESS in 5m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8fde4d15609a47ac878f5b7e7f3a1b57 : SUCCESS in 22m 19s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a77eff8bf8424c6b8586d5ccf34d56ae : SUCCESS in 14m 24s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c04e0794e6d447f5bb0751265aa1403f : SUCCESS in 7m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/70936770bf614a3090e4989b509e3f92 : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b2cdc9d1eebd43fcb3f0268a57ef48e3 : SUCCESS in 20m 17s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/12d46b84c16f4f388df77b4ce4d97145 : SUCCESS in 6m 36s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/10ea4b596d9a4b88aa6f5aaa438c1297 : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aa28b844c3b545fa989f01b29a7460da : SUCCESS in 7m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cd63e25bdeb64946a9add760cd6ba7ba : SUCCESS in 14m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e464d78d74024b3cbeaeed6d0b030808 : SUCCESS in 39m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d260bdff9bc46bfb3a2f19a87d87477 : SUCCESS in 1h 30m 14s","accounts_in_message":[],"_revision_number":1},{"id":"89b6e54fa47436d03ca4e01a2600025d9fa104a4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-18 19:21:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4fb2518da858280c70b72c3cd835e890f2b5c618","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 19:23:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6ca67e52aca3fd55f0294b2428884069305398cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 20:51:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b140ede4c77c4e3f8f52a361133e1775 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b87bf178e9e498b86c0669480d3cda4 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9dabae09025e49e98835e6e694125131 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96f8c9142df1499db56485feb7dadb62 : SUCCESS in 4m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/32f55e22018d4a2aa20efe2176a83988 : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6e312ad77b90477fa37d0e6362540a7a : SUCCESS in 4m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0181f3844c7d4288823d3a8058e6afb6 : SUCCESS in 13m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c5cdb17df9784b0e9167a5a3c53fe061 : SUCCESS in 10m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/04fd98c1ef3e4727b68f7da7e62129b9 : SUCCESS in 8m 58s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/10a42c6d2f3b4166982c2a1f03501023 : SUCCESS in 5m 01s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ebeed0c458004443b5f751dd78434453 : SUCCESS in 5m 46s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c2ecbe5c8da14207842c53c98022273b : SUCCESS in 8m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e75b4a9477a34a4fa924fcfeb97104c2 : SUCCESS in 9m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/adaa606f5f9c4ab09be1331de29e81ab : SUCCESS in 9m 19s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/50ff2d5cf2694c739fb21f501587a2e4 : SUCCESS in 14m 27s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/badd111a8c9b45a6870ea182429fe09f : SUCCESS in 4m 38s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6fc0340158b843109e459762c6987ad0 : SUCCESS in 15m 37s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5240d6ae41b9401aa199312fd521ddff : SUCCESS in 6m 18s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d76cbeff33f249b2b3e9fe092fe9d7a6 : SUCCESS in 20m 33s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/aefc2a80eb624ace85edb52e9eb33838 : SUCCESS in 13m 00s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/972ef8bc4d1b4278b1f23c20ed1b46de : SUCCESS in 5m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/89839f83bd104731b1feb0d4bb008bd8 : SUCCESS in 4m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f302fbddb2f24b91955246ba0d40a428 : SUCCESS in 24m 00s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/fb25e40134164c72bdc67a7e3cfa71de : SUCCESS in 6m 30s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/687b4e92de1b4f0da6d2d26fdc66f3fa : SUCCESS in 6m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/784af71397eb445e9fd166e7a0e1ca2a : SUCCESS in 7m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/eb8e51a276364303824d0a3e74b8c585 : SUCCESS in 12m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b03b1688f6a475b9f26858c5fdd62d8 : SUCCESS in 51m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f445323d4914b3f9f9730e06bbdbf80 : SUCCESS in 1h 16m 28s","accounts_in_message":[],"_revision_number":1},{"id":"dc8af977e8bcdbfbfb8866991baebd2fa147e944","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 20:51:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"782b242f11da1ca348f28f22db3e3a3eeb9112a8","revisions":{"782b242f11da1ca348f28f22db3e3a3eeb9112a8":{"kind":"REWORK","_number":1,"created":"2022-02-18 15:53:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/35/829935/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/829935/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/829935/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/829935/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/829935/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/829935/1"}}},"commit":{"parents":[{"commit":"189eaafac6c0707762dd258a09f96bb2fb3cfd68","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/189eaafac6c0707762dd258a09f96bb2fb3cfd68"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-18 15:53:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-18 15:53:46.000000000","tz":0},"subject":"update constraint for osc-lib to new release 2.5.0","message":"update constraint for osc-lib to new release 2.5.0\n\nmeta: version: 2.5.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: Ib40c93c5b783435cd3dcb7a357a6da8e176329b9\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I327d984e208fed09436e984cf8300db7ed369715\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/782b242f11da1ca348f28f22db3e3a3eeb9112a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/782b242f11da1ca348f28f22db3e3a3eeb9112a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
