)]}'
{"id":"openstack%2Frequirements~829936","triplet_id":"openstack%2Frequirements~master~Ie5969c24c0acd69a6a5c84854c288eb544d9a0f1","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie5969c24c0acd69a6a5c84854c288eb544d9a0f1","subject":"update constraint for heat-translator to new release 2.5.0","status":"MERGED","created":"2022-02-18 15:56:30.000000000","updated":"2022-02-18 20:54:12.000000000","submitted":"2022-02-18 20:51:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829936-new-release","meta_rev_id":"24571ee16cf22dc60db9d8e391b76bb39cb398de","_number":829936,"virtual_id_number":829936,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-18 20:51:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-18 19:21:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-18 19:21:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-18 17:03:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-18 17:24:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-18 19:21:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"358032177486f9fa0f0e5912f31c24a012274541","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-18 15:56:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15c5d2e21eefbf3155f8e495cf50bc54d3971e1f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 17:03:17.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e1f5d649f58d4f3aa4020fed5ceec9f9 : FAILURE in 17m 01s","accounts_in_message":[],"_revision_number":1},{"id":"2dc9d4ccc791206ff8ed78156f7b307c1ef374c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 17:24:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee787dcd13674582ade1e6bfc3958fea : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c9d18c92c554a84b8c1eb25c0c086b4 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7cd3c4faf8f4a6ca1d47bf1d49d73e3 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd99e559624c4dc0925f567431cddb9e : SUCCESS in 5m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f4445ce1df0842fb8d2b59f03127b687 : SUCCESS in 6m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9f65a8aeb90b4c47a65a9ed4a79ae54c : SUCCESS in 4m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9feed490a93543cd8c75df24351a55e6 : SUCCESS in 9m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/14a4132fdd1b47c4be0d0462317a16f6 : SUCCESS in 11m 12s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/06fd61c2b851421fa3ffd53a74f80e27 : SUCCESS in 6m 25s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/46033ac3019a487a8f4d94b6ce2aabcb : SUCCESS in 10m 17s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/272ad0c3320d4713ac729f402b759f5d : SUCCESS in 5m 24s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ed14499321694c2d81dabdfeea97d45c : SUCCESS in 6m 09s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8e5d44e6619d4a258a0a6e2f8be3ca58 : SUCCESS in 9m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/15fcaaf916ca4b7fb45d19a944b38644 : SUCCESS in 9m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f0b7c3b1f6fd4ca4838e65a16809e6bb : SUCCESS in 9m 20s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e8a714b9372949218bc97106ef0c6705 : SUCCESS in 12m 13s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/1f340b1167a44400ac2e6fbd7c534135 : SUCCESS in 4m 57s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ef8c01528eb046d09ac28e65de54b445 : SUCCESS in 15m 51s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a219259f2b6c4c0e8591d4ca80aca7ba : SUCCESS in 6m 20s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/84ec127b806d43e2aa18bcc658ebf9b9 : SUCCESS in 25m 57s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/decd2c9a626a43cd9d4dd6416f2a3d7f : SUCCESS in 12m 28s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/207a613c54e74a959afee81d2df05437 : SUCCESS in 6m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/465e62d6e4e24c3ca19bc8e4ea65a721 : SUCCESS in 4m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d132be2a9d3f47368149cfa52b6ceb0d : SUCCESS in 19m 53s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/faa701ae4b99470aaed3591e7dddb9be : SUCCESS in 7m 19s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f04699f1caf74e7cb6904d3836e36d35 : SUCCESS in 6m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3dbfe158ba164af79284017f0ef794ff : SUCCESS in 7m 46s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/43da7971c1b24ed9b94b79c2d14753b4 : SUCCESS in 13m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa0f457b38e04ed8aa821b00e60136ec : SUCCESS in 39m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25f93eaf8a5948d2a4e6fd90332249a7 : SUCCESS in 1h 22m 04s","accounts_in_message":[],"_revision_number":1},{"id":"296103909066c3541fb96361f605b281105e45d5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-18 19:21:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fe3ea1a78a602de56fdc76cef412eb1b9d0a29fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 19:23:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"39107fe1800dcf1062d29ff18320bcff788058c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 20:51:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4fa963d984647a1a48ee0919336f7a3 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2645547002f844888fe4c1b031d7f818 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d476c578246149a39daefbf748a0ced5 : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b345158b91094686a88b2b01c734cc76 : SUCCESS in 6m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e26e2ffe97ad40fc97a8e87e6add556e : SUCCESS in 4m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f1292fc3a0e541be935d91b72d71b6eb : SUCCESS in 4m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0590ce3fca394eb79aa56dda3a58c218 : SUCCESS in 10m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ea0d84abd9704c8e9c7a4f77b1b24123 : SUCCESS in 11m 18s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/fa9912c114ba4bc88b4e85b0f2ac3208 : SUCCESS in 10m 38s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/e01319ce028843679100f0d5d515fedb : SUCCESS in 6m 35s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/cfda5f3530d04f72aa3e13e9ae7d925c : SUCCESS in 6m 41s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0a6cc1238c4c45f1970bd3ff55605370 : SUCCESS in 11m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b86dff4b1e7442eb9748ba8327601b31 : SUCCESS in 9m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8f11dfa5bff34338af7c751d84aa2cf3 : SUCCESS in 9m 02s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/dfff70a40748464c8e8eac72bc1f631a : SUCCESS in 13m 35s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/0701718ed46f4b53969e26698fe9c5c9 : SUCCESS in 5m 52s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/07b30de87c40457195da37089228182f : SUCCESS in 16m 20s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/3db1ae6e25544e8f98e69ab026f8ea47 : SUCCESS in 6m 30s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8b632377304c481da04ec85d6760c45d : SUCCESS in 23m 54s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5bbf12de502947a89fc2e2206b3612b0 : SUCCESS in 13m 26s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/128aaaa3610e4a07bccc6d5de6ad2fea : SUCCESS in 7m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/060c737ec9ac4419aa52032e57bf62c0 : SUCCESS in 4m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3afaa03a254444af8556659e45753042 : SUCCESS in 18m 19s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/84933949dc4449e88eb295403cc56fd0 : SUCCESS in 7m 25s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/83f50344c65a49ffb52275c2cc2caeca : SUCCESS in 7m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/327808e076784156b9cfe04fa28fd872 : SUCCESS in 8m 05s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/bfa5d8d1aae140b1b4b63e4c1b96ae94 : SUCCESS in 13m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/be16d8375bfe48db8da2d302b66e0111 : SUCCESS in 46m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7cd892ffe62e4fd3974ab0d8b936c014 : SUCCESS in 1h 24m 41s","accounts_in_message":[],"_revision_number":1},{"id":"d0b209f76e9c97f7447f090d4d40deb532e9f5ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 20:51:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"24571ee16cf22dc60db9d8e391b76bb39cb398de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 20:54:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a682dc9fa2446f18c0134d98a40bd93 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e086e8ed224f4ac4a5fb98ea45e9ddcf1840f78d","revisions":{"e086e8ed224f4ac4a5fb98ea45e9ddcf1840f78d":{"kind":"REWORK","_number":1,"created":"2022-02-18 15:56:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/36/829936/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/829936/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/829936/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/829936/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/829936/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/829936/1"}}},"commit":{"parents":[{"commit":"189eaafac6c0707762dd258a09f96bb2fb3cfd68","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/189eaafac6c0707762dd258a09f96bb2fb3cfd68"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-18 15:56:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-18 15:56:27.000000000","tz":0},"subject":"update constraint for heat-translator to new release 2.5.0","message":"update constraint for heat-translator to new release 2.5.0\n\nmeta: version: 2.5.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: If564df0926f52d6322ad423cc881faafef2dce3b\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ie5969c24c0acd69a6a5c84854c288eb544d9a0f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e086e8ed224f4ac4a5fb98ea45e9ddcf1840f78d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e086e8ed224f4ac4a5fb98ea45e9ddcf1840f78d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
