)]}'
{"id":"openstack%2Frequirements~829938","triplet_id":"openstack%2Frequirements~master~I706b360820c297082559b85179c53dd1b862df5b","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I706b360820c297082559b85179c53dd1b862df5b","subject":"update constraint for monasca-common to new release 3.5.0","status":"MERGED","created":"2022-02-18 15:59:47.000000000","updated":"2022-02-18 21:04:35.000000000","submitted":"2022-02-18 21:03:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829938-new-release","meta_rev_id":"d79276a202537d7b0e90526b15cb8a67569a35c5","_number":829938,"virtual_id_number":829938,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-18 21:03:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-18 19:21:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-18 19:21:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-18 17:02:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-18 17:20:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-18 19:21:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"cc260ccde5588bd7f86e3cb7265f92e186db4f62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-18 15:59:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e023a29ab391157e2deebaefbe8313349ec08635","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 17:02:38.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/372359a8ab544cb1b7853d6e260b09b7 : FAILURE in 16m 23s","accounts_in_message":[],"_revision_number":1},{"id":"3abbf25e57abeceab88da9eb612d4946996e9132","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 17:20:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2254c9026038459fa597891d6c454363 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1173a78d240f4711b53f33172eb250bb : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44cdbc4218df4ee28796d2b19da91fa3 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3157bdf7b3946659d1d7538e55ff67b : SUCCESS in 6m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3a83e93a4eea4746b175ffb13a8267f0 : SUCCESS in 5m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6088d4c953ca44979e833f23fd52dcfc : SUCCESS in 5m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/744385113429413992dad6832582c007 : SUCCESS in 13m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fdc838ab00db4377b40e930956a2c2cb : SUCCESS in 12m 36s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/24d44ad3d79d4981a82a94a3118d3fe4 : SUCCESS in 6m 24s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7fce5cfe849e4542a66c1b4c4483ff0b : SUCCESS in 10m 40s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/000611b5ba1547eb8cbe8445bf9b91d1 : SUCCESS in 6m 33s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/82a8b5b6feee4a7b88e996f681b2858d : SUCCESS in 5m 29s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5b688edbe90b4a59b703005f23233c4a : SUCCESS in 11m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d15535a299a8495cb154187fcb121363 : SUCCESS in 8m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/df8ded4c40164f05b91526933688d444 : SUCCESS in 9m 04s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a1524d8205874893baea0e3171bc36b7 : SUCCESS in 13m 00s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/4696b5f179ec44628511db92aaade59b : SUCCESS in 5m 02s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ce45d9f9d72b40a0b84c611d64684d8a : SUCCESS in 16m 08s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/308c90a7b6294cceaa034dee0fe0006b : SUCCESS in 6m 10s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/fbd0641d797d49c8a66322ee8e75ddc6 : SUCCESS in 21m 31s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/98e1c956d4db493d9473a349ff37297e : SUCCESS in 13m 14s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0c767ec359004ecbae523dce71b7619f : SUCCESS in 6m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/34a77c91e8274191afcb880b3b5705a3 : SUCCESS in 5m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/041e7db8db3f4dfaab82e24e976cb61f : SUCCESS in 23m 40s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4553c1da86e148f094793767c38843fc : SUCCESS in 7m 31s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9b374dd0361a490bbbfcdc85024d0765 : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d666aa64bae24293aef39706705dcd6f : SUCCESS in 7m 33s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/00186b9e34214cf1a3bf8e25fca08f32 : SUCCESS in 9m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/069a2ed7f1b54bc2bfad5127f1529b5c : SUCCESS in 52m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e79d8a23d83f490fb70d8a47b4f58b0c : SUCCESS in 1h 17m 43s","accounts_in_message":[],"_revision_number":1},{"id":"c3db0c7b377d9ce119e9d3ab6821d1d6176563e9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-18 19:21:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dedd7c591c3734ddbb39eda573696e7a6c87b15f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 19:24:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"18ab2cd9aa54254f938074e827f8a53a863687e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 21:03:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f1df16467584ddaba85d1a5f9274c88 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c86ec5ddc15421196a392fc0bc21534 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96c62299ef754f2ba06d18229a25a81b : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a3b9b3d387a465db846f73e07e77499 : SUCCESS in 5m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/631c62e5549146eaa633dc310192bc80 : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/95978c52691f4737adce2a7dbf20c92b : SUCCESS in 4m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5fd1cccdba5d4145b291597753ad2273 : SUCCESS in 11m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f7fd66b056d44d31b6205b27d837febb : SUCCESS in 11m 32s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/1a4cce1810e345ddaf1d0ca6b4164632 : SUCCESS in 9m 48s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/b27c4c2ddc194417947f659c5ba65870 : SUCCESS in 5m 59s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e34c297c4bb64cad85879eff789ecfae : SUCCESS in 5m 44s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ec72422af25f43fea89a542ee9fa8b2f : SUCCESS in 11m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7f28f9b6134945f3b246a5d9bec5c87c : SUCCESS in 8m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/800031578d9c47b7b57a5e3533cc7668 : SUCCESS in 9m 24s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7d8ec4c141fc47afbf154fccd01cb4c8 : SUCCESS in 10m 09s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/63e0760c0c0f4e2cb7eace51a196662e : SUCCESS in 5m 21s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c6771eb35ea34d9ba4e7feabc31dd41f : SUCCESS in 17m 35s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/18b07166ae8c4b36a2f5f559f8f46909 : SUCCESS in 6m 04s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/81f25d52c1a242719ea7060e2db4ccfb : SUCCESS in 26m 05s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2f65abb301ab4d508ffb72bc7a488d03 : SUCCESS in 11m 26s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2742cff2fe0c4ca8b7a1fd35fb95e1e8 : SUCCESS in 6m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f76100c4a53c4bf59967a39dc2edde97 : SUCCESS in 4m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2fecfc295f8948c9834580693923b9cf : SUCCESS in 18m 15s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f0682895ceef45d0970a73931e242682 : SUCCESS in 6m 52s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9d3e95f99c374abdb341b1ee040ed2de : SUCCESS in 6m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7321403ed820408eaaa9991dcf14e068 : SUCCESS in 6m 25s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/fd28bfad6f1c4796a8517ef3c03bed8f : SUCCESS in 14m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/045262f3b0e3426c8ec98da2a9945e31 : SUCCESS in 38m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7b6675243e00497396b22c7988beec12 : SUCCESS in 1h 33m 59s","accounts_in_message":[],"_revision_number":1},{"id":"f84b7e1e99912870ef2b615edc84fff3f8e45009","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 21:03:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d79276a202537d7b0e90526b15cb8a67569a35c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 21:04:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd54f3f0aba6422aab9a11090eecf13b : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"732b93f7b2783d757862b2f1260e6f005d86481c","revisions":{"732b93f7b2783d757862b2f1260e6f005d86481c":{"kind":"REWORK","_number":1,"created":"2022-02-18 15:59:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/38/829938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/829938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/829938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/829938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/829938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/829938/1"}}},"commit":{"parents":[{"commit":"189eaafac6c0707762dd258a09f96bb2fb3cfd68","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/189eaafac6c0707762dd258a09f96bb2fb3cfd68"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-18 15:59:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-18 15:59:43.000000000","tz":0},"subject":"update constraint for monasca-common to new release 3.5.0","message":"update constraint for monasca-common to new release 3.5.0\n\nmeta: version: 3.5.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: I9560b9c6a1d5e5daf7eadfcd76e79fce7be26c40\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I706b360820c297082559b85179c53dd1b862df5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/732b93f7b2783d757862b2f1260e6f005d86481c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/732b93f7b2783d757862b2f1260e6f005d86481c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
