)]}'
{"id":"openstack%2Frequirements~829940","triplet_id":"openstack%2Frequirements~master~I3f5b44bafeeed2b7eef5885d66e16eaa275a397d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I3f5b44bafeeed2b7eef5885d66e16eaa275a397d","subject":"update constraint for monasca-statsd to new release 2.4.0","status":"MERGED","created":"2022-02-18 16:03:29.000000000","updated":"2022-02-27 01:37:19.000000000","submitted":"2022-02-27 01:35:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829940-new-release","meta_rev_id":"88269eea6b45b4ee0927e9c41dceac04d793ffac","_number":829940,"virtual_id_number":829940,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-27 01:35:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-26 23:54:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-26 23:54:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-18 17:05:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-18 17:45:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-18 19:21:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-02-26 23:54:00.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"669b50c5f3e178a37ee39e3bc3f6752587a80da3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-18 16:03:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2549fe5bbba88f4995290bad0b2544e528985de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 17:05:29.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0d5f9fad55184e21a466a3798bc7df0c : FAILURE in 18m 20s","accounts_in_message":[],"_revision_number":1},{"id":"ce412c950266c8703fc23bbc93bc48d906b135af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 17:45:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dec6347c6f34c319fb383a38847adb3 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/288d62e7d0524236b8f8ecd6f8f9bc33 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b212b9d873ee4219b0de1aedfa5fd40a : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d72de61a5b314f989dac4b05fb044997 : SUCCESS in 4m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f75f85674f874c1ebedecd2c897eda1c : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f57e8352f9b343b5b3000639396d8eb0 : SUCCESS in 4m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c4f3b5235b6945b3b13256856989d360 : FAILURE in 5m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c8fef038a53546f0b6ff7d763442e535 : FAILURE in 5m 21s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d49447b43c6e4d2eb5e6625a7b698de6 : SUCCESS in 6m 08s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6a1a33a3a5b844968d9f7a2b9e448b9f : SUCCESS in 10m 02s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/4b4c149dd09f49aaa307d8c9b2a864f5 : FAILURE in 3m 01s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/29e5819349f149678a3394fc2047ade8 : SUCCESS in 6m 11s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4b08ac1c4a2e4c0a850c504218f0425a : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b3870686e9344811a066a34d4cd959e5 : SUCCESS in 9m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/78979d21010043ba8abf62a9b8280fff : SUCCESS in 7m 31s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ee65ad23e3ce433c97feca895b8fb7c9 : SUCCESS in 12m 19s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ab5e28d4dc0843e2b0a1e2392e7a952d : SUCCESS in 4m 28s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6cda20b007e44338b617b50704c7bc7e : SUCCESS in 16m 36s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7c81a21740e44c098e223d57712e8440 : SUCCESS in 4m 59s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e02e16b13ea74b5da4fb7a07f60192cc : SUCCESS in 28m 04s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5c206c0e4f544aadaa98e75b5899aca4 : SUCCESS in 13m 14s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/8bad3299906c40faaadc6b98034491c5 : SUCCESS in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5348a6e1bdc644faa60e419274c1073e : SUCCESS in 3m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/29079de49fde457090cbbe463cfdf1f9 : SUCCESS in 22m 44s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/052af227ead64551846f097f564bb027 : SUCCESS in 6m 36s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/55caf4b0e5c143d18caef7a08b9c4a20 : SUCCESS in 5m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f48cc8daa46e41df9ccb821a0d8983a9 : SUCCESS in 7m 06s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c2e55c218b34496389f3012837bbf79d : SUCCESS in 12m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/604793b9a7d64cbeae32fbce1cd10c34 : SUCCESS in 50m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7b710d4e035941ce9ca0e90e5f81fa4e : SUCCESS in 1h 34m 23s","accounts_in_message":[],"_revision_number":1},{"id":"d10170e4713c229d0d073e2cdf761870b2613ece","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-18 19:21:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f62eb6772cbe73b420b062247ca7f7a2702857b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 19:46:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a7ea8dd6de554675a3feedb403455084 : FAILURE in 20m 56s","accounts_in_message":[],"_revision_number":1},{"id":"c206f8f1df4819b87fad86223ecfca3cd91fd2ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 20:53:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3d21cb78a57408492b1766274eecffd : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4300f24f62a482da3dcdf51d5b604a4 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b95db1aef65642cfb90e3994856264e0 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb5faf64fcc448b69ed81df54470c88f : SUCCESS in 4m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a5e8832d174f4e1a8a57bf1941084956 : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b6203a5c3a724fefbe7834de724d44d3 : SUCCESS in 4m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/469333e00cdf45d39a1ed5b83760ac40 : SUCCESS in 10m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3ff5576434f242de97538867e7745713 : SUCCESS in 12m 31s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e1666fa568dd41e8be1e441dbfa364ff : SUCCESS in 6m 04s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/bc53e983c3904a41a8b5d3e9ca97f87a : SUCCESS in 10m 05s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/8d0e3b58d40548578e58d5de262b8297 : SUCCESS in 5m 37s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8ca4537d32a2475bb5be7beb8f22a996 : SUCCESS in 6m 36s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f0f70b26f184498bb6cb4a3693eea1ff : SUCCESS in 8m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f465370d8f5e471480212e4c54786b89 : SUCCESS in 8m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c04ca0918dff4032a1ed022eece51b73 : SUCCESS in 10m 08s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d9da9826dadf4813bb90c7447ff853a3 : SUCCESS in 14m 39s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/33fe4f7001de4a09addde9e5bb8bcca8 : SUCCESS in 4m 34s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/48988a4f6e6740bf846eaa54b81340af : SUCCESS in 23m 12s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7151841412394d43b9ce1b4dcce4073e : SUCCESS in 5m 48s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6f6f25c1e96e436384c8727b8ab6dc1f : SUCCESS in 23m 20s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/da0cbc9b32f64153811ac94c79d6121a : SUCCESS in 11m 26s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/dc7f8ce654bc4acc8ea35c604c5a2cce : SUCCESS in 5m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0ec72a2b96ca4232a993aa46a865ee84 : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1fc083d960ea4bb7af50df7b5899ae09 : SUCCESS in 18m 13s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/17c1cf1e0b504465896cef6572da3655 : SUCCESS in 7m 03s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/39e6454b39b7466c94ead8e313d2e1a9 : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ae6b70d703ba4b948ad09a30ac1c6622 : SUCCESS in 6m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f0bf0ca3d45441aeb18faf7999f68fc2 : TIMED_OUT in 41m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/742aaf6a6786430ea76356b9edf1c2c4 : SUCCESS in 1h 02m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/073e7e376d7a4185b4a6bc774b8c3547 : SUCCESS in 1h 29m 19s","accounts_in_message":[],"_revision_number":1},{"id":"474f27c5d28a311c57204626fc44f75d60030d41","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-26 16:08:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ee5d187cd59ab31709cfa7c41f7666a317d990dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 16:24:37.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/16f8ff9eba02459dad1169a2925340b3 : FAILURE in 13m 52s","accounts_in_message":[],"_revision_number":1},{"id":"9d6f19ed832800f40af56eab7390300c11097c6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 17:28:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e814283c5f44cfdb47afe1e50cb0158 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/941d4c9309654dd998fabff0e1e9a17c : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb150f8947e34e67a4b4814053e41b18 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4edb9cb93f754e8dae3f10b0c1bffd82 : SUCCESS in 5m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5f80c252f2b14218a783a99941fe464a : SUCCESS in 5m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a00b6d30d2404686975c0ebf94385562 : SUCCESS in 4m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/705979bad20240aea1cfd66b289a88ad : SUCCESS in 8m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/492faa4d3d974a30878ea72562d6a9bb : SUCCESS in 10m 19s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d102939e63ef4071a2f4b21947db9b4b : SUCCESS in 6m 47s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5c7ea07260734d4d8ad748230c44d843 : SUCCESS in 12m 10s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/bea43e65f2da46178238e099385ae271 : SUCCESS in 6m 10s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/51db52c82a60496f93fa7cecaf22ac43 : SUCCESS in 7m 20s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/2eb6bc9506fd452d85adb793d379aa59 : SUCCESS in 9m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8240ffd7e2f24c7690dcdd21fce76a94 : SUCCESS in 11m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b3a11960a69f4ee39bd24c015e68121e : SUCCESS in 8m 44s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/565af75c5dc445648432e047a1b1dc07 : SUCCESS in 14m 58s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/5f665452ce714077ba507e1fd28b7211 : SUCCESS in 5m 48s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/18886d32618f4b11bb84dded63bc3f65 : SUCCESS in 9m 53s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/107bca0d838a4a2dace0c5b4dc85cece : SUCCESS in 7m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1ac92fe6de8b44c3b0717d4de06748e0 : SUCCESS in 20m 18s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/eca72d03bfad4405a448c390a015a554 : SUCCESS in 13m 52s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d5cc2564eb0f4b83a35667099d798fcf : SUCCESS in 6m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d7662c88d01f462a8473f93fba325f85 : SUCCESS in 5m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/16c1e338402c4370824529d59dc05f5a : SUCCESS in 18m 54s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c26a35760f3642318e10c4ac0cf93c98 : SUCCESS in 6m 47s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0e6bdd9e549143e9bdf3a910be98ed13 : SUCCESS in 7m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/34efc7f98a984fa8afaad568e27c7b6d : SUCCESS in 9m 46s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d141df59c4cd41199270e2c1e2a3937a : SUCCESS in 10m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b3bd55e4f0204376aea54983a572fb8a : SUCCESS in 37m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f7932bdedf4409c8c3036fe0c68899d : SUCCESS in 1h 18m 05s","accounts_in_message":[],"_revision_number":1},{"id":"c4980d4af9e567d5306abc552ad54407a5be83f4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-26 23:54:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f79e448c1736695ae949a095ae70bd5a2bb3ff78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 23:54:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"11d640473653d0909c77594f79e2123cae7e7478","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-27 01:35:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e8cff8c42c54e3098a8ba93a2544c67 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb56042cb2df41cdad8f4f28eb880b2f : SUCCESS in 7m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2eebffd25417484cbafad95690f4c849 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82ee0470466546b788bf6dbd05d0d2e0 : SUCCESS in 6m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a444f967860b4f4baddf4785c7986c33 : SUCCESS in 6m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3f4a0612502346738032987c324be765 : SUCCESS in 6m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b212ca9d250d481983724671dbe93f38 : SUCCESS in 18m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cec7d28a8c9b4da886e9a5b49b597b09 : SUCCESS in 12m 17s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/60b960f31933428ea92d2240ddb3f3cb : SUCCESS in 12m 27s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/098d2677368d4de68dfb556153c876f2 : SUCCESS in 7m 49s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c10b1294a22b4dbf86aaedfa928be204 : SUCCESS in 8m 15s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7e14c5debcbf4a84bb730511969dc254 : SUCCESS in 12m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/97f40dd6afe74bfbbd1c749d02fc372e : SUCCESS in 9m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5891b4f4d2914b438badcd6a501a8c8f : SUCCESS in 10m 44s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b195b48fc45247c785dbe42bed33f936 : SUCCESS in 13m 01s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b30d8891c6ad4878b09f8a9a98d9e6b2 : SUCCESS in 5m 32s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8fc4435eb44c419ca8544ef4565c78a2 : SUCCESS in 14m 02s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/17826f2cc9c545c099f0ef3b4dc22565 : SUCCESS in 9m 17s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/56b686129fe44c2ca6a0fcf1077ab6b3 : SUCCESS in 19m 01s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/cb42f95af370462cb07d59efa61c4795 : SUCCESS in 12m 25s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/b9455e51897e43979838e1cf46b37ade : SUCCESS in 8m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cef116ea8a4f4e7a9e9da82686b1f603 : SUCCESS in 7m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/41c537ab61b34a97a045282e47f658eb : SUCCESS in 22m 58s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/902dced591fc4f4ebdbdc507128619b2 : SUCCESS in 7m 05s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d7d959e9b0ed45f59516d4d83b6df67a : SUCCESS in 8m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3af42ff187ec4b3983d080f25d9a19e8 : SUCCESS in 9m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/dc06bb6b41134c55994330ca6db97049 : SUCCESS in 13m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82e3deb3a20349c78b430514edd9da32 : SUCCESS in 52m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7ac5c4a11f3e4f9795492920dc008137 : SUCCESS in 1h 35m 02s","accounts_in_message":[],"_revision_number":1},{"id":"1e86d3a2a1fe6d763f8c5a407f5320c181ecfe23","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-27 01:35:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"88269eea6b45b4ee0927e9c41dceac04d793ffac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-27 01:37:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/163379ed0a3e4967b6e3f2453fc3db6e : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"583aeb2ffaad1e73205c42a35f09611b6927be4b","revisions":{"583aeb2ffaad1e73205c42a35f09611b6927be4b":{"kind":"REWORK","_number":1,"created":"2022-02-18 16:03:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/829940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/829940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/829940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/829940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/829940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/829940/1"}}},"commit":{"parents":[{"commit":"189eaafac6c0707762dd258a09f96bb2fb3cfd68","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/189eaafac6c0707762dd258a09f96bb2fb3cfd68"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-18 16:03:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-18 16:03:27.000000000","tz":0},"subject":"update constraint for monasca-statsd to new release 2.4.0","message":"update constraint for monasca-statsd to new release 2.4.0\n\nmeta: version: 2.4.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: Ib67203e88646ea13dcdc67e07bc35bb747ea2957\nmeta: release:Code-Review+1: foundjem \u003cfoundjem@ieee.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I3f5b44bafeeed2b7eef5885d66e16eaa275a397d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/583aeb2ffaad1e73205c42a35f09611b6927be4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/583aeb2ffaad1e73205c42a35f09611b6927be4b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
