)]}'
{"id":"openstack%2Frequirements~830215","triplet_id":"openstack%2Frequirements~master~I25f6e9b09498638e1d8864a191a8a915c8689f23","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I25f6e9b09498638e1d8864a191a8a915c8689f23","subject":"update constraint for oslo.rootwrap to new release 6.3.1","status":"MERGED","created":"2022-02-21 15:42:24.000000000","updated":"2022-02-21 18:53:00.000000000","submitted":"2022-02-21 18:51:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830215-new-release","meta_rev_id":"47f4c8be05ae7bff143d7eecfc8787d4760cad7d","_number":830215,"virtual_id_number":830215,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-21 18:51:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-21 17:09:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-21 15:43:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-21 17:09:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-21 15:43:15.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-21 16:06:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-21 17:08:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-21 17:09:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"a5755f3fd8fa0c880e089c23756b1320546e3cc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-21 15:42:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f93927e41edc8ae5c65afc01502372573c9c0bef","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-21 15:43:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0959e18e8dd55e5a74af63afb81b9b88af9787e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 16:06:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/be865217935b4c3cafa17357b77ab1e0 : FAILURE in 13m 28s","accounts_in_message":[],"_revision_number":1},{"id":"31faf848fb52313ab28cd8abbc7580a3bc2c6076","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 17:08:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cd2737782b24bb5bf1fecd54f4c7535 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/174d4a9d033a44fe9e6ee41bdb7c43fe : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d043e3a78a5a43bfbcc711f77b04ddee : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82de5d3c62b643dfab530516ad06f3e3 : SUCCESS in 5m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3d7e89aa71364cb48b14009c86a3c612 : SUCCESS in 4m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c8b920c5a4df47b599ecff1df76dd02e : SUCCESS in 5m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fc9f4c30d2b74604a1931e711aacac8b : SUCCESS in 10m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2c246e144e754e168216a749320e018a : SUCCESS in 11m 07s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/28fbd54097d249b489c36ad94041e996 : SUCCESS in 6m 39s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e4ef43cc9ccd412e903be5bb31da7f40 : SUCCESS in 10m 56s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/e5d105178b4d44bba29c6ebe29d7c0d2 : SUCCESS in 5m 57s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2adb2df28d754037b7cadded24c3ef86 : SUCCESS in 6m 52s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a82ba0a1b09640abb0af6f9be835abbf : SUCCESS in 10m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/50ba8de1632a4903bacdaef00199fe26 : SUCCESS in 9m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/11aff0a009cd4ef698ee15b8162a307f : SUCCESS in 8m 59s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/913fbc5747d44bfa9af69a42b2148544 : SUCCESS in 13m 45s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/7602b1ff2e1147baa3f7299703b5949f : SUCCESS in 4m 42s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/34427e1befce4d9889963863a8f6ac4a : SUCCESS in 17m 48s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/041ca7567c0a4abe8b814dc01ee9b0fc : SUCCESS in 6m 08s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/739ef5bc5c384ca3b784d8e2db11835f : SUCCESS in 22m 04s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/51ce8f2bd3354813a62107ed1899beb8 : SUCCESS in 12m 30s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/131e32727dfc47bfb05fab33a8fc3ab8 : SUCCESS in 6m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/85598b35dbf74b47bbe4acca526d324d : SUCCESS in 4m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f14671e261164a70b4eda120d947297e : SUCCESS in 20m 04s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d71e094aee874a408a038ac67eece54e : SUCCESS in 7m 26s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/467ea6face9f4ab990a934cacc447170 : SUCCESS in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f96af0150bc44a0b97191ab3069f4d58 : SUCCESS in 7m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3ab58e3ae207467c8a3fa2c5004fe899 : SUCCESS in 11m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54dcec5540744cd88c2df09bbc8bb07f : SUCCESS in 41m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/652ad2704c044e778c0a76eb7fe99057 : SUCCESS in 1h 22m 28s","accounts_in_message":[],"_revision_number":1},{"id":"1edefda92ce5658af6f625b171a1abef9329eea7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-21 17:09:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"81ba17904675023d56f99e59e21aee2b6748f200","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 17:10:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c9acf66c99fcafcada9cc8919b6de7d983d55ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 18:51:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/400d3c5a49154034b3f23e428e54c641 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21a28150a74946eea794b3f27a2d1baf : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa6388efe79f4b9c996df2d9c6cff8c8 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da4585622ad74e5fbaa9291f5c22f214 : SUCCESS in 4m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/866ffddf57cb4044a9b1ae4513d6b494 : SUCCESS in 4m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7a5f0603197041ca86cbf4bcd4cea7f2 : SUCCESS in 3m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0e6f6b48387d47e5a8fffbede78189e8 : SUCCESS in 9m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/280fac3217ec464f80b2402a0b31de32 : SUCCESS in 9m 39s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/089460fd68f24ebcb02e3470f0df3343 : SUCCESS in 11m 52s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/656bc91dd72f49eb918b1a6d58adca56 : SUCCESS in 5m 15s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b4980c08061f4c4587aaef12c994bcfb : SUCCESS in 5m 38s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/950408b92eae41dfbbd877ad37ab6fd6 : SUCCESS in 10m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c4719c7291904783b38c004b33290417 : SUCCESS in 8m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3dfec61ea31c41ccbbc3759b875329b1 : SUCCESS in 10m 23s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1084146974e84966a6e64165bab7575a : SUCCESS in 13m 16s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c18f5562235242b3ba89816622a0bbb5 : SUCCESS in 6m 19s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1065ab35c5c34f32894bacf3172d6bfc : SUCCESS in 17m 36s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/05557167163545b2aad21457183b3808 : SUCCESS in 5m 45s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3b5a7c16b45f446d897af7898b500812 : SUCCESS in 21m 15s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a9890ebf422f4f42bd291ed5fe0a15fe : SUCCESS in 12m 30s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/cc710e90728644f0b9d19292008ffac3 : SUCCESS in 6m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/df22194b87ee467eab16582694437eea : SUCCESS in 4m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0c3ea809d6b14673841993f84e21e46c : SUCCESS in 20m 16s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b6bb81b9dd0144dea15ab4a3eab69b0c : SUCCESS in 7m 29s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/842773f3b7d545b3bd24ec7d76676c21 : SUCCESS in 6m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/568f8c11443a46fda13d847d1ba291b9 : SUCCESS in 7m 43s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d6045edba6514d72b889c7bb29f95352 : SUCCESS in 12m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4f5ee09975f4cb391025740c1428c6d : SUCCESS in 51m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/79ddee3ba4fb4320bfda9f817053c418 : SUCCESS in 1h 20m 18s","accounts_in_message":[],"_revision_number":1},{"id":"9a15604947d18e5c5e21f7733a856f2f9ecd5972","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 18:51:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"47f4c8be05ae7bff143d7eecfc8787d4760cad7d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 18:53:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7358d5cd07d4f2783b2e140adb6c5c4 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6356574641057e0caa2f780d6c78ff027db4fd84","revisions":{"6356574641057e0caa2f780d6c78ff027db4fd84":{"kind":"REWORK","_number":1,"created":"2022-02-21 15:42:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/830215/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/830215/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/830215/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/830215/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/830215/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/830215/1"}}},"commit":{"parents":[{"commit":"8d397aa2552234af7a11bcf180393e53d0172500","subject":"update constraint for cliff to new release 3.10.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d397aa2552234af7a11bcf180393e53d0172500"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-21 15:42:23.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-21 15:42:23.000000000","tz":0},"subject":"update constraint for oslo.rootwrap to new release 6.3.1","message":"update constraint for oslo.rootwrap to new release 6.3.1\n\nmeta: version: 6.3.1\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Douglas Viroel \u003cdviroel@redhat.com\u003e\nmeta: release:Commit: Douglas Viroel \u003cdviroel@redhat.com\u003e\nmeta: release:Change-Id: Idd2e8cf74a9fee48f900750f643eb92a69561f90\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I25f6e9b09498638e1d8864a191a8a915c8689f23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6356574641057e0caa2f780d6c78ff027db4fd84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6356574641057e0caa2f780d6c78ff027db4fd84"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
