)]}'
{"id":"openstack%2Frequirements~830224","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~Idb738bdd4c8f956c0149542186caf42c1857e578","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"Idb738bdd4c8f956c0149542186caf42c1857e578","subject":"update constraint for oslo.utils to new release 4.8.1","status":"MERGED","created":"2022-02-21 15:59:06.000000000","updated":"2022-02-21 19:44:22.000000000","submitted":"2022-02-21 19:44:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830224-new-release","meta_rev_id":"725d84eac503bfb93d315bb32c1bb0237225dacb","_number":830224,"virtual_id_number":830224,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-21 19:44:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-21 17:57:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-21 15:59:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-21 17:57:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-21 15:59:49.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-21 16:12:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-21 17:40:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-21 17:57:20.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ccf0400bd0ef9ee698f287b7a4257e145ba6536c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-21 15:59:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c7492b2dc752f363433341d71b1426aa7001b74","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-21 15:59:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"29607005cbf5dae07e28c24c3a8065181beabf6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 16:12:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a72cc44ef82f46709ce36b6e2c7d916a : FAILURE in 11m 10s","accounts_in_message":[],"_revision_number":1},{"id":"25fdccf420d347b5716b0da5ca6979cdf0f576a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 17:40:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/146cd81eee4a45019b6e002ab535cf03 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c69661b3775641f2ae02ab66fd8c9e1e : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f92cb60c6494345a86b92a65fe84baa : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94730199e2c24b9a9b05867399682160 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9dc7743035e34840ab9d49b9aabd7c74 : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/614dceac192d4ff49ceded3bb0199147 : SUCCESS in 4m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9041c312de2644b2b644c8783a53c9fd : SUCCESS in 9m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/58166fe6c8a34371af165606975db554 : SUCCESS in 11m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/91b2ffbbde184bfba9ff46908b67ae50 : SUCCESS in 11m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/babbfd2d84dc4f81ac1fcda9f70345b1 : SUCCESS in 6m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/60da5ddf619e48dd98dc55d77503e4f6 : SUCCESS in 12m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1ce77bb4f6d74794a036fe85b60ebbbb : SUCCESS in 9m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0d572680a77e477091ce3b0df783235d : SUCCESS in 8m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fdb7508e8c1c4a938c9e2e309b2731c4 : SUCCESS in 16m 04s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/2004ec411c9349ce8effecc519edb408 : SUCCESS in 5m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f744b807227e4a128b22ee4b32a628c4 : SUCCESS in 24m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/26d532db13f54cf28f8039d71fa8d83d : SUCCESS in 15m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d1bb5dcd8335456798081070d72d82c5 : SUCCESS in 4m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f2a2357e3558422f95a3c1c63ef18812 : SUCCESS in 22m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ba7f5db86b1849c28e0811d3f1b4019e : SUCCESS in 6m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/47771bd5b917467d9b0853f6dc70668c : SUCCESS in 7m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/be4c01a370c14a3e8522907e38ac0865 : SUCCESS in 12m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/29a827a6d42e4c7c8613adab1ff0133d : SUCCESS in 52m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c7ae57b4daf848eb9bb322dba8a35bd0 : SUCCESS in 1h 28m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a6789197118311dec0bb8dae60fff7c795706cc5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-21 17:57:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"121012ee44c71b9fe381b818a2324062c8fe5720","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 17:58:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4902a75032a97b0fec41f2f9566232f48a273754","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 19:44:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b6af1cbfd904283af2610b6a110df78 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46adcb96d8954d6ab6ba9783ab4e6ef9 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e72d924211044fe684960ff6f630faf4 : SUCCESS in 6m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/480ba8b43b43406eab6f99bfeeeb88dc : SUCCESS in 5m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/30e7075f45d944cd93cc13de7deb1739 : SUCCESS in 5m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0e4fbde64b3b4fbfbd5607a9872b1350 : SUCCESS in 10m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/39bcf2abc70f4137b9ec5518c0c4f961 : SUCCESS in 9m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/730de56b5954444caaa1eb551f1ef517 : SUCCESS in 11m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7e5693849b51445fa4fc0bf2312cd145 : SUCCESS in 6m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e5612386bd9840188d2f29ffddd37086 : SUCCESS in 11m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5351bd7759014b80a4bdc1b2ad86f33c : SUCCESS in 9m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fe3f0aebe99145f0a8719658229d1fc6 : SUCCESS in 7m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/abb08d9a52b048cc9f9cdc8fd20964e6 : SUCCESS in 14m 30s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/bd92c84eeac74d27acead6fed7a37f27 : SUCCESS in 5m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6af255919acd442f88b092e242f78c3f : SUCCESS in 35m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5476eaa86e7a4faa8d28e59782068dcb : SUCCESS in 15m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e872d6c33dc64a1b99ccce933528dcc4 : SUCCESS in 5m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1c0ca675316843f7a1c752010b7dc7d8 : SUCCESS in 19m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cef82eb1d92a4d93878b1161eed987d3 : SUCCESS in 5m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/49a4eac11d214a3789fbc791e5db848c : SUCCESS in 8m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b417ee2dfd7740c196feb02ded88105d : SUCCESS in 13m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5479b08cd98b46f786bccd308644ce3b : SUCCESS in 1h 00m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/959c1ad6e6a14f44aaac0ee8902a5d48 : SUCCESS in 1h 11m 37s","accounts_in_message":[],"_revision_number":1},{"id":"725d84eac503bfb93d315bb32c1bb0237225dacb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 19:44:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d141ae5509c80c8e5d44dc2a939d082db559b69","revisions":{"3d141ae5509c80c8e5d44dc2a939d082db559b69":{"kind":"REWORK","_number":1,"created":"2022-02-21 15:59:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/24/830224/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/24/830224/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/830224/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/830224/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/830224/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/24/830224/1"}}},"commit":{"parents":[{"commit":"266c0209939179dca1869a96201d72999c272eae","subject":"Merge \"Bump pyroute2 version to 0.6.5\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/266c0209939179dca1869a96201d72999c272eae"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-21 15:59:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-21 15:59:03.000000000","tz":0},"subject":"update constraint for oslo.utils to new release 4.8.1","message":"update constraint for oslo.utils to new release 4.8.1\n\nmeta: version: 4.8.1\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I7cf61d3851876359e7b7a412b13420168affe28f\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Idb738bdd4c8f956c0149542186caf42c1857e578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d141ae5509c80c8e5d44dc2a939d082db559b69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d141ae5509c80c8e5d44dc2a939d082db559b69"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
