)]}'
{"id":"openstack%2Frequirements~830346","triplet_id":"openstack%2Frequirements~master~Ib824125cd0a3ad921151e00b0d2abf8bf218d6b9","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib824125cd0a3ad921151e00b0d2abf8bf218d6b9","subject":"update constraint for aodhclient to new release 2.4.1","status":"MERGED","created":"2022-02-22 10:29:19.000000000","updated":"2022-02-22 19:14:13.000000000","submitted":"2022-02-22 19:12:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830346-new-release","meta_rev_id":"463af36cefa599e49057ea16856e0268a604b41f","_number":830346,"virtual_id_number":830346,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-22 19:12:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-22 17:23:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-22 11:05:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-22 17:23:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-22 10:48:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-22 11:05:28.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-22 12:07:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-22 17:23:19.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c0d55027a40eb74ea0ce6eea14d56157e1a53a60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-22 10:29:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"525373710edce7f67a5e935bbfdcc93af3060198","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 10:48:24.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/45463a5835e24281885fb6ac380128a2 : FAILURE in 16m 25s","accounts_in_message":[],"_revision_number":1},{"id":"81b8666cfa9a186040dd987627dc2cba28707c84","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-22 11:05:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"32617a3e85e94ee98dc9903a59dfd13bf87b693b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 12:07:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/209e1aa4daea4ff39b39dbcafcb659f4 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ee7ed35d8aa413bbd536a1d0b903620 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25bf21b7d6634f039939f40f77492df6 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c36bad3da6c48038f59634b1ab09072 : SUCCESS in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b44c8c2177d3401bb5f5801c09a66cbd : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8dffca64310c4b6aa85d5fe3a6a44238 : SUCCESS in 3m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fe0fa8a23c0048ee9a4f268780e18dba : SUCCESS in 9m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f7697fae419f49689c1c3e65905d13db : SUCCESS in 9m 29s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/6db98ee93e6f4a73872743737b1caa2a : SUCCESS in 6m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/62a2b734701d4d1cb08e7aff25742edd : SUCCESS in 9m 22s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/46e89465ac324a92b7e9d33fba3797ca : SUCCESS in 5m 31s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3c868ecfa59d478799f83cddf1a657b8 : SUCCESS in 6m 17s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a6700b62fd1c4e72863c1618ed745b98 : SUCCESS in 10m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9e182f93aa7c49bda8ed79d5df99aa8e : SUCCESS in 10m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/845fa0498ad249759d59ead146d38253 : SUCCESS in 8m 25s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/11e52009ae294ee8a51b2fb1f33ddc2a : SUCCESS in 11m 19s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a78be101312540009db6894254ac2102 : SUCCESS in 3m 43s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/25f403cc83644969921d042521b8cbcb : SUCCESS in 25m 46s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/80ff533080c4425497baa984b7700ce1 : SUCCESS in 4m 55s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/869d3de80f154875ae56e0ddd9a646f9 : SUCCESS in 27m 26s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/99323cdedf6f4bd681bc3ae7e52f9df9 : SUCCESS in 15m 13s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/702848cc8e42432c838cffd52bdee5b0 : SUCCESS in 5m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/80d4872f4c9c4505b326c384d7a2598b : SUCCESS in 4m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fbdeec6c80e9427393932a11231ccb58 : SUCCESS in 21m 48s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a849538ea8534ed9b6f83433bea34d90 : SUCCESS in 5m 37s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/dee19767d72b4bb687c480deb1bbbd41 : SUCCESS in 6m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/163a7dca9cac4557aa0165dd2228c1ae : SUCCESS in 7m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f1a854ea59cc40af832bc0a69360e23b : SUCCESS in 11m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31e2895416ce4d9fb6d23607e4695a51 : SUCCESS in 47m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/053db2da6c49404e89ae40992bc88d68 : SUCCESS in 1h 34m 57s","accounts_in_message":[],"_revision_number":1},{"id":"18b2daf489fa1b963f75620c101f23e018929890","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-22 17:23:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4441d5459b0918dfdd3c9d4d98bb081dc20728f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 17:25:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"282ba8ba548b63247c3d195cc85be5fa03de1e28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 19:12:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cce1382b9b764ccebe030e0340c8cb39 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3c5394b8cf94a41893f5381a972e60a : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/826b09136be04f71a796a8f7435836c9 : SUCCESS in 7m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd2bfaafffb84dd69a8e2522a4b9e7f9 : SUCCESS in 7m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f3f1905d2d6a4188bc76dffd918d063f : SUCCESS in 8m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4e739be817994f5c88d8c76e7b422c15 : SUCCESS in 7m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f00303c002cc459ba198ef83cf4546a5 : SUCCESS in 12m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1f936e1f501b4bb3be79eec5c30c934a : SUCCESS in 11m 22s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a2a303595d364bd38d6e8b620c47e257 : SUCCESS in 14m 10s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/f042caf76f7d4893abff41eac16cace2 : SUCCESS in 7m 27s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b6b8775ac0d34176a8871307ac1b4cac : SUCCESS in 9m 09s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/990b98f8487b4807ae1ff17e06d31a6f : SUCCESS in 13m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1c19a40d5baf47e59c5dbbb59812f5c0 : SUCCESS in 12m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0eb1cca498d445a6b6bb911aca935595 : SUCCESS in 10m 35s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/06c5f6cb278e47d282bde5c354fb5ba9 : SUCCESS in 21m 55s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/60482bd785204ab18d27e79df4eae115 : SUCCESS in 6m 28s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/16eef8d4a4d14182b890369400094341 : SUCCESS in 18m 57s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/462efadddf334542a985535917e176ae : SUCCESS in 7m 35s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4f74046ed3204357a83c8e7f63e885d4 : SUCCESS in 26m 39s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4f1b2ddd1b4a41eeb6699c6c2383ba8d : SUCCESS in 14m 58s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/250c53cf606642dcbc2a417b0788529f : SUCCESS in 7m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bd96198a87b640b292f5a0752044132f : SUCCESS in 6m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0b5664326d2d4153b562e6b8144ac942 : SUCCESS in 22m 38s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/680a74fe7c76477e841d7e11a8f24b3d : SUCCESS in 9m 31s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/18cb939428244819b99e00708d3f66f9 : SUCCESS in 9m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f344b99917e741b3b538dab77ca5fd0f : SUCCESS in 11m 13s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2660cfb1988b4f2580c23dce97ab8ea3 : SUCCESS in 14m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0798d54005554a9fad8879921611e8e1 : SUCCESS in 47m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8053b370d9dc42ad98529ae6672c645d : SUCCESS in 1h 43m 43s","accounts_in_message":[],"_revision_number":1},{"id":"5dd22bb99d3cc46a5388bfdad9e96f3b6e8c5384","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 19:12:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"463af36cefa599e49057ea16856e0268a604b41f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 19:14:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c8fd1f058d24adf9210322c9436ea24 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e17661c593052e81c4e18153737c20c0cb51ad27","revisions":{"e17661c593052e81c4e18153737c20c0cb51ad27":{"kind":"REWORK","_number":1,"created":"2022-02-22 10:29:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/830346/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/830346/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/830346/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/830346/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/830346/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/830346/1"}}},"commit":{"parents":[{"commit":"6356574641057e0caa2f780d6c78ff027db4fd84","subject":"update constraint for oslo.rootwrap to new release 6.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6356574641057e0caa2f780d6c78ff027db4fd84"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-22 10:29:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-22 10:29:16.000000000","tz":0},"subject":"update constraint for aodhclient to new release 2.4.1","message":"update constraint for aodhclient to new release 2.4.1\n\nmeta: version: 2.4.1\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I49686c62547461bb1bc480dadb1a261e98eecf42\nmeta: release:Code-Review+1: Matthias Runge \u003cmrunge@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ib824125cd0a3ad921151e00b0d2abf8bf218d6b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e17661c593052e81c4e18153737c20c0cb51ad27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e17661c593052e81c4e18153737c20c0cb51ad27"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
