)]}'
{"id":"openstack%2Frequirements~830379","triplet_id":"openstack%2Frequirements~master~Ic472ac593bdbc353f45b5e6bd325b1d63aa0643d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic472ac593bdbc353f45b5e6bd325b1d63aa0643d","subject":"update constraint for python-masakariclient to new release 7.2.0","status":"MERGED","created":"2022-02-22 11:34:32.000000000","updated":"2022-02-26 00:09:17.000000000","submitted":"2022-02-26 00:07:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830379-new-release","meta_rev_id":"fe5a85338b3be4f68fc1874ece490667f277485b","_number":830379,"virtual_id_number":830379,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-26 00:07:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-25 20:49:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-22 12:40:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-25 20:49:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-22 12:20:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-22 12:40:32.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-22 13:25:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-22 17:23:45.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-02-25 20:49:40.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"21b435c7659507df2471fb9b6545360b1ffbef33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-22 11:34:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5ae648074c57fa65f5e26f6a1fba615c69a91f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 12:20:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3982b24c735e4527b1d80f62db0c8f87 : FAILURE in 16m 03s","accounts_in_message":[],"_revision_number":1},{"id":"f3f1a8f543772e1eabdc32394b7d1d74ef8492a2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-22 12:40:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"693ae283497454a94990e38db843b1556c2458f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 13:25:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b5f582bfa354d229b85202fd9bff7b4 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c4cfb9c2d454cf59b1d0c79b033823f : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae5d734c32b448789ce49b11477edb5a : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07b80933788d4bb2ad3a06ba8e30464f : SUCCESS in 4m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3d399f702e8b4b45a39abe31014579d8 : SUCCESS in 6m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0616951de9424736b95d8452b5c1da06 : SUCCESS in 3m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6d72d72dac954671b99ee37b9bbfa6e5 : SUCCESS in 10m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/659ebff7bd924106aa14e286e8c7cf48 : SUCCESS in 11m 54s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/0824f4bc5afe41a588973769463804ef : SUCCESS in 6m 08s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/19a4fca5b0ad4d118902b82a8112077e : SUCCESS in 11m 10s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/cb279432d84148fa8c1e5c63c96db2b5 : SUCCESS in 5m 49s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/65a89b7e91e9452e82588c6912d162ea : SUCCESS in 5m 59s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/240b1258ffa94179b2eb4d7e490658a9 : SUCCESS in 9m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2a807f9fc36d45c3a0dff0f4ae90a23f : SUCCESS in 8m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ea14c4c50e474e3892269309158aea75 : SUCCESS in 11m 28s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ac2d25a7c9b641189c5cdb648e82f5aa : SUCCESS in 11m 37s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d7cedfeb4a124a509a56dc057a3716f5 : SUCCESS in 4m 10s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7f8e108aa33440909c68b325373a9481 : SUCCESS in 19m 42s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6a150ad89f544b89bd9d2fca49a42c41 : SUCCESS in 6m 14s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/bc0dfe5fee4147819bac6645336a4a77 : SUCCESS in 25m 28s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/60cd00170d274bfa9eeeec76805eee71 : SUCCESS in 13m 23s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/adc0f7c2fa1b4635b44364c773b29aaf : SUCCESS in 5m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/600fe8efb83d46cbbf51cb999a16a7ee : SUCCESS in 4m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cf07785b4a274bcab0519cd14a6e0925 : FAILURE in 22m 11s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/205882c04da448dab5e97998ddf761a1 : SUCCESS in 6m 05s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1a0046a7dcd44893bdb823382b8e34ee : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fbe352bac1e1417b8bc1b4e6654fbe6a : SUCCESS in 7m 14s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6848a1a408974ffa8179ec69d93ee02b : SUCCESS in 12m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e37cdbbd5c0b41738ee34bdde44cf775 : SUCCESS in 51m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/46596b6fe5694befa169b154ac8285b0 : SUCCESS in 1h 44m 13s","accounts_in_message":[],"_revision_number":1},{"id":"8a40ef16322fc960b36ab2382a37e7f4d6579fd4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-22 17:23:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"665f226fa35165a472200aa84e8e1760ac96a2fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 17:53:13.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e66479b80773429b99366651a10f2bc2 : FAILURE in 18m 03s","accounts_in_message":[],"_revision_number":1},{"id":"04d96aac9676b81d17e20542f57b0cc91c3a100d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 19:30:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bb1031b60e4495f846c3fe1a69a01da : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05d865ab6858407598ebce7a8a701608 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70b9a9ddca8b495a9c6b83d48b3a2c7c : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af9b116cbb1a41ec8cd12a602282c233 : SUCCESS in 4m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/69a7ace8f236414698b495afb403ad7f : SUCCESS in 4m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d0e1dffa40164811939007803d3b6cfc : SUCCESS in 4m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dfdbfbaabb3847139457b1e648271bbe : SUCCESS in 9m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bad50d150d404a7b97d1ca28c86d7021 : SUCCESS in 9m 48s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/38434a01b4a14e34a76c9dfc3699c26c : SUCCESS in 7m 05s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c9f5150d2e6a48e9807b1dcf2cdbb975 : SUCCESS in 10m 42s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/249e2ac3ad944cff87664c6f207fd193 : SUCCESS in 6m 01s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4168cbc5e41c46aeb6db14518939ee08 : SUCCESS in 7m 08s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/901f6327da584426b13b71207407f665 : SUCCESS in 11m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fa83f2a4399a4a1bbc1d13cd0cf5b30c : SUCCESS in 11m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a821c402278b4f4d8facf75681bdf071 : SUCCESS in 10m 39s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ddafa722638b47ecba48b6f016e24d6f : SUCCESS in 14m 45s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b70624014eb7461e8d275f0d325a3701 : SUCCESS in 5m 21s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/488171b4fe0c483689ae6a0a2768ced2 : SUCCESS in 18m 14s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a765b328c52b4b2388a2fb7bdd03b36d : SUCCESS in 6m 44s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d38c6ee4b2384dc0af5e5ff4db78350f : SUCCESS in 27m 50s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b174f42d397846009f38c87f6126e5c3 : SUCCESS in 14m 09s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0faafa5ae2b24814a111e0c630d33cb2 : SUCCESS in 6m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/39e9e8fbff74470ebcb78a77e5c11246 : SUCCESS in 5m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8e05dfd39eec44ffbcc7423e78c2457c : SUCCESS in 20m 54s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f40385450eb24d568de70bfec2d8976c : SUCCESS in 8m 04s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e0b3dde9766a46b8bf4857e395acb57c : SUCCESS in 6m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a86b78c3548e44abbd88d55616ab9710 : SUCCESS in 7m 53s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a0b8282be68145ef8678ae5002ab0d58 : TIMED_OUT in 41m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c68f25deb154d0993d12f5a369e7309 : SUCCESS in 40m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2d96e9fea5db488aba858253fd1b2dd8 : SUCCESS in 1h 36m 12s","accounts_in_message":[],"_revision_number":1},{"id":"99d37470d61a146460628d7f2e076509ffb27410","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-22 19:49:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"31994f4204a528589243d3b89922c90b134296ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 20:11:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/db674065fd3d4ef7bdd2616a6b244627 : FAILURE in 13m 41s","accounts_in_message":[],"_revision_number":1},{"id":"61a3405f8de2f4fb015bf9a19da2cba47ed1adea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 20:28:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15bcfc5bacd34a34a6fb05933db28351 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19eda549a26546638dbd7af30eb05c77 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e7bb26d9bb3492eb3e0855d3deb8f63 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d2c5319ce2a4939be192b2261bf2d89 : SUCCESS in 5m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d9d10803a1a747eabc05445c03a60e00 : SUCCESS in 5m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0b92bb0b65c84634bd8e062970e74782 : SUCCESS in 5m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/720d2b655f8244638385ea0647a97d0c : SUCCESS in 10m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/168ba29cc1774d06830acf44effb7850 : SUCCESS in 9m 41s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e087febbcabd4253b8d8fc36761ad252 : SUCCESS in 6m 57s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4056918f0aa740c9861e0f47d613d40b : SUCCESS in 10m 41s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/00db199530614b0193debdbd198a2ecb : SUCCESS in 6m 11s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7fca0727c7f94fbf86125ad53114c8a3 : SUCCESS in 6m 57s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3aa8b0de371245e98775e5c6b1c3ce87 : SUCCESS in 11m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bc851bae0ce44faabbcf8156be32abe0 : SUCCESS in 8m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2ca4b2d0bf4b44a38317d7bf4a22c135 : SUCCESS in 9m 24s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/3a32a9f1af7f4faca6dc442f8f1dfb07 : SUCCESS in 11m 51s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/0f60dc5992f74442a98a5fd03a47cca4 : SUCCESS in 5m 26s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b5df7fde5e444982849e5367bf74fddb : SUCCESS in 18m 38s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b48474d50489432787ada783abeec975 : SUCCESS in 5m 51s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/39bdd23fd1a7487db3dd09349e73c550 : SUCCESS in 26m 11s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/477ef5dfb3014ea8be9b00280974651e : SUCCESS in 12m 32s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/1222d6920734475da028cd23916077bc : SUCCESS in 6m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8900aa8d7422468fbcfca6ad3891c1fe : SUCCESS in 5m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/482c66b2ad534841b6bebd0176bec88f : SUCCESS in 21m 43s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f94adb75972b40e2a1fe694e9c15ef58 : SUCCESS in 6m 38s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e2437670161243a1b73d29b5a7ebdb76 : SUCCESS in 5m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/62c03d306eed4b98a92696c321ae1b25 : SUCCESS in 6m 22s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1fe7bdf33e52451db56e42d5c573141e : SUCCESS in 12m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b084f320992d46808cf7cf10fe3ae4ef : SUCCESS in 34m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/68c099d9a05342e0adfb2df815c688e0 : FAILURE in 19m 02s","accounts_in_message":[],"_revision_number":1},{"id":"d6f9f44c21a5f46088788f43526c32ba6f996374","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-25 14:54:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a4b2e8ff63f0a6b6c0dcc9d452e160b48cb7e8cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 15:30:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6862763332f64255b8637de7f798e9a2 : FAILURE in 16m 28s","accounts_in_message":[],"_revision_number":1},{"id":"c1f3607428123cb6460a33633147b1b2dcb51d85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 17:02:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daf247747f86472b9b4e863cd31dbf3d : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c607ecca1e364959adc0283b4b2cb42e : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebef62894c954b728941eb54126ee928 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a2923a887a4463d9ac4165fde977be9 : SUCCESS in 8m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/098f5c9f3d234f139441f00c666643ba : SUCCESS in 5m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ca2ab1481195469ab1e2aa7e560851cd : SUCCESS in 4m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b69e53276ee94867a6fdb14f9aef6402 : SUCCESS in 12m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a6ed2417fb2e4072a49955d924b158a9 : SUCCESS in 9m 32s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/2c7f2a79f8034010a8c5e60812785586 : SUCCESS in 7m 51s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c9fbab11b72c4ab7aca29e01fa0b67fc : SUCCESS in 12m 10s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/4c6f5b7aafec4340a4ffb0bab0af0ff1 : SUCCESS in 5m 57s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/bacf783e380e44d4811218d5ad222d0e : SUCCESS in 8m 11s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1c18462e7a704503aabfd22421babd75 : SUCCESS in 11m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b7d235cc4d13445bb4e4ad84de0ada53 : SUCCESS in 10m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/edea1b5ec3dd4129b20f452fc8a856c3 : SUCCESS in 8m 53s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/df538d95d23d4bcebeea1abcc1f5f72e : SUCCESS in 16m 08s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/3f954cc1926241c99ad146e3c645b60c : SUCCESS in 5m 24s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/64af63e571314882a77e448ec4eb55f9 : SUCCESS in 17m 15s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a8b951b6c10444ce87b664fccefb5cdf : SUCCESS in 8m 14s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2d5aee90dcfb451cb2928cf2e929578c : SUCCESS in 21m 54s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/14e4752f348c41b2aa97a8e5a4ef103c : SUCCESS in 15m 45s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d6848a4e97634505b0f21bc6b499a3c7 : SUCCESS in 8m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2997d43580644da98181b33e4145a1bc : SUCCESS in 5m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4fefd4887bb34ef1aee7bc06281714ab : SUCCESS in 25m 23s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7682176ba78a4d54bbef541cec55a214 : SUCCESS in 7m 59s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c75fda74a3e64184b91317162ffe1b48 : SUCCESS in 8m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7fa2cc77d57a4319b4dc6204cfb3cf04 : SUCCESS in 9m 42s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6e635b6e4fb44f5d814a847b782e0b67 : SUCCESS in 13m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c51ef4a519c2480fa08adf23bdd87fa1 : SUCCESS in 40m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bec85f4c67a34aefac972c24722bee61 : SUCCESS in 1h 49m 11s","accounts_in_message":[],"_revision_number":1},{"id":"ae6cc355b966a47721f409a975c19eccb31db205","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-25 20:49:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1b6dbad432bb7125309d70e1ab2b0dac1b4fbe57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 20:52:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6fe44a830af36ef93352bf9f69a00c247d66c301","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 00:07:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21caf889169e47fc8fcaca13c0283c07 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fdccea54edb4e6ab75568406fb8ceb6 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24b25d30248c4dd0be9b6974f644d26c : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e7d7d307a06428381629cd9738d18a9 : SUCCESS in 5m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5161bd32d69a4dcbb8558d2fc8279b6b : SUCCESS in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4aefa793e53d47c783209a553076fbcc : SUCCESS in 4m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9a3a279a4d2f437ebf1d7f41305c77e2 : SUCCESS in 9m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cb19fd09086d4241a017d7f8ca3d4bde : SUCCESS in 10m 58s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/fce71cda0f22446ebce91cc04b63b940 : SUCCESS in 10m 35s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/44410abb5d8645cf90e4bfeca977f25b : SUCCESS in 4m 34s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/bb7e8a99fc474d9b948992d39ffa7c9a : SUCCESS in 5m 28s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/04d978ef8d3945c385a9bc42538f94f1 : SUCCESS in 12m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/744bdf159fe642cba26d153a3bc90165 : SUCCESS in 8m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1da23c61eeb74c3e8971fb723c57b165 : SUCCESS in 9m 04s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4f8bb51b3b11485fa049ac2ff983a367 : SUCCESS in 13m 25s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/5bd9e1fbb46f43f28117fb7cd9aee78c : SUCCESS in 5m 23s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/915d9500baf64d5988110cca2b3481ff : SUCCESS in 16m 44s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2b4949385740422f8aabcf7e0db28487 : SUCCESS in 5m 36s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/dc4954ddb53b4a21855495554f551bd6 : SUCCESS in 25m 55s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ee74e326e13b434ea36ed5f33af86c93 : SUCCESS in 11m 30s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/9767b771b8be4889b61a187b0943fa1f : SUCCESS in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/96fc049e80a14389983681f6df3c286a : SUCCESS in 3m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/de12e745478a45e0a27a949322b6fa49 : SUCCESS in 19m 51s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/38fbfddbcf6345ce9b13749fd799bf8a : SUCCESS in 6m 55s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/05e66a14acee4861958197049bc5dc92 : SUCCESS in 6m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3350fa8563e648a0a7ed2569a4fb1fd4 : SUCCESS in 7m 47s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/67b5db64c9674da8b5f905236c61fe51 : SUCCESS in 11m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4eadc0f02daf40d3ab0fe99c0b609704 : SUCCESS in 48m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2f2a0ad90aec4bca982f57486af30632 : SUCCESS in 1h 19m 23s","accounts_in_message":[],"_revision_number":1},{"id":"102f9114e495fbf29f05703055881a3d60370e1d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 00:07:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fe5a85338b3be4f68fc1874ece490667f277485b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 00:09:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cdaba1fc2144312aabf6bd37288d3da : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"888c4191649311b7368b63c2aad2128b46ab89a1","revisions":{"888c4191649311b7368b63c2aad2128b46ab89a1":{"kind":"REWORK","_number":1,"created":"2022-02-22 11:34:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/79/830379/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/830379/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/830379/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/830379/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/830379/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/830379/1"}}},"commit":{"parents":[{"commit":"6356574641057e0caa2f780d6c78ff027db4fd84","subject":"update constraint for oslo.rootwrap to new release 6.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6356574641057e0caa2f780d6c78ff027db4fd84"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-22 11:34:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-22 11:34:28.000000000","tz":0},"subject":"update constraint for python-masakariclient to new release 7.2.0","message":"update constraint for python-masakariclient to new release 7.2.0\n\nmeta: version: 7.2.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Iad2dba6434b6d96075ddc6e3d9400dd434f02c65\nmeta: release:Code-Review+1: suzhengwei \u003csuzhengwei@inspur.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ic472ac593bdbc353f45b5e6bd325b1d63aa0643d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/888c4191649311b7368b63c2aad2128b46ab89a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/888c4191649311b7368b63c2aad2128b46ab89a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
