)]}'
{"id":"openstack%2Frequirements~830382","triplet_id":"openstack%2Frequirements~master~Ie3974b67f5fa1829647831e0ab9eaaa9b3cb161c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie3974b67f5fa1829647831e0ab9eaaa9b3cb161c","subject":"update constraint for python-solumclient to new release 3.5.0","status":"MERGED","created":"2022-02-22 11:39:57.000000000","updated":"2022-02-22 21:29:09.000000000","submitted":"2022-02-22 21:27:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830382-new-release","meta_rev_id":"f63fa94207848104877dc3dabd98cb8c884b83ae","_number":830382,"virtual_id_number":830382,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-22 21:27:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-22 19:48:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-22 12:41:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-22 19:48:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-22 12:41:39.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-22 12:53:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-22 13:04:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-22 17:23:47.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-02-22 19:48:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"de734c778ef0dc6af32bfd2a40da2567327298e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-22 11:39:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87fac7d17d11e584225f8aa2bd824a56fa89542a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-22 12:41:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"827e57d3923f4933466e8e293eb7d86db3e6f317","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 12:53:07.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/842cf2a24b74424d8546ac3a7bb55b20 : FAILURE in 20m 02s","accounts_in_message":[],"_revision_number":1},{"id":"3a9150c184416509e041af410f8d0d8fa248a723","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 13:04:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4861ae1568d44ca8a0bd7b011ee6a8b : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/059f6042761440d58d48d3fee9228fa0 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36e412f7a25f4c338370661149d611d7 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3755d5f23268463aba66f63fa6a463bc : SUCCESS in 4m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3efc65dacb7e4e43a2ee94d9ac8deda9 : SUCCESS in 4m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4f91309e55274dfc9dce85d7fee6911d : SUCCESS in 3m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6c29dd3fc28c4160b89988b06dc8f2e1 : SUCCESS in 9m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1fc32ec74a824ec9a8abe58abf7a829b : SUCCESS in 9m 48s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b5f539753fe549aebcfa877b478dfde9 : SUCCESS in 6m 07s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/06015bc0efab4be2adf5337df6f1072f : SUCCESS in 10m 20s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/6861ad4c863642369b35ac63127c516f : POST_FAILURE in 5m 36s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/856668a53f1b47b5ad15842525a51583 : SUCCESS in 6m 09s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/939d3908588747ea92f132d497a2e391 : SUCCESS in 9m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/20a573c0459441cba445191482321ca4 : SUCCESS in 9m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7bcc7ede5b3f4950b3ce22eb0e3739ab : SUCCESS in 7m 24s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/fbca8c5803b24b458758621cebfe0952 : SUCCESS in 12m 11s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a9bd72d611f1461eaa94eebfd53e2acc : SUCCESS in 4m 09s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6a5888f8085b4f4eba257a0c75b46ebe : SUCCESS in 17m 54s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/81591b01b658439fb103fbe6b83ee38a : SUCCESS in 6m 27s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/eeb7c7e114c844b7aefb2569a15f24ce : SUCCESS in 22m 59s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6bc25539c76b43cab081c003c99e0561 : SUCCESS in 11m 35s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/103870fcb89c4486a18079829f105642 : SUCCESS in 6m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b8c1a28daf3b4e23bc13e24dbef22c37 : SUCCESS in 4m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7992d264dc614c6baf7e71954204cea8 : SUCCESS in 21m 57s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c2b77771e1684683ae1ca6641b711b8d : SUCCESS in 6m 08s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/cba74f9b1fe44eb898638e382c5bf305 : SUCCESS in 6m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a89b52e6974d40fa855ef670161721dc : SUCCESS in 6m 21s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0d9bd21f6ad740e28c38a035779d8157 : SUCCESS in 12m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a6fc9dcf1e3441798787dceecce34057 : SUCCESS in 39m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e0e214d760ec440284f2637f6a687857 : FAILURE in 1h 16m 05s","accounts_in_message":[],"_revision_number":1},{"id":"f94f0ea3766a14786a3cd3c08594c2b2e5b7b81f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-22 17:23:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dca0cc5a8c6d91a46c22b5fc6ee95141b64eef4d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 17:55:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7b01ffe018f84d10b28ed11fe1953313 : FAILURE in 15m 07s","accounts_in_message":[],"_revision_number":1},{"id":"001379d7b649906496f9296d8c25254a32a76d16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 18:59:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a97be2cd4fea4235babf9164c718ee87 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff0b21ce0ec94f77943d5d5fb2236320 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be540256237b4796aa0665a78eb6a014 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7ea3e54cff34708808906f14476f4c8 : SUCCESS in 6m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/53037400a28b4885b4319211c9f83ee2 : SUCCESS in 5m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d4543623947842729e856e7ee239c8da : SUCCESS in 5m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/319643468266455e870256014c4e2748 : SUCCESS in 10m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7c80c7e67a024ac39585dbef78a21b16 : SUCCESS in 9m 50s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/3f4f76dda8724928a1af95f64235b11a : SUCCESS in 6m 51s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/473ab76a16b84ab59778cc7843df21a6 : SUCCESS in 12m 18s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/aec132dc18144c6a94ca46e6a4435df9 : SUCCESS in 6m 37s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/91f3eef7d9174be0b7cb72214efb8109 : SUCCESS in 7m 07s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/22ccaac93274438f99f51baa93c96641 : SUCCESS in 11m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/940365ca228142f98b4f61eee6a496d2 : SUCCESS in 7m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9ea9f462e98e4cd4bdcaff6a3f6f4a08 : SUCCESS in 8m 50s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9e96e84e4f3c4a3fade10defcee7d8bf : SUCCESS in 11m 27s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/9a57e9dbed734d2dafc7d5351c9dd013 : SUCCESS in 4m 37s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4d2de9f4d4184fbb9c89e9908a6e7c7b : SUCCESS in 24m 30s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/815afffab7924ca5bfa24f7ded2a0b09 : SUCCESS in 6m 34s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b4729475ed47439492f022ab61e14cc7 : SUCCESS in 22m 26s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/609bad51ea2644a586869de21c5fc72d : SUCCESS in 13m 44s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/69e5a908036e493289e92278ff317d49 : SUCCESS in 6m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1d65475eeab44466ac1084d727780523 : SUCCESS in 5m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6640b08de3f24d6abbf933ec704caa0f : SUCCESS in 19m 49s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4312fa86399b47c8a3f6aabccb388e69 : SUCCESS in 6m 31s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/eb0d5dbca6d14089a23182726e06c653 : SUCCESS in 6m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eb21e92b4e334e8ab05f473bf011bc38 : SUCCESS in 7m 29s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/282bf9f0214e4df2a8f5f0c78e567e95 : SUCCESS in 12m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed669c9a386f4fa0a398011d0bf26404 : SUCCESS in 36m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a0c2f015b74b4048ac04dc6db65f609b : SUCCESS in 1h 03m 14s","accounts_in_message":[],"_revision_number":1},{"id":"23cdef4c965e0ad72f0edaba36f41c82253bcff3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-22 19:48:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2f26b2bbad3068937aa01dd72a0b12e0d99e331f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 19:50:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"47f9555cf6e7d2fc1c5524452684736375930c04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 20:07:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f99c2d59fdff4a0185bfff53824d25fc : FAILURE in 15m 13s","accounts_in_message":[],"_revision_number":1},{"id":"c583518f16b617f059e0afbabac656ad7253ab1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 20:56:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1758cefe54e2457a8afabb0380c95bbd : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6e1da6be3c9456e9576d5bfbd738709 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51c4fab3c7054a45a6bc33438d78efb4 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c0a2c2e244c456184540f35d5d662fb : SUCCESS in 5m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f744bdfedd7f4c1bad0ae573ec3a59c5 : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5f7e5a04cbef4fc496c8a2fe25acef8c : SUCCESS in 4m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9e9800b7cd72483d9ca9694fefbd5c21 : SUCCESS in 10m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b13e8d7bfac34626a69b197a7a66fb64 : SUCCESS in 11m 41s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a08c256487d4441d8cd71469ed3097e0 : SUCCESS in 6m 32s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/84a6858519364e29a203d94c95face4f : SUCCESS in 9m 57s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/769fed15c4924182968a1f08a3c7b2f5 : SUCCESS in 5m 56s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5eb4b590ee5a46a98379f2c805b9dcb2 : SUCCESS in 6m 09s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9459ba90d84d49ebb5b56f79eb863a6e : SUCCESS in 11m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/39a0586651d3448387de9ab6b032a0c9 : SUCCESS in 9m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9179f4c516754d9f80d159a3e417688e : SUCCESS in 10m 07s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/48018f82152447bdb316e1cd460df8b7 : SUCCESS in 11m 29s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/3a8658b515ff48f597e3bc24b77c8b0a : SUCCESS in 5m 03s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4a04a244d7b246bca06d681ba8e16c52 : SUCCESS in 17m 15s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e88cda697e874b74b19d3b4f3d6c783c : SUCCESS in 6m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2a0ec6532fdc477998a97e3b78ae780b : SUCCESS in 24m 55s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6790ff786a12442e93fe092703bd6b3c : SUCCESS in 11m 57s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/b7e9434a6ac6452587caf33b1ea36b7e : SUCCESS in 6m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1e8bf09a22b349ac81b3a67d1f4c4aee : SUCCESS in 5m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f26bd0bc29af4eb4a4a8fffdc16953fc : SUCCESS in 18m 33s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7b3309e00e354baf83e021f1db7a8048 : SUCCESS in 6m 12s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3f94dde561ad41589236914ac66cd4e1 : SUCCESS in 6m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/04ec62994ba24f6dadb5ebc3385a8027 : SUCCESS in 8m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e4d8010dd3b645d090975e886d1622e6 : SUCCESS in 12m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/309399cc4918490698e58dcd2c0f2fda : POST_FAILURE in 20m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/24d9c6dbd3ee4494933b8bb05cea1ead : SUCCESS in 1h 03m 15s","accounts_in_message":[],"_revision_number":1},{"id":"b8152cb92ed23da16767b10e7b0a9d4105632b9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 21:27:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16fab63bf211445699043913a4a0b106 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e25ef3ee110f411f8c18e3cbe6d02a44 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a55d7a3df7b9411280fb0e2173c4ed23 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abb13412418f4a2889cb200e335aabfe : SUCCESS in 6m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c5597e58da9c4518af37799a2c3463db : SUCCESS in 5m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/16dd913833194d488911728623bdd9bb : SUCCESS in 4m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/923fcc5d4ba9457ba78a0aae99449cbd : SUCCESS in 10m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/99c94dfa859848df948ba87890afdff8 : SUCCESS in 10m 20s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a9692a2661a14bb7a273aae1b030b35b : SUCCESS in 11m 52s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/752fa45432484dfaab4a48e3167ac1e4 : SUCCESS in 5m 43s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f2eff9fd521d47acb9c7527a3f501244 : SUCCESS in 6m 59s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/58e39381c13d4bd6bb3db40745ea26d2 : SUCCESS in 13m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/45e34d83e46f46bfb90762e570e01c83 : SUCCESS in 10m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2f4b907395b041fc9bf34cee397a61b9 : SUCCESS in 9m 01s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c3798ce265e04ced88b4862413cb89ba : SUCCESS in 14m 13s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/cedfc57214ff46d5b3bcf2b4d7ed6f37 : SUCCESS in 6m 32s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/96e42b10adc646f3b2007174cbc25632 : SUCCESS in 16m 31s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/072610c1e4c146128f99b9c5c2a1fece : SUCCESS in 5m 44s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/468c32bac2024356be6e0e9a8e345a76 : SUCCESS in 25m 55s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/40f825fa42a040c2aafe4c4bc2c947b3 : SUCCESS in 13m 29s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/cb46181f4c8b40dda22dcd71b046fa53 : SUCCESS in 7m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3a74db911992490a933a3528f2cfa7ab : SUCCESS in 5m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5e14a573957d4be797ce7c5648ea5b7c : SUCCESS in 22m 20s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d8138ccf5e864139acc493a889aad28d : SUCCESS in 7m 19s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1de18cd1811e46f39ab185f12749a582 : SUCCESS in 7m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/58616a68203b42808a74191a4f3b7042 : SUCCESS in 8m 02s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6f3fe1eb988e4220801e8606cc54526a : SUCCESS in 13m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3acbdf8c5227410f8fbbb03b54bb0480 : SUCCESS in 41m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/42272df1f89c49acaa5cb763ceed397e : SUCCESS in 1h 34m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9e54f9fdfb85f117cc4f8d1b691fcf637519a76d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 21:27:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f63fa94207848104877dc3dabd98cb8c884b83ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 21:29:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ec1bcc49bc142c2aea54106614c7648 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c322e4e79d0c88f2df0bc6ab94b2a4e70b6358aa","revisions":{"c322e4e79d0c88f2df0bc6ab94b2a4e70b6358aa":{"kind":"REWORK","_number":1,"created":"2022-02-22 11:39:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/830382/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/830382/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/830382/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/830382/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/830382/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/830382/1"}}},"commit":{"parents":[{"commit":"6356574641057e0caa2f780d6c78ff027db4fd84","subject":"update constraint for oslo.rootwrap to new release 6.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6356574641057e0caa2f780d6c78ff027db4fd84"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-22 11:39:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-22 11:39:57.000000000","tz":0},"subject":"update constraint for python-solumclient to new release 3.5.0","message":"update constraint for python-solumclient to new release 3.5.0\n\nmeta: version: 3.5.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ia81730cd46f9515a81488ca4cc9d69f7ea1793f0\nmeta: release:Code-Review+1: zhurong \u003caaronzhu1121@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ie3974b67f5fa1829647831e0ab9eaaa9b3cb161c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c322e4e79d0c88f2df0bc6ab94b2a4e70b6358aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c322e4e79d0c88f2df0bc6ab94b2a4e70b6358aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
