)]}'
{"id":"openstack%2Frequirements~830385","triplet_id":"openstack%2Frequirements~master~I94550e6d622686883e6ebd609a7a890bcaa4a692","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I94550e6d622686883e6ebd609a7a890bcaa4a692","subject":"update constraint for python-cyborgclient to new release 1.7.0","status":"MERGED","created":"2022-02-22 11:42:51.000000000","updated":"2022-02-22 19:12:23.000000000","submitted":"2022-02-22 19:12:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830385-new-release","meta_rev_id":"5ae6aa73a8d2c5811234291d2137976b708a6b30","_number":830385,"virtual_id_number":830385,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-22 19:12:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-22 17:23:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-22 12:41:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-22 17:23:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-22 12:41:44.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-22 13:12:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-22 13:12:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-22 17:23:21.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"0e39b8fcfb0e2890dcc2c4d311e141b8d33065ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-22 11:42:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b995853f338772e98d10f78ee1ee9c05abdeb58d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-22 12:41:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"edb7826c429fcae67be568d9c53a283129a204ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 13:12:13.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/dda5a24ebe6343a18c973b4f5ec648f0 : FAILURE in 15m 38s","accounts_in_message":[],"_revision_number":1},{"id":"c2139f8ca269a9f5834f2edc5f1f8b691a0be056","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 13:12:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96370e0169bc4c33a722787a91b665af : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b66959aa8d944b4bfa8485278cce39d : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f94c44c54fc41c5b7f3e0b53c7bf8bb : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4054e26958d4378935852706f6d54e7 : SUCCESS in 5m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a337739369941cea25ed1170346e5a3 : SUCCESS in 5m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/806c2050e45e41abbc73fc09a858fb98 : SUCCESS in 5m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e500cda248f94cc1bd12886bf4e5fc02 : SUCCESS in 10m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/695284fcd9b8469cabac683dcb511f6c : SUCCESS in 11m 39s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/706f6cd27b2a4419b70ecc35eb73e008 : SUCCESS in 7m 28s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2325233922064e269f92449c70e6b67c : SUCCESS in 11m 23s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/def00cad12e44c33900913cae2e80a3a : SUCCESS in 5m 54s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2aaad31987614c47826fc1846a9aa5b7 : SUCCESS in 7m 05s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b34d6b54ba7348928062d308a84fffdf : SUCCESS in 10m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/667773529062411e9584bfa1a74f9061 : SUCCESS in 10m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c5b7d3878623482ab517c298eee18957 : SUCCESS in 9m 13s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0535165ad3c64134a57206edc1227a25 : SUCCESS in 12m 11s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/f9635847d49b40c095b05f082a4a2d4e : SUCCESS in 5m 20s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e873dbe4021c4fdba75e99ea5d73404b : SUCCESS in 18m 24s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1de3e6b31c4d4286b9ea86b9c3e0de1f : SUCCESS in 6m 38s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e52b0c0ea09049cb9142c216d474fd9b : SUCCESS in 21m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c80ee56610f44acb86b40b9a171300f1 : SUCCESS in 13m 13s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/5dcf1a193d1d4fda9bbd9d0e5e2fd5ef : SUCCESS in 7m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ec301aaa082241c086123d8dc6aead48 : SUCCESS in 4m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/07a56c464b5a4fb890a97ade6f50a819 : SUCCESS in 22m 10s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e2c761705b5e47d69ee0c45d24d8e90c : SUCCESS in 8m 04s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/16350238e87c4119b8cd491a04416603 : SUCCESS in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3dfca23e86cd4e86a92907e9fc8dce47 : SUCCESS in 8m 34s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/787286e1dd974afeb2a9f3a800fda24a : SUCCESS in 12m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71441434840649ed806bba32c0e8275c : SUCCESS in 53m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b3c9918cef3a4d99abe96fa778f2763e : SUCCESS in 1h 23m 10s","accounts_in_message":[],"_revision_number":1},{"id":"e4f83b73ac91551a0e18269cb7599d1bc54357fb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-22 17:23:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f3635dce9e42b4760c168718d0d2273ca54b8ce5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 17:28:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"064c120ed28a70c9fc8b0e6306f09178b7de2ad8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 19:12:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8baeac0ab2f4efeb9794e058e8b6f1b : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56cc48db8e0f43558f0f1c557a1d5733 : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1a01d4260af4ea3be2f37a4863f6ddb : SUCCESS in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/017942eb05734f3391c09f1af9529748 : SUCCESS in 6m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c39bfe17a3574583aea27918b94ec2be : SUCCESS in 6m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b5053198e6674ad295e335fd70a04ccb : SUCCESS in 6m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dc78ada833da4db4aa0b280fe36e6fa3 : SUCCESS in 12m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d9b0259c263d4a9986ad14e7521281a3 : SUCCESS in 10m 41s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/44a375094d6a4ff7bcac5cc84d7fdc5c : SUCCESS in 12m 54s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/682347d0e7bc4e09b9940b9de17e7585 : SUCCESS in 8m 00s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/bcc5059c2c3641a89bd9017166027ca1 : SUCCESS in 8m 01s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/50ccea41006242afba2659122f8a5494 : SUCCESS in 13m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bfedfa24d37e439889f18cbc710c4178 : SUCCESS in 11m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/697f46f5d7274d3b842f3229931fef37 : SUCCESS in 11m 12s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4ec562ab3a684649be9efa52fbe5422e : SUCCESS in 14m 01s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2422b1529abf493288d71d56c1d47e18 : SUCCESS in 6m 50s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5d38cad3dced457993357f7116c1d788 : SUCCESS in 26m 59s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a0d211dec6834cda8de58177d8e3ec06 : SUCCESS in 8m 39s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/03a5b063c2a9496dae732bee9299ee95 : SUCCESS in 25m 50s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4cee7cffa4d7447dba2bdd1776871691 : SUCCESS in 17m 19s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/5b8073e62f7047d5a871a94b1cfe2c1b : SUCCESS in 9m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5f5bc115cffb40dfa4028a9733a07440 : SUCCESS in 7m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cb22c36c68ac4a7b975b7c741ec45396 : SUCCESS in 23m 55s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1fa08bbf877144548e694f47931ef7ae : SUCCESS in 9m 44s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/eac801742efc4c9fbaf1fead8ca011c9 : SUCCESS in 8m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7c30245df40d452c84e669996274bcd0 : SUCCESS in 9m 47s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d36c50c174f0462a8691295d20ea8c99 : SUCCESS in 14m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c088fcb653d4fd5a89df1bbe3db7929 : SUCCESS in 51m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/70030113c22e4384b8508da8923f6ae1 : SUCCESS in 1h 28m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5ae6aa73a8d2c5811234291d2137976b708a6b30","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 19:12:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"75e7b1d5d5fdfe9df747bceea8d614c431c624d6","revisions":{"75e7b1d5d5fdfe9df747bceea8d614c431c624d6":{"kind":"REWORK","_number":1,"created":"2022-02-22 11:42:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/85/830385/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/830385/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/830385/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/830385/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/830385/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/830385/1"}}},"commit":{"parents":[{"commit":"6356574641057e0caa2f780d6c78ff027db4fd84","subject":"update constraint for oslo.rootwrap to new release 6.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6356574641057e0caa2f780d6c78ff027db4fd84"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-22 11:42:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-22 11:42:48.000000000","tz":0},"subject":"update constraint for python-cyborgclient to new release 1.7.0","message":"update constraint for python-cyborgclient to new release 1.7.0\n\nmeta: version: 1.7.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I0d29855785f747a5a573330ece4099edec32ed59\nmeta: release:Code-Review+1: Brin Zhang \u003czhangbailin@inspur.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I94550e6d622686883e6ebd609a7a890bcaa4a692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/75e7b1d5d5fdfe9df747bceea8d614c431c624d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/75e7b1d5d5fdfe9df747bceea8d614c431c624d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
