)]}'
{"id":"openstack%2Frequirements~830387","triplet_id":"openstack%2Frequirements~master~I766c86a6219564e0c467f74b39528b71b9d7a3b9","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I766c86a6219564e0c467f74b39528b71b9d7a3b9","subject":"update constraint for python-troveclient to new release 7.2.0","status":"MERGED","created":"2022-02-22 11:43:10.000000000","updated":"2022-02-22 21:35:04.000000000","submitted":"2022-02-22 21:33:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830387-new-release","meta_rev_id":"56cb3c96af4609149419b30f41723d6fba03fd8c","_number":830387,"virtual_id_number":830387,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-22 21:33:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-22 19:49:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-22 19:49:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-22 13:12:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-22 13:23:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-22 17:23:47.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-02-22 19:49:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"685fad06e9b1cccae5872414f8eb325aa7332600","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-22 11:43:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de33a114d8ae3dd14681871dc52b720ab750425e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 13:12:40.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5392a73764994a7f93f210437250d5f4 : FAILURE in 15m 29s","accounts_in_message":[],"_revision_number":1},{"id":"06a80acba201559d28be549ee1624e20c59d8ad8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 13:23:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c52d77b2fe664f76ab90bb2813474a24 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/598a060390ca4e84bb0e5ea0f1126e48 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3aef097993842f4925566c7dd39b2b8 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5e151436a1c40c1be3c6e5dc2829e37 : SUCCESS in 4m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/06e041f8e23a4697a0fe216cdcbdb550 : SUCCESS in 5m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7d76a9c8ef4343d7855a57b295eff5e2 : SUCCESS in 5m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2a2c3fd6c06243ae85ad4ff4cee879ae : SUCCESS in 9m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c0e0e71df53c44f6b9dcd45457a4f3af : SUCCESS in 10m 02s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/aa34878a13384687ab2fa5f36fb7b04a : SUCCESS in 6m 59s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/24629260a7e94876911bb407add9178c : SUCCESS in 11m 00s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/852eade1c36f4047afb8098973d8094a : SUCCESS in 6m 14s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a2ae99e6cda4468782dee9923b76b6e9 : SUCCESS in 8m 26s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/888bc1ef2b0441c5b9ed45c37c20d2c8 : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9fa6fd867e774972bfbe94370183056d : SUCCESS in 12m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/09bb29429a774474b56b96b0cdbcee05 : SUCCESS in 10m 45s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4a5e3aeeef8044cc9de947a99f27854e : SUCCESS in 17m 50s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/95ebf4653d9442b9b7c4c3aff32c3bfb : SUCCESS in 6m 11s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e0515a95313c4483b248d4d7c80f45cd : SUCCESS in 18m 34s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6a26d5f9d53744a995fddab3e05b7640 : SUCCESS in 6m 21s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/5b67b243c47a46629de9ec43c2466b8d : SUCCESS in 26m 00s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0b0ef853c1e84512b8ade773561c5c13 : SUCCESS in 13m 23s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/1d45cbba4ab346af8dfae22cc7933b14 : SUCCESS in 6m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d9d4a4802f10404eb958fe2d9711be14 : SUCCESS in 5m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4568bce2b8e6449bbb37201cbf69f645 : SUCCESS in 24m 09s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/aee329ffa6ea4155bfd356f81067e0bc : SUCCESS in 8m 39s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/fb9fdd91fca24d4385bcf7b5e3a6894d : SUCCESS in 6m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bb8854f061654c9a9cd94c910cbef1e7 : SUCCESS in 7m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/08d553fedaf445ffb0f5ee0311828db0 : TIMED_OUT in 41m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a3ac859924e42bdb10dbe6818ada2b1 : SUCCESS in 46m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d658b612ab254d6c9145e2c33a5dc083 : SUCCESS in 1h 30m 00s","accounts_in_message":[],"_revision_number":1},{"id":"46f190577237b0b0043dc9a54b819674f36a46e8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-22 17:23:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5949be1b94ae81f7611870fcca0a62aa7fb42fea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 17:53:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e35964c2d0c247f0b0a3bcdd41d38382 : FAILURE in 18m 25s","accounts_in_message":[],"_revision_number":1},{"id":"ee873e408d6abd93834c981cac1154b3d2712ad8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 19:19:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f322ea2447da483bb7a34e5715235312 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f50ec7be0514f17a32644aaff1d543a : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/971d43f052a84f379fb0f5cdc224cec7 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d612d384ece14a68b528e20c78fd7cbb : SUCCESS in 5m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4402755798f64aa5afa66f486efe374f : SUCCESS in 5m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/482a92bce2fa46538cbf7bdfad844aa8 : SUCCESS in 4m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f16f928cfd1c4f62881c0c446ac23fc3 : SUCCESS in 9m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ae060ea70a1440e1af9d618c0933e53c : SUCCESS in 10m 29s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/fcffa88ca6d94450b9542335563b1811 : SUCCESS in 7m 23s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/41487a027c394842aed86550f2ef226d : SUCCESS in 11m 29s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/7116e712b9fb4d718e3c34682e180167 : SUCCESS in 6m 52s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/400b51148e944cc4acc112084efcf8e4 : SUCCESS in 6m 04s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/38cc31922f664aae9a7ce46c5ecde421 : SUCCESS in 11m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/747a2507bed743c4b63c5977382af41a : SUCCESS in 8m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2db628cccfdc4d57a0c2ccf83c765983 : SUCCESS in 10m 27s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7240f61f6e054f07997b6dce3d72d9bd : SUCCESS in 13m 59s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/3638e0cae6ed4c46aeee9a7ca277e336 : SUCCESS in 5m 32s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8256633d8f1f4fd19283fba2250f5692 : SUCCESS in 17m 51s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/293c76d390f94a8fbd86d0a0b13e357e : SUCCESS in 7m 19s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c5bd18cb9cff48e38fa2bab56198fdb1 : SUCCESS in 27m 22s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/54afa1946db846f8b4f11b5de6e5cc1f : SUCCESS in 14m 58s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/b0fb539fa5b847f29a3697f43c28d32b : SUCCESS in 6m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/88c3e451e4b343509ecc1bb22d31c063 : SUCCESS in 5m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cbcdbf0f1b914ae1bca07f5c8a40f0b4 : SUCCESS in 19m 20s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/0b5d96e0111b4eab8f47581211c9f4ed : SUCCESS in 8m 41s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/544b826a0f76416bb6209329d59a9c5e : SUCCESS in 7m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/70cfe4e6f9c54dc2b9450fe107486dd3 : SUCCESS in 8m 21s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2ed8179b7bfb419f9677a035836432ae : SUCCESS in 15m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/50b78f190a9248a88a60270b8888a4ad : SUCCESS in 44m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/404951f89f3547c9a1f8517ac73f3248 : SUCCESS in 1h 21m 24s","accounts_in_message":[],"_revision_number":1},{"id":"4ac87030db291627b182f3278bbb571992fada29","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-22 19:49:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"25efcf90cd9d5a2fe85ef2b3b75fc567f2eaca33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 19:51:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1493e4b6612a12c241c518ca1fe423569520171c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 21:33:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8b0f842e1d140629f6d22e8203e479f : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52f935ebcbe94726b565d85e4bb34858 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75be0bb837f04bed98a00bc47cea75ab : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dcfc239ee88b4c47ab84fe2f19f935b6 : SUCCESS in 5m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/deca09bd6a374c56906f150dd462cb0c : SUCCESS in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2ca6bd6a33174eed831d6530efc716d0 : SUCCESS in 4m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ba8da7defb904905923fa2c2a6af84f3 : SUCCESS in 9m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1dc0427388ff45e1861c70a305758b4a : SUCCESS in 9m 34s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/85855cc0cf2445c6a036f1dc8ffb6785 : SUCCESS in 14m 05s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/740a798f5400400a95e9e72faa4686ae : SUCCESS in 5m 22s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9f06b1b6c3a84d28ac5f7645b8e6b54a : SUCCESS in 5m 46s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9d77091ac77c4f1faeebc900db8c0e57 : SUCCESS in 10m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/482e797bbbc64b5cafd0309cd16db3e7 : SUCCESS in 11m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cbbd6441760947b4b88abd863b6eb8c3 : SUCCESS in 11m 06s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5b14627e3be7472b8b3ad31c0968fe1f : SUCCESS in 14m 15s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/15cd53370eb24b6abdf907df191b8aa7 : SUCCESS in 6m 46s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0fff6ae77fb44036bd59c659724d62f5 : SUCCESS in 16m 22s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8db061b651ec4c67ac508140f2deffbc : SUCCESS in 5m 18s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1679374b56c94f529fc447a7b64cae3f : SUCCESS in 24m 33s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6a7141c4640847d5b26686fb9e5e4516 : SUCCESS in 13m 18s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ee7b7fa49b564a749a84643359de6139 : SUCCESS in 6m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/186f6e91e48d41de99f671731fa110a5 : SUCCESS in 4m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d586c9f5aa054d9d843c7b3b14489b8c : SUCCESS in 24m 33s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d3548912ea8748d38e244f110e0320d8 : SUCCESS in 6m 02s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4450629811ef408fbc066f41e02c0328 : SUCCESS in 6m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/533ba3ac2049450e8a0471f72fb75c14 : SUCCESS in 6m 49s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4d3ec452365d40e8864458055a7ab40d : SUCCESS in 13m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0f7d678b34d48c6a58fee9ffd4d7704 : SUCCESS in 45m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a8fd2afff8e46a3b321da63322aac78 : SUCCESS in 1h 13m 25s","accounts_in_message":[],"_revision_number":1},{"id":"bd382b2d3cb04eb83b84fdbc75e3c9ce0e9d1df7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 21:33:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"56cb3c96af4609149419b30f41723d6fba03fd8c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 21:35:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fbf370b40014d55bf65d1d92618296b : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7176d196ed4c1720d961a96fc34b29280ce025fb","revisions":{"7176d196ed4c1720d961a96fc34b29280ce025fb":{"kind":"REWORK","_number":1,"created":"2022-02-22 11:43:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/830387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/830387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/830387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/830387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/830387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/830387/1"}}},"commit":{"parents":[{"commit":"6356574641057e0caa2f780d6c78ff027db4fd84","subject":"update constraint for oslo.rootwrap to new release 6.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6356574641057e0caa2f780d6c78ff027db4fd84"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-22 11:43:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-22 11:43:08.000000000","tz":0},"subject":"update constraint for python-troveclient to new release 7.2.0","message":"update constraint for python-troveclient to new release 7.2.0\n\nmeta: version: 7.2.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I2c81c116dd357a155a6aa6e7d1c05e14c094cbf3\nmeta: release:Code-Review+1: Lingxian Kong \u003canlin.kong@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I766c86a6219564e0c467f74b39528b71b9d7a3b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7176d196ed4c1720d961a96fc34b29280ce025fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7176d196ed4c1720d961a96fc34b29280ce025fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
