)]}'
{"id":"openstack%2Frequirements~830395","triplet_id":"openstack%2Frequirements~master~I24644269c4c4063f0ff1e47c164218a0ad7402d7","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I24644269c4c4063f0ff1e47c164218a0ad7402d7","subject":"update constraint for python-octaviaclient to new release 2.5.0","status":"MERGED","created":"2022-02-22 11:57:19.000000000","updated":"2022-02-22 19:12:28.000000000","submitted":"2022-02-22 19:12:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830395-new-release","meta_rev_id":"67b3d246bd77cdef725597a75043addf12bedb38","_number":830395,"virtual_id_number":830395,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-22 19:12:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-22 17:23:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-22 12:42:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-22 17:23:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-22 12:42:05.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-22 13:22:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-22 17:23:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"10b3a74ad51e335545a2a0f3ebf5a5f9d2e520bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-22 11:57:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f44100414f6849c0f0cb23acaa246f8f3ea182e1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-22 12:42:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"78817bdcc0f0d9483ba21273b68ac9ff39c60660","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 13:22:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9dba270f53d480985324a865270e0c0 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/302701d67f0d40ed8a64c22a842b2ad1 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f837122849454a838c518ac9e5b2cb84 : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/230958bb304d41fa8480d09e5431a8cf : SUCCESS in 5m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/be455cf5f2944da790642e6221895a9d : SUCCESS in 4m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/96b1763c664d4a308571b89e095fab21 : SUCCESS in 4m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/18c018bde0ea4bcab5b707eca466bc3f : SUCCESS in 8m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/db3418184ca64ea784cca14cea7378ea : SUCCESS in 8m 03s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/9562e4e5daee4789832072f418e4caa6 : SUCCESS in 6m 58s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d224e71bb12f4fcf904315ee985891e1 : SUCCESS in 11m 32s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/3576baeb72f7407ca306bd9538c8aa54 : SUCCESS in 6m 24s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/19ba8720956f45d69ae3291246ee126e : SUCCESS in 7m 08s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/11b1931188b14cc79e2d20e33ffd3822 : SUCCESS in 9m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/82bff7ffbde746b190241b968867ac8b : SUCCESS in 9m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/974dcd3fd7c9467ea00a726bb743ffc4 : SUCCESS in 10m 32s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/3c159fbc1f09478e895be0c090849b59 : SUCCESS in 14m 16s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c6b8d01a818349359cae914079f4fa6d : SUCCESS in 4m 58s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d6d9ed2e83e74f8e90c90cdb6c9141b7 : SUCCESS in 17m 25s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7fa13ec1efe74867bc772ab23b12eb11 : SUCCESS in 6m 22s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/70e4bc83a01f4d4f86fd97be46109ecf : SUCCESS in 23m 08s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/24baccf4a0e1484d9e19b41430fc7588 : SUCCESS in 13m 45s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/7f2e02226dfb4a59bea2f4594b5be672 : SUCCESS in 7m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4c1ea3fac3994f2bbbecca0bfd8aab4f : SUCCESS in 5m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/011b6739d607421f9e9150a91b4bddf7 : SUCCESS in 21m 05s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/ee634cd9d5e6430189872fc1095863f7 : SUCCESS in 6m 22s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/acf9363261ff4142b03ea0a1a6cf3b74 : SUCCESS in 6m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0eda9b32b80849da8a8f1b4adbdd763b : SUCCESS in 7m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b542cde7beb14f74ae78bf6951675b38 : SUCCESS in 14m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3cd8fb87d7041a5b04454bb55fd15d6 : SUCCESS in 49m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9709d2dbc6e4c908ada6ae54b1ddc15 : SUCCESS in 1h 20m 44s","accounts_in_message":[],"_revision_number":1},{"id":"425dd2317c4fb9f73ecbbc40ac0c77faab78c181","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 13:31:45.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a729bb75dd3344a9abc69d5726f1b3e8 : FAILURE in 17m 08s","accounts_in_message":[],"_revision_number":1},{"id":"677cc478a40741f364a7afe5bf5505ca005e47ce","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-22 17:23:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9de8bc015f53262b1e7833748dd3cf1fc3d0007e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 17:26:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3d782016043600102727080b15bfa41effc77368","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 19:12:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f702a33467a4ce18b7af5c60ae51c24 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3798cb76791e46f0921b9e62828911d6 : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d0abe8573ed4701bfe0440ba4c843c5 : SUCCESS in 7m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95a35ede8aa94e369883c2ee5063c65f : SUCCESS in 8m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0fa53b126b5947fa9caf3331e3d6d4e9 : SUCCESS in 8m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/65dbd5cb11e04ecf8e39ed30e808ac9d : SUCCESS in 6m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a575aaf1d0fd4aefa665866005cfbc55 : SUCCESS in 11m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1de7d5c1a4ad4ccea4acbb52186e33d9 : SUCCESS in 17m 32s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8ae1960865754d5eb363256303a1741d : SUCCESS in 12m 47s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/80fc2d6efaed4f2cb93edcfd81f1070f : SUCCESS in 7m 45s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/29dd8b7bbc754941a4593fb130dff087 : SUCCESS in 8m 19s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/476cabb0c2264a3281516cee7ae86422 : SUCCESS in 14m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9c35212f7afb4633b14440a62005dd4f : SUCCESS in 10m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ccd30e5cb7d84973aa185a62c1b71632 : SUCCESS in 11m 28s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a94db033e470430590c7f87871bb27d7 : SUCCESS in 15m 16s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/1c5fb84b45624e0687b0c3aa2bdd9180 : SUCCESS in 7m 50s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5472564d36724d4c98215df9673aedc9 : SUCCESS in 19m 40s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/151e18faf1304081b5fa5792e60d038c : SUCCESS in 7m 53s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f7e9d2c7be4045a4910217a0ac339462 : SUCCESS in 28m 44s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d2eace6e267a40ffab1155df21659224 : SUCCESS in 16m 31s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4eeddf4cf5424aa1b40fcb13f89d8ae2 : SUCCESS in 8m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2b1e3c5104204355a5499f3858ecdaa6 : SUCCESS in 6m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f02119980536456da06f0591f989e262 : SUCCESS in 21m 47s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4119066140b54bc1803d3c2e37894474 : SUCCESS in 8m 04s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9b7611d8ed63400589776d1dcd3d50b5 : SUCCESS in 8m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/855039cd19834145b2767b019d449d12 : SUCCESS in 11m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7ecefa68fe024528b98add9b0267cdcd : SUCCESS in 14m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6755565521784aec9b16a6c1ba5fcb86 : SUCCESS in 47m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f5f2eae7eb94a42863489868bb1b817 : SUCCESS in 1h 20m 59s","accounts_in_message":[],"_revision_number":1},{"id":"67b3d246bd77cdef725597a75043addf12bedb38","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 19:12:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3a8d40173ac2c66df283da104e2d568f4a6a88f6","revisions":{"3a8d40173ac2c66df283da104e2d568f4a6a88f6":{"kind":"REWORK","_number":1,"created":"2022-02-22 11:57:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/95/830395/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/830395/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/830395/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/830395/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/830395/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/830395/1"}}},"commit":{"parents":[{"commit":"6356574641057e0caa2f780d6c78ff027db4fd84","subject":"update constraint for oslo.rootwrap to new release 6.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6356574641057e0caa2f780d6c78ff027db4fd84"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-22 11:57:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-22 11:57:16.000000000","tz":0},"subject":"update constraint for python-octaviaclient to new release 2.5.0","message":"update constraint for python-octaviaclient to new release 2.5.0\n\nmeta: version: 2.5.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta: release:Change-Id: If61e32afedeeb4c2e82314160b9f7c4e2bf4caac\nmeta: release:Code-Review+1: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta: release:Code-Review+1: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I24644269c4c4063f0ff1e47c164218a0ad7402d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3a8d40173ac2c66df283da104e2d568f4a6a88f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3a8d40173ac2c66df283da104e2d568f4a6a88f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
