)]}'
{"id":"openstack%2Frequirements~830442","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I478ddc6623ea0d7f0da7df4b6603f459f43c7226","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"I478ddc6623ea0d7f0da7df4b6603f459f43c7226","subject":"update constraint for oslo.privsep to new release 2.5.1","status":"MERGED","created":"2022-02-22 15:21:41.000000000","updated":"2022-02-22 19:32:00.000000000","submitted":"2022-02-22 19:32:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830442-new-release","meta_rev_id":"06426d1766e8cca0ea20c6425ed528af7518e347","_number":830442,"virtual_id_number":830442,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-22 19:32:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-22 17:23:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-22 15:51:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-22 17:23:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-22 15:48:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-22 15:51:18.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-22 17:01:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-22 17:23:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"a61fc11a2236c96866fc6bfe6486927c8fe0dc17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-22 15:21:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4de7c348b20db27cb37e6d43b6b53e8480b28dd7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 15:48:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/73a67aa739fa43ada031113166e78822 : FAILURE in 16m 31s","accounts_in_message":[],"_revision_number":1},{"id":"9cfc564406e059197e9fd983b52c56f01d50fb70","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-22 15:51:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"60aa3e48275da848630d8a7b11092468827f814d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 17:01:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd9998c7fd3f45b0a7eeeba101341587 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c6a2d69f18c41cbbdf47c583dbd92e8 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1b5dc33c6fc4ecaab5d83aab088097b : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2c9ed26f19a4f64ba077b393d0a6776 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d9bbf9d852a24ab480d9c3254ff8fb59 : SUCCESS in 4m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e5d50b113436483fadc3fce48b9b5a4c : SUCCESS in 5m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ecca3055e09448b894fd4225af88e3cb : SUCCESS in 11m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4545fc4c8c364c899e8f1dd3f836c28e : SUCCESS in 10m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a4bf85cce20147ec91e98c2662f79ed0 : SUCCESS in 14m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d74dc0f3daed430d9e45000466e97382 : SUCCESS in 7m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1c8c81f435d14a8e85ba32c76ccc8133 : SUCCESS in 11m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/44b4b57955c541699fc7cf3fe54a6686 : SUCCESS in 10m 00s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2e111919856742ea883f951e859ff743 : SUCCESS in 9m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3628a4d70cc74b9b91d6400b973b4782 : SUCCESS in 15m 03s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/e2b93ae514ea4dbe81ae418a49e7bb72 : SUCCESS in 6m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3d44ce789d164e808fcc9ccac0dd1805 : SUCCESS in 25m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5afe3de349104b09bf7db65fb7d284cd : SUCCESS in 13m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a0699015af54474b8ff8212b0c18b00e : SUCCESS in 6m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/925507d725664b1db7645e6f54d0b843 : SUCCESS in 21m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b8ddbd22d0124e2e80c00146ca6d6342 : SUCCESS in 7m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/24f972879b03475e912297efce86b23b : SUCCESS in 7m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4dfd06376d7346c3a98dd0c710b4caff : SUCCESS in 15m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb68224b628a4be6b37eb298c7ad2ecf : SUCCESS in 58m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1b41376ae82845daa755a10fc3393d6c : SUCCESS in 1h 26m 15s","accounts_in_message":[],"_revision_number":1},{"id":"c09e9b1ee440adfcb39273d5ad7c9d8b2de5975e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-22 17:23:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0229179e091a7a29c9bbbad60908ccc6f0cbeb67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 17:27:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bbe549d4bb83a1566316a68fd7daf987e27fdafd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 19:32:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c20fe6ed1314e1da29567da281af059 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bef77667911409e9374dd0c4f0d35f6 : SUCCESS in 8m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc63d79d8cb34952b05a0e5becd24450 : SUCCESS in 10m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3bcbdb462c714aba88392c82f777bda6 : SUCCESS in 8m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b6122deee32741c9ad285163ba6e5e64 : SUCCESS in 8m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a556e0bc95a94366ab2b068f3a58b53c : SUCCESS in 13m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/221f6c0a5e144cce9a885e1a4fd0263e : SUCCESS in 11m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b26cb1b31d09445bad79d58eaaf83cf1 : SUCCESS in 13m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fb609b6277994421af59ee130c264232 : SUCCESS in 7m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8c2a5a0df96b40c2a947bd35bf359b5f : SUCCESS in 17m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/17be6111cf694882a581e8a5f4d00944 : SUCCESS in 12m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/56cdd5c66a2d43e58a8acc48e639cfe0 : SUCCESS in 11m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/25e83229fb884d048cb5b4d2b5eeb355 : SUCCESS in 19m 37s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/80e3167fa8414c5ba8d3b4a344d48235 : SUCCESS in 8m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/653a8d8004da4c3cbd459996a4f7620f : SUCCESS in 27m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f954fa14edfe47bba670897cfa5d007a : SUCCESS in 17m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0d6d30ff9fe245a3978b24e2fe5092a6 : SUCCESS in 7m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5154e5ee7d1947189d2a86680d4a8bd9 : SUCCESS in 21m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1b99ec2d67a34e8ead3119ddb02d7346 : SUCCESS in 8m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/252f8c6ef3e84024bc6bda1e4cf20d4b : SUCCESS in 10m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ec59df0bb7034167a7e1cbc46a2a8293 : SUCCESS in 14m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8fa57b451ec41b9b57b6a1d199e447a : SUCCESS in 51m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9cc082b3068449d1949870c93f8de125 : SUCCESS in 1h 49m 45s","accounts_in_message":[],"_revision_number":1},{"id":"06426d1766e8cca0ea20c6425ed528af7518e347","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 19:32:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"79d78f2469f73ebb984da8bbe1fcbaadb6f7d311","revisions":{"79d78f2469f73ebb984da8bbe1fcbaadb6f7d311":{"kind":"REWORK","_number":1,"created":"2022-02-22 15:21:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/42/830442/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/42/830442/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/830442/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/830442/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/830442/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/42/830442/1"}}},"commit":{"parents":[{"commit":"3d141ae5509c80c8e5d44dc2a939d082db559b69","subject":"update constraint for oslo.utils to new release 4.8.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d141ae5509c80c8e5d44dc2a939d082db559b69"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-22 15:21:40.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-22 15:21:40.000000000","tz":0},"subject":"update constraint for oslo.privsep to new release 2.5.1","message":"update constraint for oslo.privsep to new release 2.5.1\n\nmeta: version: 2.5.1\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I920ee7bcca3cbf3e4051b3a3098beed53b1c969d\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I478ddc6623ea0d7f0da7df4b6603f459f43c7226\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/79d78f2469f73ebb984da8bbe1fcbaadb6f7d311"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/79d78f2469f73ebb984da8bbe1fcbaadb6f7d311"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
