)]}'
{"id":"openstack%2Frequirements~830443","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I16e2e10d0b595fdbe86e1fd3a1e14bbdb3078941","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"I16e2e10d0b595fdbe86e1fd3a1e14bbdb3078941","subject":"update constraint for automaton to new release 2.3.1","status":"MERGED","created":"2022-02-22 15:22:17.000000000","updated":"2022-02-25 20:36:22.000000000","submitted":"2022-02-25 20:36:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830443-new-release","meta_rev_id":"6b6cf20cbfc36cdef735ba2107d160368aa39d22","_number":830443,"virtual_id_number":830443,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-25 20:36:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-22 19:48:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-22 15:51:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-22 19:48:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-22 15:51:11.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-22 16:06:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-22 17:00:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-22 17:23:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-02-22 19:48:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"37fb18296aae33839e40c1aa36220f619de3a454","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-22 15:22:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3edf26112ff62334108f50fb4190a2620de7a174","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-22 15:51:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ec983a4047bb93eebc9c3c1ebc73b8cbde47c257","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 16:06:06.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b7c1bc77c4d04d5dbf386b42061ab7d0 : FAILURE in 15m 06s","accounts_in_message":[],"_revision_number":1},{"id":"fd08a46dd191e809af87d761c0047bf0bf453286","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 17:00:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82ff10da327a42d198e8ab62d4dea296 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04670ca94da347fc9d69e5f99d4f7908 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8db7497ea2364bcaaac003db989630a9 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7997b09894c42c4bbe2062311ff7c30 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ed7dd141b713448a86222bb835e93126 : SUCCESS in 5m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9dcfc9a39e4c447e8e920dc1c7470fd8 : SUCCESS in 5m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e0ebab5a403441dab14dc92a76ab701b : SUCCESS in 10m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/05be448452c2461abbf12f81fad75a9c : SUCCESS in 9m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/13a5e7cdb9b54bad95fadd5ffc4a3794 : SUCCESS in 11m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f0ce1d4fdc164d45a8e4c50ec98575e6 : SUCCESS in 7m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0ffe7ff06f654eea906335d279ad80e8 : SUCCESS in 10m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b2a231b953074685ac406e22a5af14f1 : SUCCESS in 11m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/557dd6b82e634405b0c838d82c025f9c : SUCCESS in 11m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/66283699826b424ea06360d8922d7b7a : SUCCESS in 22m 27s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/b352199859fd498daeab9a222ca58f1f : SUCCESS in 6m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d8e3285f77f649da835ddee0c0d3ff31 : SUCCESS in 30m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/40c5562cc33341128348d35322a87944 : SUCCESS in 15m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/eebfff1220564ae9a4d46f58e7309732 : SUCCESS in 5m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/33b92e22ac054bae8d995f2b99e9aa47 : SUCCESS in 27m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/786c8e5bc4914e9c8078599dcd9b82ee : SUCCESS in 6m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/869d945c190348fc912a4df834f55e7d : SUCCESS in 8m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c49cca846cda4a7794f1d2c1c5ef0e22 : SUCCESS in 15m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46289e742c9446b79171ca6d30d0ab67 : SUCCESS in 56m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5858b48e9103495bbaa14cab531be323 : FAILURE in 1h 19m 44s","accounts_in_message":[],"_revision_number":1},{"id":"62d32e1039848f40c736e0c49d49b9a970d4a41a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-22 17:23:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"63230d3f3a4b88e1dea7fc97b2198624ac9b523d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 17:53:02.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9b578d4ac3524bc8abfe0d35d219b54e : FAILURE in 17m 55s","accounts_in_message":[],"_revision_number":1},{"id":"90d8938f23a9bb882cdd10951680d197a45ec1f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 19:16:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d33d029fadda4bbe810b968aaa94e7b9 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09944ae2459b4cb4a08efc76f3bab1ea : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23ec3576778b44bd848b162a26c01257 : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c6c8599c4934817953dd8d695f5151b : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d587c18a73c04ae6819570643e138565 : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/abf50ac30aa04be4911dbf32a1f9f24a : SUCCESS in 5m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c21a94eef8ff463a9596ad1a6f031d82 : SUCCESS in 9m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6e7ed0d5c3bd44be93ce8128d6804823 : SUCCESS in 8m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cde86d7661284938a815b2126bb71b9c : SUCCESS in 10m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/402c8f35193940f0ae5d58faace21f38 : SUCCESS in 5m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/800bb72d365540768cd3585c7dc89f02 : SUCCESS in 11m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ef03f5fcd4b64c1cbf253293bc1c329c : SUCCESS in 8m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/595cb006e7de472883903a50addec884 : SUCCESS in 8m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/37adec39fb1d405d94537e72bb0acfac : SUCCESS in 22m 44s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/63bd25d034354aa7be4bdb58f5ca0ea1 : SUCCESS in 6m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/78dfde559e4c4c939d35e991fad854c0 : SUCCESS in 26m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ebf6946152a94ad5918ba6ec0f40aa55 : SUCCESS in 13m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ab5891dbf5954f918197cc34015cb141 : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e8219ad815b24c48a0b5f5e9e253af1e : SUCCESS in 19m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8ee8bb772a20406bb630501cbc2c15f5 : SUCCESS in 6m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/15dcb1b175e6446e89ddedd40c8aebbf : SUCCESS in 7m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a00f2be33972437083c9a0c8b8c27237 : SUCCESS in 14m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a02fe6d4a1c4bcb9a701f2460c55c80 : SUCCESS in 54m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f5962ba154f4a97bb57ec8612bbdfa5 : SUCCESS in 1h 17m 29s","accounts_in_message":[],"_revision_number":1},{"id":"4df6085d195d5e58d447a7dbababc06f30281da3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-22 19:48:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1d4fd5f181b1b33b837fadbef67e9934b7d38f85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 19:50:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"86e9f31fd40297b096320e3f72007db5eaf20085","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 20:07:41.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9cc893e7afad480ea367401d62d6e995 : FAILURE in 15m 53s","accounts_in_message":[],"_revision_number":1},{"id":"57171e68dc7dbdaa88d4b70971ff8171cc91e064","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 20:56:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e65ff53f02d846648e997e7875c62c23 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb104bd3e9c54372af384d43f697257f : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/faa2ca5479dc447a9f12bd2776c158f8 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/552f2afacfdc40808128c17fbca0276d : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f66bede6ea4d4b4fa5777faa0f74a1ce : SUCCESS in 4m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bdd1de9b10f84e2bbf31944c893fcde6 : SUCCESS in 4m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6fe4131a66e94f35b7f777a4b1582fd3 : SUCCESS in 9m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cf1e1923c1c14e45b1453697e90a466f : SUCCESS in 9m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d9abf01571f049458d15d6e056903504 : SUCCESS in 10m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/38915141b5954861af642e245ee5b378 : SUCCESS in 7m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e011bdf7a0444ef8bf9567322fe62ee0 : SUCCESS in 9m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/87e4d95257684a8db03746af9c04f404 : SUCCESS in 8m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c86219f291b446a28a6e7ff546854063 : SUCCESS in 9m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c6f66b86ec2c483db920b44dfad48638 : SUCCESS in 17m 11s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/0aaf231e867a4b5badd212e50754a9af : SUCCESS in 7m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/52f5c701492f46cb83f264bddd35726c : SUCCESS in 31m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c79d77da4c5b428db9fba85078db38cd : SUCCESS in 12m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/de9743f2fc154b2792336a2e2e858da1 : SUCCESS in 4m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/43ede06743ca47e391fbee492b59218a : SUCCESS in 18m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cd2d6feb9bae4199b17f5afc077aea5b : SUCCESS in 6m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bd988f66c1a742bab6f8f06af891aa81 : SUCCESS in 7m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/42c72ce6fdf54cfd9a8e9219b29e7c79 : SUCCESS in 11m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dbf06b3359764355a63785d80eedc413 : SUCCESS in 41m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/03d07e4528e04aed832cbb4ee488e791 : SUCCESS in 58m 38s","accounts_in_message":[],"_revision_number":1},{"id":"83b3143e9f4a42b55620a4e79c285497307ee2a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 21:27:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a41c64d05aec476d9efc2dfa9be58143 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba56f4404e3047eb9b107aa777c0a348 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecc49e52757646a288223b48a6eddd9a : SUCCESS in 6m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6ff2c763ecb54b6897ab47304f3ce0dc : SUCCESS in 6m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2fe350d166c1489c9e51e3ca4ae4ddc9 : SUCCESS in 6m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bd6422c57b3e488894ff0e4ec15b174a : SUCCESS in 9m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/20617bb237df42bcb9155bd5de58171f : SUCCESS in 9m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d9a864dffb0a47dda3ed324de017fc35 : SUCCESS in 10m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/23cbc2f9d9ce42c0af03fef6ddf61b45 : SUCCESS in 7m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e5285de5f6da414fbb1fcde0db9a67d2 : SUCCESS in 10m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f9738bb705d04df7be71b14f031f1621 : SUCCESS in 9m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9bd55ff933304bf19d4f287d8c1b2001 : SUCCESS in 9m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c61c262bd6ce4cba858d94a04f5dc7dd : SUCCESS in 19m 04s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/1636b708efc149e5b0e4a12402c7e324 : SUCCESS in 7m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/872aab0695ec44e5a264babab226af9d : SUCCESS in 38m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0c2890be7414442b9415dd9c90db4abf : SUCCESS in 14m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d2ca622bfee645338d25d014a123c3e6 : SUCCESS in 6m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/de6453a3bde14d09ac205ade7838b84b : FAILURE in 24m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e9776cecda0c4682b041b1a359bb86a4 : SUCCESS in 6m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ff217f4f86cd4c11a333ba9abd5959c0 : SUCCESS in 7m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dc7a9cbe3be14ea0bc09d191aff0e800 : SUCCESS in 12m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4e11ce9966b4ce8a4c7c881c6cab4ea : SUCCESS in 1h 04m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/185927618f0e41c8b0a8383642ac4ce7 : SUCCESS in 58m 59s","accounts_in_message":[],"_revision_number":1},{"id":"0e3455f842d8907bd1e6303a7057fa934a06a1e0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-25 14:55:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b240aaaa0e569fb0845afdb5b2f44a79ce45a8f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 15:37:21.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5c353eeb5c144109bfd8e0c78936d7e5 : FAILURE in 16m 09s","accounts_in_message":[],"_revision_number":1},{"id":"94163d50f37d719ab9dfdae2e337c42b47114d0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 16:04:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4de1d77e605e42e0bed65fc0be8d962e : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e20ed7bbf2142ebb81f953f26d610b1 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/240c867d14bc4ca789955a2d3b7ef31d : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37f31cab4db743d59c7b62cb9a121ce9 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5bc592bbc1fc43919a3ec7e437b2c282 : SUCCESS in 4m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5865a85e2d9c4af09875d5f8742e0ebb : SUCCESS in 5m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/eb2ef1541e214b7d89cc7b090c7e8da8 : SUCCESS in 8m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6de78a6b059b485ca718a84f10b97068 : SUCCESS in 9m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/12383bd995e4494faaf369847b3f77eb : SUCCESS in 11m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d834009639734935b5fe6f5a7482058f : SUCCESS in 8m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bfc1c22015ea4ac5ad586834093bf62f : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/78388d73bf5346f3965d9e7f64fe457d : SUCCESS in 11m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6212751811d54246a51feb0fd23e10dc : SUCCESS in 8m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f8f916d489b64c448aa5673054c47c21 : SUCCESS in 16m 48s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/f1b1af9c71774142963990379cfeb734 : SUCCESS in 9m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/64fbb66e59ee4297850c2683258c8927 : SUCCESS in 28m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4be85c8f9715489a97feb845931b14bd : SUCCESS in 12m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/04b8db2469e74d7aa5b803244922db2b : SUCCESS in 8m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/69c157aaecb449faa795cfbd13bd8611 : SUCCESS in 20m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cdeaaeae66ea474b8e342f43c6a65707 : SUCCESS in 7m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/986171f144db4ceea856062cbc3ad4c0 : SUCCESS in 8m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/74bbbe5f322b4beba5854957cf610366 : SUCCESS in 11m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aead765bf54943cc9b88cdd19b6ae8f4 : SUCCESS in 55m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5fb41b1fb5324c24a598d480e893445b : SUCCESS in 1h 05m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7a03cea2dbf624577b6caae0b4b2ddfa34c25af4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 16:07:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4342c622eeb2f95041ecad8fe03b53da3a59efaf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 20:36:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d761097f7f434f12974a93bc75a448e3 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dfedd95d1983481f9dea61ce4ec7820d : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fccec2afbbf473388ac9d16ee012e68 : SUCCESS in 7m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7e496453b25847aebe3485270b77ab3d : SUCCESS in 5m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/abb5e717b2b2444d84119bb2540d41b9 : SUCCESS in 6m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/305768d93f454da69c3184d84fff3b11 : SUCCESS in 10m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8d5382d0e5f5416fbc0d675ca93caf5c : SUCCESS in 9m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9fe73e2f95974b5daf76f2ee88ab99fd : SUCCESS in 11m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c7cd187ffad347d1b80e173a04d5bb59 : SUCCESS in 7m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a8266f8ba0f3450fab9a773e3f5cb5b2 : SUCCESS in 12m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5ab34f8bb6434d20bb2185ded2946e8d : SUCCESS in 9m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e72a72d78058452a8c0cff7e6e151158 : SUCCESS in 10m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/738fd24ac80e48e89b47b9305cbd1f5e : SUCCESS in 19m 22s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/fe56b13bec5f4805baa0099d5f5abfcd : SUCCESS in 8m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/940c27271a754ce882a3fb17cd742610 : SUCCESS in 25m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/972ea77597f149b1a3438aeefd5ede3c : SUCCESS in 14m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1af952e7bb5d436c81e851a78eae157c : SUCCESS in 6m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6f49fa3ed936499c862fb1d3844ed77b : SUCCESS in 21m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/81ea3f5e61ed4193b0c27fcbf26690bb : SUCCESS in 7m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bd440f27f48548c6b816d36fc55fb269 : SUCCESS in 8m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8b38f83d3c43432c93ea95be1efba466 : SUCCESS in 16m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8340d6e35815466e89894f1642b7e825 : SUCCESS in 58m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7081dd53b0504370b5b78dd512eab024 : SUCCESS in 1h 49m 42s","accounts_in_message":[],"_revision_number":1},{"id":"6b6cf20cbfc36cdef735ba2107d160368aa39d22","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 20:36:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc125243cd1c001abeeed59035700bb5844e8d51","revisions":{"fc125243cd1c001abeeed59035700bb5844e8d51":{"kind":"REWORK","_number":1,"created":"2022-02-22 15:22:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/43/830443/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/830443/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/830443/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/830443/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/830443/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/830443/1"}}},"commit":{"parents":[{"commit":"3d141ae5509c80c8e5d44dc2a939d082db559b69","subject":"update constraint for oslo.utils to new release 4.8.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d141ae5509c80c8e5d44dc2a939d082db559b69"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-22 15:22:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-22 15:22:16.000000000","tz":0},"subject":"update constraint for automaton to new release 2.3.1","message":"update constraint for automaton to new release 2.3.1\n\nmeta: version: 2.3.1\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I920ee7bcca3cbf3e4051b3a3098beed53b1c969d\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I16e2e10d0b595fdbe86e1fd3a1e14bbdb3078941\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc125243cd1c001abeeed59035700bb5844e8d51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc125243cd1c001abeeed59035700bb5844e8d51"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
