)]}'
{"id":"openstack%2Frequirements~830445","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I22123c5de8a802cebd2c8677e84bd1f093723bb1","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"I22123c5de8a802cebd2c8677e84bd1f093723bb1","subject":"update constraint for oslo.middleware to new release 4.2.1","status":"MERGED","created":"2022-02-22 15:25:26.000000000","updated":"2022-02-22 19:32:06.000000000","submitted":"2022-02-22 19:32:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830445-new-release","meta_rev_id":"2ef16956d30b5166163a66b20df807995de789b2","_number":830445,"virtual_id_number":830445,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-22 19:32:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-22 17:23:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-22 15:50:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-22 17:23:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-22 15:50:58.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-22 16:22:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-22 17:04:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-22 17:23:26.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"653bd48667b2ec300b74bdd65acbac53b1e6800f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-22 15:25:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dfbc5a737024a5d162a72074f16da05df5d2c841","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-22 15:50:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"12919c405adf720ea0cf69b247490c8bd4ed4efa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 16:22:13.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/25c98778583646d7871dcae559d7f78d : FAILURE in 14m 03s","accounts_in_message":[],"_revision_number":1},{"id":"49bce4a3c4151833b4c88d6fffc853c76573b36b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 17:04:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18cda839e8714b36b405be3dfe22ca79 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc8f7b6a3713446bb0128079395a556c : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4f3ebcf6c864ca1bb1df5bbba0e3718 : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3d61eb01ab34284954d6bf58740472d : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/37526fecdf7742bdaf1be9cd00326a40 : SUCCESS in 5m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d847c628d5cc44f287c8079de79f4826 : SUCCESS in 5m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c60f8d80374a4bfdb519bbec225b16ca : SUCCESS in 12m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1a5cb80c9ff34294b32a6574eb9a285c : SUCCESS in 9m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/350e30e721f64fa998d1e14deb0bd9b6 : SUCCESS in 10m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ab88d9fa497744ef857f16a3b5e88bb6 : SUCCESS in 7m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/05c2e8cc0de8404ba016ae1baba66c9e : SUCCESS in 12m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c62187aa45984e5087923619cbbafc30 : SUCCESS in 9m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0a2ba1ec78ed479296e76539de7d2ba8 : SUCCESS in 9m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/44ab034ac03144ea8c331680af96162f : SUCCESS in 17m 26s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/c90b463a76054973adb2e5ef4d603c39 : SUCCESS in 6m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2c2c68679c3d4b9b8e9eb5e0f236c931 : SUCCESS in 25m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f172ce41b46e4a959ab517809d18c74f : SUCCESS in 17m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5806fcc4a5684460967cbc012b33824f : SUCCESS in 5m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fa19f3c5aeef4596ab8711673da5b880 : SUCCESS in 18m 53s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/46050e399bff402cb336c9d3c6fb5824 : SUCCESS in 6m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2c3277da7e0e4dde9f1eb6adc4390223 : SUCCESS in 7m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bd8d67b1e7df4c73a2474c3d9f994c07 : SUCCESS in 13m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5dea05e11a64230a8363240e389b52b : SUCCESS in 53m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b6c7e355772b49ce9e34882f729648b9 : SUCCESS in 1h 22m 10s","accounts_in_message":[],"_revision_number":1},{"id":"373ba76df6adbf04f2c246511e2cb47244817dea","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-22 17:23:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"04cec88522ed9d3bb77d197772b113215144dffb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 17:28:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"77c4342743ee829ec4bc54687aa97ad26814bb03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 19:32:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6a4e28774484fcaa95a93a4f5582849 : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26562de4ba8b4907b1595778f9cb649d : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b40e99049bf2497b93d9f4ede0ba005f : SUCCESS in 5m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1615102c9a4348c393befbe8e85c1b35 : SUCCESS in 7m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/516f90a08a894c49bfd0cdcb03cc217d : SUCCESS in 5m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e7a43bfe40014af0820d305280e9b02e : SUCCESS in 12m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fcdd98e7e1734211b2271b0c241ea9d8 : SUCCESS in 12m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5df221a3fe5e4b0daaee3d4e91fe634c : SUCCESS in 14m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/60f9a8d1851043aeb099fb75b31ebe66 : SUCCESS in 9m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/87b3807d2c7340c4b3b9fdd023500b23 : SUCCESS in 13m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bf37bf732ec44491953cc4ca83dc07b0 : SUCCESS in 11m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c623f3b4ae9d4326b97183c58278eac5 : SUCCESS in 12m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/af1ab4f4a8614de6998884e8f678c43d : SUCCESS in 15m 17s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/be463c0acf2e4a95b43a95a85d6f0e16 : SUCCESS in 7m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e9d5f186144142e9b74edb203b4ca829 : SUCCESS in 25m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2906e6c59071424297c02fa902df3e7f : SUCCESS in 16m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d5cdf39c82c947a19058725fc2b77ccc : SUCCESS in 7m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bb9200c68a8f447593f9904a75346fb4 : SUCCESS in 21m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/20e88ca622624c8a9a86b732caddfb2a : SUCCESS in 8m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b1bcb14d670f433387185a756aac11b5 : SUCCESS in 10m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c8ee347173244de1af8cf39385f7be9e : SUCCESS in 15m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/715b61524ecf4d939397c37be3d6ab4b : SUCCESS in 54m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5c02209c00d446baaa1f161f9b5da14a : SUCCESS in 1h 30m 14s","accounts_in_message":[],"_revision_number":1},{"id":"2ef16956d30b5166163a66b20df807995de789b2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 19:32:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"97e1c7dbb478ece0ab33ab2d66c63e8fef531a20","revisions":{"97e1c7dbb478ece0ab33ab2d66c63e8fef531a20":{"kind":"REWORK","_number":1,"created":"2022-02-22 15:25:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/830445/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/830445/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/830445/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/830445/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/830445/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/830445/1"}}},"commit":{"parents":[{"commit":"3d141ae5509c80c8e5d44dc2a939d082db559b69","subject":"update constraint for oslo.utils to new release 4.8.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d141ae5509c80c8e5d44dc2a939d082db559b69"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-22 15:25:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-22 15:25:25.000000000","tz":0},"subject":"update constraint for oslo.middleware to new release 4.2.1","message":"update constraint for oslo.middleware to new release 4.2.1\n\nmeta: version: 4.2.1\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I920ee7bcca3cbf3e4051b3a3098beed53b1c969d\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I22123c5de8a802cebd2c8677e84bd1f093723bb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/97e1c7dbb478ece0ab33ab2d66c63e8fef531a20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/97e1c7dbb478ece0ab33ab2d66c63e8fef531a20"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
