)]}'
{"id":"openstack%2Frequirements~830448","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~If82c190c0b006a186ecf5dc2b4f89a41485f9a3b","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"If82c190c0b006a186ecf5dc2b4f89a41485f9a3b","subject":"update constraint for oslo.limit to new release 1.3.1","status":"MERGED","created":"2022-02-22 15:25:44.000000000","updated":"2022-02-22 19:32:12.000000000","submitted":"2022-02-22 19:32:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830448-new-release","meta_rev_id":"ca787c8d7b41cd78ad1707480adb9151ec22e649","_number":830448,"virtual_id_number":830448,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-22 19:32:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-22 17:23:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-22 15:50:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-22 17:23:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-22 15:50:47.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-22 16:27:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-22 17:15:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-22 17:23:27.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"4a5cf19d406b644c0a5aae4e17f562cc0fbb3cde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-22 15:25:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e621fb5abbbe087411739f635358c1cad5a986d6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-22 15:50:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"656844bdc23a3b89018d7fdcafed68f08325f7c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 16:27:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/989010bc3f184b7fa977a8c19c94335b : FAILURE in 16m 23s","accounts_in_message":[],"_revision_number":1},{"id":"b78cd5afcebf01eb3e75842843ac8808890aeb6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 17:15:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2b50f2e50fa4d3392237be42e5bb2e0 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8321df94528f49c6abbafca96b740ebf : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e66aafc711784c708a77e814d7738b31 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21be03339da5403e8306bf748f259884 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/36ca403c152c43f49ffaf989b9efb209 : SUCCESS in 6m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/edab22d26c0a44c1b6a96502c380050d : SUCCESS in 6m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/745fa6199705486c9d8892bdfd63575d : SUCCESS in 9m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/824ec3edc2b04ab9a6e510a1527950f1 : SUCCESS in 9m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2a513a195b374ea68b1dd05c1f81185b : SUCCESS in 10m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/671c855d115d4e30b7254427a058c9a2 : SUCCESS in 8m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/505270f17c91412daf25727b1e557d24 : SUCCESS in 10m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eb47c23d181a40fab92ea3cb4de9eb6f : SUCCESS in 10m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f808f17e82df48fba9af6f1ea609738d : SUCCESS in 11m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ecfcea22443747e4810cda3965a5c3b4 : SUCCESS in 13m 56s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/9f28f252c6c14e6a9043d8044d742147 : SUCCESS in 7m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3ca00421a2b84cef9ea9a80821cac509 : SUCCESS in 35m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7dd2ca847a014ae3ba73464dfc9c0ba6 : SUCCESS in 14m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e725daf57da944d7abcd36667002d045 : SUCCESS in 4m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a46f029b9e8c4684a1f211fb80e09e5a : SUCCESS in 20m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/64aab8a0819743cbb3ae35b35d88b8fc : SUCCESS in 6m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/701074150073475e8885d380d8924ef8 : SUCCESS in 8m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/aca48366044d42b5a95f619712a792bc : SUCCESS in 13m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/abf1f91a96334c6fa0654640dfe81da0 : SUCCESS in 53m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e9b857d32d9478bb1d662aad4a6eeb7 : SUCCESS in 1h 27m 41s","accounts_in_message":[],"_revision_number":1},{"id":"9e706dd7983022574447b38fb8281231ee701c95","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-22 17:23:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"50fe2fcc79ac11d684181cee78da906b0e175274","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 17:28:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aa4c70ad7c2b6357017f15264832fd14932a947b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 19:32:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c490bdcb6b14d4c8654f54145f1ffd6 : SUCCESS in 7m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4eb9be07186e441898236879e8500d24 : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb585a2989754d6981a684c7295a34b2 : SUCCESS in 7m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9c300cab38a74db6ab927974eee67c6a : SUCCESS in 5m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cf613674f8b340fbb777382549e2a9f6 : SUCCESS in 6m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7a4a64a6d8e74f9f8d84e30988b4f644 : SUCCESS in 10m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a73ee9001d3e45999249e26ba5678d59 : SUCCESS in 9m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/757dedb065974df8959870ac7e49a97c : SUCCESS in 13m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f98af044f4084152ba518d220993f338 : SUCCESS in 9m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/78d4930db55f415bb93ac8e6bdbdda9f : SUCCESS in 13m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/54396fe58ee340848594fdcaae778471 : SUCCESS in 11m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4f1df003fec94dbb91a654ba95f629c1 : SUCCESS in 11m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2a68fd7cdf664be59034cbaa94ff8e0c : SUCCESS in 21m 22s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/3df53c2a58b0427ca5f1ee573b395261 : SUCCESS in 6m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3f9e5de579024f409c383289251ed926 : SUCCESS in 30m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a96f98d9ac2f4a11ba4ee1d6107682d4 : SUCCESS in 14m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/915fdb75e84241c7985c443f18321ef1 : SUCCESS in 6m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/36542f9067134813a3b59e8510965780 : SUCCESS in 19m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/111867d05735404592a5b9c4d49e176a : SUCCESS in 7m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d8ba2c4c57f2470587730a2d25898a74 : SUCCESS in 9m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/45cb9f3328e94367a4b034fadd53bde3 : SUCCESS in 15m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd51eb62a2ee4d218008da2db961364b : SUCCESS in 56m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a6e59abf9a974c65bb4bc7dde4481a50 : SUCCESS in 1h 18m 43s","accounts_in_message":[],"_revision_number":1},{"id":"ca787c8d7b41cd78ad1707480adb9151ec22e649","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 19:32:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"315c073cd895d122e1fcadbed570a1ef9106252e","revisions":{"315c073cd895d122e1fcadbed570a1ef9106252e":{"kind":"REWORK","_number":1,"created":"2022-02-22 15:25:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/830448/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/830448/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/830448/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/830448/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/830448/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/830448/1"}}},"commit":{"parents":[{"commit":"3d141ae5509c80c8e5d44dc2a939d082db559b69","subject":"update constraint for oslo.utils to new release 4.8.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d141ae5509c80c8e5d44dc2a939d082db559b69"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-22 15:25:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-22 15:25:43.000000000","tz":0},"subject":"update constraint for oslo.limit to new release 1.3.1","message":"update constraint for oslo.limit to new release 1.3.1\n\nmeta: version: 1.3.1\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I920ee7bcca3cbf3e4051b3a3098beed53b1c969d\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: If82c190c0b006a186ecf5dc2b4f89a41485f9a3b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/315c073cd895d122e1fcadbed570a1ef9106252e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/315c073cd895d122e1fcadbed570a1ef9106252e"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
