)]}'
{"id":"openstack%2Frequirements~830450","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I1e2daf4bfa432731c332d779d4e9cb821f4c6813","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"I1e2daf4bfa432731c332d779d4e9cb821f4c6813","subject":"update constraint for oslo.versionedobjects to new release 2.4.1","status":"MERGED","created":"2022-02-22 15:28:50.000000000","updated":"2022-02-22 19:32:16.000000000","submitted":"2022-02-22 19:32:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830450-new-release","meta_rev_id":"ebc225e1c51270416df2f80f3777d902252ed2f6","_number":830450,"virtual_id_number":830450,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-22 19:32:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-22 17:23:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-22 15:50:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-22 17:23:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-22 15:50:30.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-22 16:38:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-22 17:12:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-22 17:23:29.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"6453b18097bc1c591c0f8f6fc5675f90970df3bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-22 15:28:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ed5812960a4cc0f664ca8e1b52e60b796651e3f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-22 15:50:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b3f119f9838d1956aafeb821be35675963fbff55","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 16:38:13.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/99d728342e434552b6aa2273317c897a : FAILURE in 13m 48s","accounts_in_message":[],"_revision_number":1},{"id":"dccc2718424c57a22c8aab52bd9113b199aec16c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 17:12:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e9a4c7d8fd941a687d232bec93c457d : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fff8301b54b74282a7022d14702c9252 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e02a58d351894654ae6b88d035c19363 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05aa195be0084ecfa0d14bbe81c13954 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9d602728da9340abba95310b7fc2154d : SUCCESS in 5m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1b4e6ca86ebb4955bee8b1f896dd78b4 : SUCCESS in 5m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8fb1d1ca2dd7445980d309c63be49b3a : SUCCESS in 9m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/da2b8ead493e4d5eb74f5cfc21735b69 : SUCCESS in 8m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5d8269613d6144ec85597334a5b02872 : SUCCESS in 15m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1aaa60a73b8c4ad4883a3bc51283a8a8 : SUCCESS in 7m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/aeb1506bc64d4b8caa0a9db0246343b5 : SUCCESS in 12m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/524e5ca74bbb4eda937167989b2fb6c4 : SUCCESS in 10m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4a4d3bd5c925409889b8153d30eb256f : SUCCESS in 10m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6e561985b35b40578da761cd56d23b09 : SUCCESS in 19m 55s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/959b1a28263545ffa019ce5c1dfcb64a : SUCCESS in 6m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cf4d65f8d1384cacb20c9016544134b7 : SUCCESS in 23m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/aa5d8383034945a8b037b50c54a751d0 : SUCCESS in 14m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e24eecbc0ae846ed8f005283792231de : SUCCESS in 5m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5c63b5fede844fb5a190402db7ee340e : SUCCESS in 19m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/304353f79de34663aa17a185915182cf : SUCCESS in 7m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/57f198f1962546a0acbe154779922baf : SUCCESS in 7m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5588d80c71e14957bf1db35e544961a4 : SUCCESS in 14m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c9893a8de25e4b33ad15818bc406cc57 : SUCCESS in 56m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f7e713d717b4b339a859f3531d36854 : SUCCESS in 1h 24m 28s","accounts_in_message":[],"_revision_number":1},{"id":"b6ffdf76cb94b70c34d7f2095d347cbcdc2b4c5f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-22 17:23:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a37599c8675812ce4bc0f3793e24808426d48143","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 17:28:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e30e11e8b58203d367bd28294aa9b9d5b9c489fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 19:32:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/695254db389d48f59c2abde6dbc4506b : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51fdd99f313e4562b7895b4bcbfb70d3 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04af2195ada14acca0dcf4112c311837 : SUCCESS in 7m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fbfc5147ad1a44c6ab795709d742009f : SUCCESS in 6m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5090cbbafbc248ce8dcd99b8339990c6 : SUCCESS in 6m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f36cea1630284fd2849612ffebb01c6c : SUCCESS in 11m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6ae5274a2fc246708f8e726ae2e7b0b0 : SUCCESS in 11m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2ba8958f7e2d40a7ab13de498cc67783 : SUCCESS in 13m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0d6bf9b1a697494cbbcddac5436a3b1a : SUCCESS in 7m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5d7685cc695b4991a7f4629dbb7529da : SUCCESS in 12m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/86fbd98945b94614a6d7b2f36bad8787 : SUCCESS in 10m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e359c006ca434283b152664453e4a5bc : SUCCESS in 12m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fa7d8579891c49c698d8745112eb0150 : SUCCESS in 14m 25s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/caa2f62aab124687a11ca61f78bad1c5 : SUCCESS in 7m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/187289eb1f5e415fba5177a41f1be433 : SUCCESS in 29m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/266d1ad0443044898d9c1a1918815e78 : SUCCESS in 14m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ca524d4bf1464237a332a2adfdb20251 : SUCCESS in 5m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1b69a9e85f294638abef70ed8c7167e2 : SUCCESS in 21m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cff090555d25442191b4fb47ea8ad5ce : SUCCESS in 8m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d5870410d1fb4eebad87c9219536477e : SUCCESS in 9m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4ef23ce297474382aabdc078c21e3fd8 : SUCCESS in 13m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8d2456e2b9f473dab703f37d20ca87f : SUCCESS in 56m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d52f292f98d846f79707307f4691a223 : SUCCESS in 1h 26m 21s","accounts_in_message":[],"_revision_number":1},{"id":"ebc225e1c51270416df2f80f3777d902252ed2f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 19:32:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a477e84406d23f55a2e9934386fe2a8b7c96534b","revisions":{"a477e84406d23f55a2e9934386fe2a8b7c96534b":{"kind":"REWORK","_number":1,"created":"2022-02-22 15:28:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/50/830450/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/830450/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/830450/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/830450/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/830450/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/830450/1"}}},"commit":{"parents":[{"commit":"3d141ae5509c80c8e5d44dc2a939d082db559b69","subject":"update constraint for oslo.utils to new release 4.8.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d141ae5509c80c8e5d44dc2a939d082db559b69"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-22 15:28:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-22 15:28:49.000000000","tz":0},"subject":"update constraint for oslo.versionedobjects to new release 2.4.1","message":"update constraint for oslo.versionedobjects to new release 2.4.1\n\nmeta: version: 2.4.1\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I920ee7bcca3cbf3e4051b3a3098beed53b1c969d\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I1e2daf4bfa432731c332d779d4e9cb821f4c6813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a477e84406d23f55a2e9934386fe2a8b7c96534b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a477e84406d23f55a2e9934386fe2a8b7c96534b"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
