)]}'
{"id":"openstack%2Frequirements~830451","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I460cf69df88049a627a6c276d0e68bdd697f1767","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"I460cf69df88049a627a6c276d0e68bdd697f1767","subject":"update constraint for oslo.metrics to new release 0.2.2","status":"MERGED","created":"2022-02-22 15:29:35.000000000","updated":"2022-02-22 19:32:19.000000000","submitted":"2022-02-22 19:32:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830451-new-release","meta_rev_id":"20262becd5269275a6e5237404232c810d4d6147","_number":830451,"virtual_id_number":830451,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-22 19:32:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-22 17:23:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-22 15:50:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-22 17:23:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-22 15:50:24.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-22 16:41:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-22 17:19:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-22 17:23:30.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"8000d7970f66c61b3028c59df8b45b5e8eb79d6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-22 15:29:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"020d7e84fc139c737f0f376215ba0dc7e6ca2b6c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-22 15:50:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6a53ee29ccd42d1fbd4d32eb421beb3ca4b9d0c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 16:41:40.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d6159f788a1043cc8a2d0c9b27fe2866 : FAILURE in 16m 39s","accounts_in_message":[],"_revision_number":1},{"id":"7f32f281ef4b4fc827b42ef9a3a9638e807f9fd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 17:19:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5121ed7e49534180b78542a4c09f912b : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49e530479fd14c818a68cdc9edfc442a : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/770fefe2e078486da3dc27dfa61454f5 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/beb798517f1a4a5d8ec237de50d27471 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/91bd6f5afc6649d2afec7d29dc079757 : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/238384531987411b98ba62d63590a15f : SUCCESS in 4m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/af8fd937befd49099b0d136840f09e1b : SUCCESS in 11m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/55f0447ddc9c4760bf57939866f742ef : SUCCESS in 11m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9e64561a1dae45b2b4f1e81117433d80 : SUCCESS in 11m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d637c59e99b944f88cbb5dbc71161223 : SUCCESS in 7m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/80609f60128a4ab08ecb399cb178ed48 : SUCCESS in 12m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/940f985acb844c469a013b7e06bd3d5e : SUCCESS in 9m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/dc843af31c274f24bf0bbd4f2479ab6f : SUCCESS in 11m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/eecc8c72aa3a48dabbd8baa408b1d833 : SUCCESS in 17m 29s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/19ffd7f01d1e49b582982906525c8207 : SUCCESS in 5m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4b483326703640fdb26ac9c1bf51b9ba : SUCCESS in 45m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dbf05301e26c482e9df6073de7d013fb : SUCCESS in 13m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/33441b357d0c4c618d38a590ab0997e0 : SUCCESS in 4m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ccec70fe21bd41c5adaad15863e6e016 : SUCCESS in 22m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0ad4f6b725f74e9f8016f1774ada5382 : SUCCESS in 7m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0cdc0bd83cd841669bf4b3ff1d7dfa39 : SUCCESS in 7m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2bf4dfd7b00749b2a38774d2d42d4042 : SUCCESS in 13m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f53f2c7d1bff40a294aba6bb706740c8 : SUCCESS in 1h 06m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/623c322ede354f218f8c7b07c2f3f79b : SUCCESS in 1h 27m 12s","accounts_in_message":[],"_revision_number":1},{"id":"a078976e3edde6377016c4e43456ffe2165df643","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-22 17:23:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"942eaa09d94f362c1fb6a9350ca4efac7edc4227","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 17:29:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8322aafdb1f13ed379b21f1b3ce488952ca68450","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 19:32:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/321896a84b054c6b8ad4908a76a8fdd1 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96d141b5d5c5430ebe80492fa44cac1a : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76dc85a8dfdb4daca0aed3b883bffdc2 : SUCCESS in 6m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/40e07d4e3fd442909eaac8fcd7da8b2b : SUCCESS in 5m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/948d275c0fb24d76baf11df382c63fe0 : SUCCESS in 7m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d3e078cc0733437ebfec40b00f0f83eb : SUCCESS in 12m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a37baadf33dd4adba5d46ebedeb55db7 : SUCCESS in 12m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3b7d94467d05468484fc68407cb23129 : SUCCESS in 12m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/03358abede72451a89f0336da92727d0 : SUCCESS in 8m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d319b5e7fe33408ba44895d10702c44d : SUCCESS in 12m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e56f1c69e6144f20b7c6ce01cfe73a18 : SUCCESS in 12m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2e6fe322cb4f461aaae18b9bb919e1e9 : SUCCESS in 12m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/23747f917482484d95f830582483d3fb : SUCCESS in 20m 22s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/8e9de0ae559c4f3e8bd7cd2024a7dd88 : SUCCESS in 7m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7499c091412f433d9c1b00ab6c9c45fb : SUCCESS in 27m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a968504169874c60955d471f5ddeb820 : SUCCESS in 13m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/69c3c92d16274fdda942258ae4c87e44 : SUCCESS in 7m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/03299f95a9a64cdbaaac15431b861ef1 : SUCCESS in 21m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/92ffc78abc434f24a2fe79ac18924178 : SUCCESS in 7m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3c89b4ab2889460da2412533717f67b5 : SUCCESS in 9m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2b28a3f4e5384b9ca8d5b1ef6f8ee7d7 : SUCCESS in 13m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b9212dc86c54b709fc8b54f0b7ca143 : SUCCESS in 1h 11m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bc839e93ce68465ab2aaa287efe6f5ad : SUCCESS in 1h 25m 35s","accounts_in_message":[],"_revision_number":1},{"id":"20262becd5269275a6e5237404232c810d4d6147","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 19:32:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d7e212948c4466983b7f1b1d709f827f1fd03108","revisions":{"d7e212948c4466983b7f1b1d709f827f1fd03108":{"kind":"REWORK","_number":1,"created":"2022-02-22 15:29:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/51/830451/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/830451/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/830451/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/830451/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/830451/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/830451/1"}}},"commit":{"parents":[{"commit":"3d141ae5509c80c8e5d44dc2a939d082db559b69","subject":"update constraint for oslo.utils to new release 4.8.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d141ae5509c80c8e5d44dc2a939d082db559b69"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-22 15:29:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-22 15:29:34.000000000","tz":0},"subject":"update constraint for oslo.metrics to new release 0.2.2","message":"update constraint for oslo.metrics to new release 0.2.2\n\nmeta: version: 0.2.2\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I920ee7bcca3cbf3e4051b3a3098beed53b1c969d\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I460cf69df88049a627a6c276d0e68bdd697f1767\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d7e212948c4466983b7f1b1d709f827f1fd03108"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d7e212948c4466983b7f1b1d709f827f1fd03108"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
