)]}'
{"id":"openstack%2Frequirements~830452","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I7c6609984c2991db6be7c2463b153edb912c6944","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"I7c6609984c2991db6be7c2463b153edb912c6944","subject":"update constraint for oslo.cache to new release 2.7.1","status":"MERGED","created":"2022-02-22 15:30:12.000000000","updated":"2022-02-22 21:33:03.000000000","submitted":"2022-02-22 21:33:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830452-new-release","meta_rev_id":"3940174089298c19ca1af2b3661cf69c242346fc","_number":830452,"virtual_id_number":830452,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-22 21:33:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-22 19:48:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-22 15:50:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-22 19:48:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-22 15:50:18.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-22 16:46:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-22 17:25:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-22 19:48:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"52d5826a4d7891821677584d1bfdb83f75c0a62b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-22 15:30:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bfd5342700c2f9f35b4177e386f34747c6609d3d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-22 15:50:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"eaecaf7baadc128c83e34423884ba91577a888c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 16:46:26.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/98894a1fcb7c43e4bab5d407aee7dce3 : FAILURE in 16m 41s","accounts_in_message":[],"_revision_number":1},{"id":"824012426f0923e8d60f5ad50f662bffbed218e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 17:25:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed56b8bf748d4bddbd18294545bf2e0e : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c40ba01ea26b468b988a888991aa5b99 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba2514f628d54138974f0b48b1f9b9eb : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b83e70295a2423bb4d67e9a997df245 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e0959792bef243d384ffc18f0eb82873 : SUCCESS in 5m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9352dd7eecd84e07bb35ec768054ee00 : SUCCESS in 5m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/da649af168c74155885e625e1e8fc779 : SUCCESS in 9m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/37112951885249ffbf6c8178a8f4db52 : SUCCESS in 8m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d7226088265e4c3ea2fb58036e3e0744 : SUCCESS in 11m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c57de9b7607241c2b2c7816e92dc715d : SUCCESS in 8m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4e099b0116cc481e99dee08ed21798c1 : SUCCESS in 12m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3c5fc70f5e51478c8cf6930c03e9492b : SUCCESS in 9m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f6e36fc71f304fd0b7c30db2b24de711 : SUCCESS in 11m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a184d6ee1c62444587dfd5741119384b : SUCCESS in 15m 35s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/ecca4177b99b4002be98daaee884475d : SUCCESS in 6m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c714288df2f2498dbb52a066893706a8 : SUCCESS in 26m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4acc7355d9d8452aa2b8555aa9af5a2b : SUCCESS in 16m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1b3190e3abca4314b17b0fb9f8a9c3d3 : SUCCESS in 4m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/031c62a6bbfd4edea518dbcde0d87e3b : SUCCESS in 20m 58s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b961ef3e448445448841a8d5427eb337 : SUCCESS in 7m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/827dcfb505f9432ea33a88a093497803 : SUCCESS in 6m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f06e4b674bdf4957a61dc762215e4823 : SUCCESS in 14m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94d80c85fa7449c3aa180ed94c64bc83 : SUCCESS in 51m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa156f36d6714ff39dcee3a67911207d : SUCCESS in 1h 31m 22s","accounts_in_message":[],"_revision_number":1},{"id":"01befd6d9354465c6617bc6d0c771a394b1bebfa","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-22 19:48:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"09bc4fde04ad690bb531bf29fa7b21c8d07cd15d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 19:50:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"17c2165523abca1fcb41c5f037b0bfa597276a1c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 20:07:56.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9290c80ce0554fd4a473a06db9bbaa05 : FAILURE in 16m 08s","accounts_in_message":[],"_revision_number":1},{"id":"5ff9d3e52b4511e361b4835a932cd79bffa31cc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 21:16:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5afa4630ccf4d84a92370d0272bd5fe : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9961109ae2454f1bac5410e524413162 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a661fce25e3d4530a48f30f1c2f9d729 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06ef12c52e5f4df5bf91e791368b42f7 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/39cc8413dc254790b18c4b3f708e0941 : SUCCESS in 6m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cdd55e37093e403d8ef35be0cf71c6ca : SUCCESS in 4m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e058c319fccf47e0b22f8ed0b57e9df3 : SUCCESS in 9m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d3409bfacfce45e2a43083e1d5a58b73 : SUCCESS in 7m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d3a08ec1fad1439f8c4ee16dfa0e26bb : SUCCESS in 10m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/30d08ccab5c34f8fa4f50219cc763720 : SUCCESS in 5m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1d73845bd8324336af54901fceb89fe4 : SUCCESS in 21m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2728dcfbab7e40e3a90ce962fc0a688f : SUCCESS in 11m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bd348e7ec489470186704f22835dfc24 : SUCCESS in 7m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/be0e271870624dd186b8cc8478d6f0bf : SUCCESS in 12m 22s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/bae684d0625d45a595f4e55da542d7f1 : SUCCESS in 6m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/751f9f26efee4db88d9ad0a04086564b : SUCCESS in 28m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ebe14ea2a81440bf95627f00e85d62ef : SUCCESS in 12m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/45a97ac2ecb546bfbaaca5e07be1483d : SUCCESS in 6m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/56912f79e00a49bf9479acc6e5c0daca : SUCCESS in 17m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c93c160fb83a4c2285c1639cd29c4496 : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c849228b19f2425584f4f8478d806004 : SUCCESS in 7m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/12f1d13c640c4b8d94bfba5ee7c1be76 : SUCCESS in 12m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/93607109bf764f5f95a7fe129b47f3de : SUCCESS in 38m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dddc4d80914844b9a7cf6d24eb0747e5 : SUCCESS in 1h 20m 05s","accounts_in_message":[],"_revision_number":1},{"id":"37d4d8b2a7d93eef01ce7260afe18456ea86bbe1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 21:33:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d92accb8b4c4f05a86dcf91e5f34bcc : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e1561522f76427cacdb0aeed6a9423e : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72029dd53b084e31872623631ee24b72 : SUCCESS in 5m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bd0041c81055491da06eeba81e9402b9 : SUCCESS in 4m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/039a8234804e4d3fad4aa254ab4ce888 : SUCCESS in 4m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2d445fb82348413c8ed5357e4d47b32a : SUCCESS in 8m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/28a24449e70a42479edcc910b5dfad17 : SUCCESS in 8m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6e2f9fe5ed48447db13f6b9bdc450fec : SUCCESS in 9m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ca0de0e7f27b4bff8cdddcd9b266a29d : SUCCESS in 6m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/15822bac148b45f8a1d4e1fe3b3305ce : SUCCESS in 11m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8097a5f99d4641c2b19aab0f4f132c5d : SUCCESS in 9m 18s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/14c036c1af75473481ef31939b42d030 : SUCCESS in 11m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/aed293e4d43847c3a29455f7966783a1 : SUCCESS in 27m 07s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/6ad08d91ecfd46af967cdc32e437cfa2 : SUCCESS in 8m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a8e5c13cf16a4000bdba4d3f10eafb0a : SUCCESS in 24m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ed2c722b6e164340b7ebb966c7bce5bd : SUCCESS in 11m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a907bcd651da4ace81f3774a25f9a37e : SUCCESS in 4m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7346977aa37a4834925d15beae639d86 : SUCCESS in 20m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9dcb004805d545ee8d307e8f3f3e8d21 : SUCCESS in 6m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5690186f2a464f5e8226877be2f73ae0 : SUCCESS in 7m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b8c768c6d31c43848a0a84864787b4aa : SUCCESS in 12m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6603d5d8bd1f4b9fa4581e3b32a251ea : SUCCESS in 47m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8fded18813934413806654c773408792 : SUCCESS in 1h 15m 00s","accounts_in_message":[],"_revision_number":1},{"id":"3940174089298c19ca1af2b3661cf69c242346fc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 21:33:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ff3c808cae011111751a13d54dcd4cb411e5091","revisions":{"5ff3c808cae011111751a13d54dcd4cb411e5091":{"kind":"REWORK","_number":1,"created":"2022-02-22 15:30:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/830452/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/830452/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/830452/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/830452/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/830452/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/830452/1"}}},"commit":{"parents":[{"commit":"3d141ae5509c80c8e5d44dc2a939d082db559b69","subject":"update constraint for oslo.utils to new release 4.8.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d141ae5509c80c8e5d44dc2a939d082db559b69"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-22 15:30:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-22 15:30:11.000000000","tz":0},"subject":"update constraint for oslo.cache to new release 2.7.1","message":"update constraint for oslo.cache to new release 2.7.1\n\nmeta: version: 2.7.1\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I920ee7bcca3cbf3e4051b3a3098beed53b1c969d\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I7c6609984c2991db6be7c2463b153edb912c6944\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ff3c808cae011111751a13d54dcd4cb411e5091"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ff3c808cae011111751a13d54dcd4cb411e5091"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
