)]}'
{"id":"openstack%2Frequirements~830682","triplet_id":"openstack%2Frequirements~master~Id586063638082ea3f3bbe0fc0810a81748d7df04","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id586063638082ea3f3bbe0fc0810a81748d7df04","subject":"update constraint for python-designateclient to new release 4.5.0","status":"MERGED","created":"2022-02-23 19:27:10.000000000","updated":"2022-02-23 22:23:28.000000000","submitted":"2022-02-23 22:21:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830682-new-release","meta_rev_id":"679d3139067db3794223006cae6e9ab403d315fb","_number":830682,"virtual_id_number":830682,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-23 22:21:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-23 21:06:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-23 21:06:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-23 19:41:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-23 20:42:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-23 21:06:13.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"38d0659b8e1c631e7099161489b2bc97208d18cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-23 19:27:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"020e7642a6b7d589658372e02714de16fd27ae92","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 19:41:56.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3a32eaa5a23b443799c24aeebdf4cde1 : FAILURE in 11m 56s","accounts_in_message":[],"_revision_number":1},{"id":"aba3d23fade569a111a62881c2b9cb482f29ebee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 20:42:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad7368c70bfd4b81aa48a1434ce2c091 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76d361182686416dba5a0f1b62aa47be : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33067362c17948d9a0031a8c14f582aa : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9ed22a4213144959e39ac823725a872 : SUCCESS in 4m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fd6bc4296080482f8c68b49979b38891 : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/df7930bfef3d4e5fa4498dce9284941a : SUCCESS in 3m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fe25f2f2b5c545fe89439d0894f3f573 : SUCCESS in 9m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a1460830823646f0825ee88480da081f : SUCCESS in 8m 30s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/222b5dbae26d4cf0a5a94dd575ecf525 : SUCCESS in 7m 22s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5e8807de06e44347baf65d51f1400d5f : SUCCESS in 10m 36s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/afe34e2e4c594cd58fc99fc1010c6ccb : SUCCESS in 5m 22s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/79a7a4e602594e62ae69464a66a672c5 : SUCCESS in 5m 47s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9c4863e7fd64413c942f766ebc185524 : SUCCESS in 11m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e437ca8bd82c42d2976c3780998d9c44 : SUCCESS in 7m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9c4937344db24e05bb34d52dc24b5687 : SUCCESS in 8m 41s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/957c8ff160964329b2a180ed175788fc : SUCCESS in 12m 59s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/6ee18dffd74447c2b73aaf353fc3eba9 : SUCCESS in 4m 03s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6d26a1e437da4852b7e3a37adac4329b : SUCCESS in 17m 21s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/561d47a16fcf4e2aa021ae3a926f0e79 : SUCCESS in 5m 34s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/fd51ddac385d4736b826c673adfc0fca : SUCCESS in 20m 09s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0fac85c3e8d94dddba93f4821b1380df : SUCCESS in 10m 37s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0db11c79039f4d1ca3875c8823fbe872 : SUCCESS in 5m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/45d82f3d062c435f841bcf422127cbce : SUCCESS in 3m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a771df13f4044ed2afe3437ebd2831e8 : SUCCESS in 17m 29s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/488e3144f5c94232978e9e56f417b557 : SUCCESS in 6m 28s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d36a9f3e98904fad8cb3372d4373d4b9 : SUCCESS in 7m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ba18b701a713466da5935facf082150d : SUCCESS in 7m 09s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/319dc7e5aad54066a866848a6cd75b69 : SUCCESS in 10m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/390a32cdda2746c79629a44c52613174 : SUCCESS in 37m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a691db58ba2d42b499667634ef753d07 : SUCCESS in 1h 12m 20s","accounts_in_message":[],"_revision_number":1},{"id":"f6becf0ae78a66bac5257997795ce80680d8056a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-23 21:06:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3a6f766bab97b9d0de6921d21dbf277ffa5de2ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 21:07:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"16d5aa0e1a900629e128b1690c9d117e0da8e6f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 22:21:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/894c850ef54e40d7822c53a64678c1b9 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfb8239d314e43b489e75212407b7242 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6493949a26d444f4a8820077f7a49011 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0da0ae59ab574795b62f9dd742673fa6 : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e36606a1a2b44375820c6117878300ec : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5fcc631007314d4194a5f536e85346fe : SUCCESS in 4m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/48f23e31457946ad848e3d425e4b5ff2 : SUCCESS in 11m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d54241a5e9644457bc8770c63303c183 : SUCCESS in 9m 14s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/997ecf45966c4d4493c55afa41e9f101 : SUCCESS in 11m 57s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/eece6fda7e854912a6167f69ba9b9ec9 : SUCCESS in 5m 08s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ceaaf32a6c6e4e9f9c29c52a00dc5c7d : SUCCESS in 6m 29s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1a80c35dd56a4ffc91c2f4aee9987ace : SUCCESS in 11m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8b59c66f84424e72987576507fcb2592 : SUCCESS in 8m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/97b9dcf2ca05480593e60e7ab5a33a1b : SUCCESS in 9m 26s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/3532ff96af3d47e9bc175758331c7679 : SUCCESS in 16m 10s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/49ccac1d32004067877b491ab0a18084 : SUCCESS in 4m 40s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/572c26640cc845e0a36ae715f66d0b82 : SUCCESS in 18m 42s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ba1af5e729d74889b50b1d04244d8e29 : SUCCESS in 5m 05s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4155226dd8344d4aab9b1ec3eedc5938 : SUCCESS in 27m 04s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e656a18a806a4a018b1a2a3e7fcf52aa : SUCCESS in 12m 35s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0a0b2ba99c1d49fa8b3e8fc3cf2fb9c7 : SUCCESS in 6m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b28e5a87ba1443498a1241dc9e6b7743 : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1b74a00be85a4c0084f96381ef744dea : SUCCESS in 25m 11s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b6b70c91f1874322b40f7bfe775c7cf5 : SUCCESS in 5m 50s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/52272cc1384a4ccbbc3886abacdb0933 : SUCCESS in 6m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/37506280c23b47fc85285a706e25669c : SUCCESS in 6m 28s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/79a54fb731a948e0ac80250022feba96 : SUCCESS in 11m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3dbc174ef2845b682717255d21338bd : SUCCESS in 53m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c554e8ddbf94e3da798dbd926b70d3c : SUCCESS in 1h 07m 30s","accounts_in_message":[],"_revision_number":1},{"id":"be90ede9efd97b0b5d4be893a1edec7e9b3f2ef5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 22:21:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"679d3139067db3794223006cae6e9ab403d315fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 22:23:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ffa6f4af2b14fd4910db02e89ee0d9c : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4c26c59194333b0d559c8d494993e15e529e09e6","revisions":{"4c26c59194333b0d559c8d494993e15e529e09e6":{"kind":"REWORK","_number":1,"created":"2022-02-23 19:27:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/830682/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/830682/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/830682/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/830682/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/830682/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/830682/1"}}},"commit":{"parents":[{"commit":"e342fb64cd8e54d8eb8f3bd7899e26b4f240a1ca","subject":"Merge \"update constraint for python-troveclient to new release 7.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e342fb64cd8e54d8eb8f3bd7899e26b4f240a1ca"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-23 19:27:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-23 19:27:09.000000000","tz":0},"subject":"update constraint for python-designateclient to new release 4.5.0","message":"update constraint for python-designateclient to new release 4.5.0\n\nmeta: version: 4.5.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I4bb03dcc41c3a9deb9a59843a7cc58731528084a\nmeta: release:Code-Review+1: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Id586063638082ea3f3bbe0fc0810a81748d7df04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c26c59194333b0d559c8d494993e15e529e09e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c26c59194333b0d559c8d494993e15e529e09e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
