)]}'
{"id":"openstack%2Frequirements~830815","triplet_id":"openstack%2Frequirements~master~Ib5329e07ef6e92ac56d753bd4a1928cd94de336c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib5329e07ef6e92ac56d753bd4a1928cd94de336c","subject":"update constraint for python-blazarclient to new release 3.4.0","status":"MERGED","created":"2022-02-24 12:58:09.000000000","updated":"2022-02-26 10:31:24.000000000","submitted":"2022-02-26 10:29:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830815-new-release","meta_rev_id":"31a4477e7fc7548fbd16849c14f3c7036c24fd50","_number":830815,"virtual_id_number":830815,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-26 10:29:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2022-02-24 20:21:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2022-02-24 20:21:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-24 13:05:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2022-02-24 20:21:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2022-02-24 20:21:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-24 13:05:20.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-24 13:19:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-24 14:42:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-24 20:21:04.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2022-02-24 20:21:59.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"121ab8b4870495344a653fb7a568109ad8781862","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-24 12:58:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"151b097fa7a0099c92025c5e55cdc11f6512e28b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-24 13:05:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8c7b15848a4a23ef09c5673625d3ce7fc584404e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 13:19:57.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/bbcb7a1651c640b39fd26c45e6b2e725 : FAILURE in 15m 38s","accounts_in_message":[],"_revision_number":1},{"id":"f4ba3827fec74961b4806ae55541df95a3abd3da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 14:42:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e10883023ff4c6e9fd1ed64b0e3c24c : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0a9a8e3836e457682d5bd796037f82e : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/337dbede9ff544079bc32b671986991d : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c89452bf7b14c62874d61c6649143cb : SUCCESS in 7m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0196876e1a9b4448a2d319c0a584ce06 : SUCCESS in 5m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9664c67f697141a8a23a2dcf85473c6f : SUCCESS in 6m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/176bc341e2d94d85bdcad45c67fe41d5 : SUCCESS in 9m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4ae44dc6b0934c839dea830b0ebad837 : SUCCESS in 12m 32s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d89efd3614bc41a89c029ad2a9cbbd74 : SUCCESS in 6m 00s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/77a89bcb52f847539f2afdb17a92cd79 : SUCCESS in 14m 45s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/ffae0a6048104f56aae72f455646ee07 : SUCCESS in 5m 03s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/13930616b0ca4db4a2800c39a98cc742 : SUCCESS in 7m 53s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/397852cacb014e9b82d11f7d5193d368 : SUCCESS in 9m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4eda311d614f4145a3f1841856cec196 : SUCCESS in 11m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3851c256b0be454e9b3a62f5b8062acf : SUCCESS in 8m 52s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/18cde66f7e134df984f53940be5435d0 : SUCCESS in 15m 40s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/332f2dbffd3b4ccfa57c18ecae1a7d39 : SUCCESS in 4m 58s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/3e7b6196dd7e4fe8bc3982b2976fa082 : SUCCESS in 15m 32s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f48096e0256940f48b2dd9c00308eadb : SUCCESS in 6m 17s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/22c3716e1f074191866e0f865dce695f : SUCCESS in 26m 27s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8695de3dbbf24c07bc0c6904f09a2a9d : SUCCESS in 13m 57s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d133d22e56f64fd49dc991519aa9d27c : SUCCESS in 5m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bd7191bea0e54b88b3af84b45943d597 : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0a2a0bdc4d6a4222abc27ab680f09d47 : SUCCESS in 18m 47s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/8bd4a8a227c043e08000ed072f77752f : SUCCESS in 6m 48s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8e2f2cceb06d454e85cfd6dbbdeea61d : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/982b29131adc4fd48c49080cbc7c88d1 : SUCCESS in 7m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/23cc72ebd8f9447d8c47690ace9bb8cb : TIMED_OUT in 41m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f91254d00ee4c60aeb1039720fa0912 : SUCCESS in 42m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/676de7a91ec94b74b397ca41dff60d84 : SUCCESS in 1h 36m 27s","accounts_in_message":[],"_revision_number":1},{"id":"2c2847b74e941d0e731e3cd7e157c76680f8b549","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2022-02-24 20:21:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7893beaf5d85b6a61bfa71cc23aeaa279c43a4fa","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-24 20:21:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f4cf9bc8e0f896676e62bc775a131e940bf0c53b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 20:50:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2cb140eeae0f485d8f75568d9bd8b37d : FAILURE in 15m 07s","accounts_in_message":[],"_revision_number":1},{"id":"211e08d4f16be3cc76f6bd79fcb8b2e5ed3a3ab3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 22:09:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/427519ec60ba48c688793a5cf33a7b78 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/645cbadf09a8432d8204b22ee5552d5f : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/566c36943df74e989cec3b6e827522dc : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/576bee047a424873b8b96929b62641ad : SUCCESS in 4m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c29efd5ae3b149cfa5389f7ea8098664 : SUCCESS in 5m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/49edf9a134384a8d98d2ba807ba666d3 : SUCCESS in 4m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ded1f6bce97a49b896dd2ab236d997f2 : SUCCESS in 8m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2140d76a841747529a5085440ccf71e4 : SUCCESS in 9m 05s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/9d4be1e60fe44e8da137ad17862c97c9 : SUCCESS in 5m 32s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/84a3c517b2b34efe84ccd4ad5fe8060e : SUCCESS in 12m 31s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/e869e6ff8cc642f6881832ced401b4f7 : SUCCESS in 5m 47s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ea95a7f11b1e4e30a373f7c5ea5fdeca : SUCCESS in 5m 31s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/2b722e127c5341c18bd296dcf2351e33 : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3657fe088f45433abce867ae93e6fd66 : SUCCESS in 9m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4b4512d1f4f54b35870d237fa62fe4af : SUCCESS in 8m 11s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/dfa37b498d834924a32befe8581f2162 : SUCCESS in 22m 20s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/99a65f2a2c394e4ca38f8f33e249d991 : SUCCESS in 4m 37s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2fd371830f0341769a31e2e5c78f820c : FAILURE in 27m 32s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/617b0487510f454cb40e723c68b43447 : SUCCESS in 6m 12s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d6154cb3d940497a872868e89ce5b8bf : SUCCESS in 29m 07s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c5a3d3a41959407bbb4414d52cb5ddba : SUCCESS in 15m 47s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/6df6da6f58144e81b050ccc121d0e1fe : SUCCESS in 5m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/826850020de643309cb72fb368522e50 : SUCCESS in 3m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/75c2d7847cb2459d81037bcf1f6f0bec : SUCCESS in 22m 01s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/9b389aac6ae04c1ab30bbcccab4a93c0 : SUCCESS in 7m 25s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1cbeef29efa243dfbe94002460f41b07 : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4367cbb2e4d545efbb9b9575ed50c1b2 : SUCCESS in 6m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/93557b0a59f046eead95056754b40007 : TIMED_OUT in 41m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5aa17d97af48427fa5915b50c5268b28 : SUCCESS in 44m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/101f4e1106ed4a3e8664d99108232f96 : SUCCESS in 1h 39m 31s","accounts_in_message":[],"_revision_number":1},{"id":"d4af30153f820323e4a53bd1eb832f668376ba63","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-26 01:55:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9107ac95ec2e340ca668ad33594a895ada0aa7bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 06:08:14.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/471e8166aa6b4dfc8113a18f149a2405 : FAILURE in 15m 50s","accounts_in_message":[],"_revision_number":1},{"id":"28c6ca7fefd21262d65a540c97cc42e77037d054","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 07:13:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d08907c4c82040d386c8040361fe9846 : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07dfc6c3df7e42348fd36b5b435e1e4b : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c44f47b7f0bb4d97a60e43ef1f20a54c : SUCCESS in 8m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f02fe9ab220487cb94ebda418734430 : SUCCESS in 8m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f15e485a4b334f6db0a2ba9e184fa674 : SUCCESS in 7m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d86adfec299f442bb0dde046c5faa965 : SUCCESS in 8m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/41a6bafb1f3f4597a18cd5f3c93ab4c0 : SUCCESS in 18m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0b4686eb29a44535a762a49653e6b4c9 : SUCCESS in 11m 37s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/13e3b632b1a143c3a37992dfb86aeedf : SUCCESS in 9m 22s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a1adee0ea3394bc0a6a60b8809abd989 : SUCCESS in 14m 25s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/e7fcc5e7cd1c4bc5a8d75b411fab9655 : SUCCESS in 8m 05s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4e21b98d94e94ae99ec62bf141aa0964 : SUCCESS in 8m 50s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/2a0e67aad78f4068bf75e63f5314ca32 : SUCCESS in 12m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ee2fae42d3bf4886ba2e39926af6d774 : SUCCESS in 11m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a64f39b7c1f8446dae96c9dfde906e03 : SUCCESS in 11m 51s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/311816fb93b440bfab0c87a354064e05 : SUCCESS in 13m 48s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/f1ce57fe2ebe4624a4c8d610b9343ab2 : SUCCESS in 7m 27s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/800676b603e6474889d6c4cbeb3abf1f : SUCCESS in 13m 07s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e5b21bb32c2e49b8bb9c52284792b982 : SUCCESS in 8m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/fdee7e0c00564943a685aa69158987f0 : SUCCESS in 24m 07s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8fba3564eaf24971b33f2e6be7168c35 : SUCCESS in 15m 05s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c449c0b596854da18bd08bfeb46d5d1d : SUCCESS in 8m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/78d38b71ad4146a88ada98443bf33510 : SUCCESS in 6m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/517e3012e58c4ed49abf3893e7726ed9 : SUCCESS in 20m 47s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/aa91956905d3485cb16532bf83ca9ea1 : SUCCESS in 8m 59s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2da9778386c54fcd99846c8638c728df : SUCCESS in 7m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4482d173b10c454484ce9b01dd78147c : SUCCESS in 9m 31s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d2feb28b3cb74c9090420820a0d9c832 : SUCCESS in 15m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/32a672ffd32f4b02a8ec451b6ec5d194 : SUCCESS in 44m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a766e995b98d4ad796b7d6c346870100 : SUCCESS in 1h 22m 12s","accounts_in_message":[],"_revision_number":1},{"id":"d944020739fef9c003a1b6e17222740ae78c88a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 07:14:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aeb8152ef0770ff8bd55fee0342dd56b51d62c54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 10:29:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/394232b04a8b4d5c9a752bcd39439e09 : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/988e332fa74b4ccbbad577899c95c1ab : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a37c5e807eb64d96a6bd753279397505 : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1abaf123fd64555a7463bc8c6242f5d : SUCCESS in 5m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/80419103fee94e53876f5e5280f90677 : SUCCESS in 7m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3ee9bb9a2c064240b8d5d6cfe4aff27f : SUCCESS in 4m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fd4b5eb3a32347e290c086d4cf227f20 : SUCCESS in 12m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/944238647f6d49eb8837b0e4010284a1 : SUCCESS in 13m 42s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9ed8c410a7cf42b280f26134361cc422 : SUCCESS in 11m 13s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/ee23f288e30b425cb04a1c3f7329bafc : SUCCESS in 7m 22s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5e7c6c5ed00540649aa19d54f338edb3 : SUCCESS in 7m 02s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5a4fd5319aad443991e0d32084bd5244 : SUCCESS in 9m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/18ddbb6930b944488a5ca3c6cbd8d9d8 : SUCCESS in 10m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/79fe0366b63046f6a2defd2235008d5a : SUCCESS in 11m 16s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/fd98ed5b971940038f3fb59434e20329 : SUCCESS in 13m 41s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/4331c5e818434560bee0c8cd07ca87c2 : SUCCESS in 5m 27s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/530ed40a040c45db9f623363cadeb3d9 : SUCCESS in 9m 10s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/cc2f5c3823464157a5b4d9aa343da6e4 : SUCCESS in 6m 17s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/de59cf7ac0f44bd5a697dcf40e1b91db : SUCCESS in 26m 21s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/85973c9dd5c0435e9f68780ca1144396 : SUCCESS in 14m 21s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ca204e9f1ba74d748f4d6b4894a92ef5 : SUCCESS in 7m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f150ce80cd754de097680969dc29c2bf : SUCCESS in 8m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7449a55357fc476e9be938d9d61f153f : SUCCESS in 21m 21s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f3be5c7028a74f30b4a24b35e841aa0e : SUCCESS in 8m 30s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b62d037dcc0e4ee19736f28069221cf6 : SUCCESS in 8m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/043983df1d8647d3a01728bc82dd4d20 : SUCCESS in 8m 23s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/75b7aa4be3dd4d2ba1a6b103a406a4c5 : SUCCESS in 13m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/588ab26596d14f58941608ce3c511af2 : SUCCESS in 41m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb5993ad5aeb4fe2b2359e1cfc4124f5 : SUCCESS in 1h 26m 12s","accounts_in_message":[],"_revision_number":1},{"id":"3d7af3f2409bf5579279b49392621eeff8e6fa14","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 10:29:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"31a4477e7fc7548fbd16849c14f3c7036c24fd50","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 10:31:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4439199b8f84c70becb37a5dea515b1 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e4dd5fe226ac1e2db263de0a179f86cfffb0b449","revisions":{"e4dd5fe226ac1e2db263de0a179f86cfffb0b449":{"kind":"REWORK","_number":1,"created":"2022-02-24 12:58:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/830815/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/830815/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/830815/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/830815/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/830815/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/830815/1"}}},"commit":{"parents":[{"commit":"4c26c59194333b0d559c8d494993e15e529e09e6","subject":"update constraint for python-designateclient to new release 4.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c26c59194333b0d559c8d494993e15e529e09e6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-24 12:58:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-24 12:58:06.000000000","tz":0},"subject":"update constraint for python-blazarclient to new release 3.4.0","message":"update constraint for python-blazarclient to new release 3.4.0\n\nmeta: version: 3.4.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nmeta: release:Change-Id: I0c5d5a890c1040b3e599fd2b7d4e4921925128bf\nmeta: release:Code-Review+1: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ib5329e07ef6e92ac56d753bd4a1928cd94de336c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e4dd5fe226ac1e2db263de0a179f86cfffb0b449"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e4dd5fe226ac1e2db263de0a179f86cfffb0b449"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
