)]}'
{"id":"openstack%2Frequirements~830816","triplet_id":"openstack%2Frequirements~master~Id44b342bf3aed481bcba67b1b1c15a3ab693c3a2","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id44b342bf3aed481bcba67b1b1c15a3ab693c3a2","subject":"update constraint for python-zaqarclient to new release 2.3.0","status":"MERGED","created":"2022-02-24 13:03:15.000000000","updated":"2022-02-26 11:33:54.000000000","submitted":"2022-02-26 11:31:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830816-new-release","meta_rev_id":"3d0b1f3e3f589cc4ab9c2d936fe79bf03cbf8d4c","_number":830816,"virtual_id_number":830816,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-26 11:31:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-24 18:52:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-24 13:05:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-24 18:52:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-24 13:05:14.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-24 13:19:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-24 14:50:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-24 18:52:44.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"26cddef44a24b463f08e862ba03822ce141f7657","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-24 13:03:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ca5bd41214cba3885ab5c1ebdf588f181bf384e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-24 13:05:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"75c2a8bb04ea3ffe38fddea3808364542f9e4a47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 13:19:01.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/45c82a49aa0f4ef7894f61773678b3f4 : FAILURE in 13m 18s","accounts_in_message":[],"_revision_number":1},{"id":"03922bc3c6bfe2a101a29c567a0219d0fe326cf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 14:50:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e051c62db03439bacaa473dc83b68c0 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b949305353df48ed87333d3e7fbf990d : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd1d8e166fef4707934e403a59d3fbca : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6b8c55562a142e3b2054c73d00df613 : SUCCESS in 4m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4a5acc372b0b4a28bac8e6e400cae651 : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/577cb586b4fa4c4a81f802c77f3adbe2 : SUCCESS in 4m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d03948dab90e4f048f6805e87b61180f : SUCCESS in 9m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/020fa4fd0bd54f76a55ebe24a2574c18 : SUCCESS in 10m 05s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/780e9bb337cf401fbd31d4d1a8ca67b6 : SUCCESS in 6m 31s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/da1d23e0efa443d48f8bb5074f469306 : SUCCESS in 10m 32s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/c87067aa266a4b82892903c0d3258bf8 : SUCCESS in 4m 51s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ae1711d3006343f787936e09e9eb956d : SUCCESS in 5m 57s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/05331c04a1294455b4f10811068f13b0 : SUCCESS in 9m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ba3661d9dcbc463e8b6e3277606c7fb5 : SUCCESS in 8m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ae842d214a304f0e8312d35acc7ae54e : SUCCESS in 8m 53s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/13bbb33170d04a01a8b3145cb6c7df3c : SUCCESS in 14m 10s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/8316409c14b0450c96660b0c8d521668 : SUCCESS in 5m 26s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/60a7659082a64bb2b2d37a7f418c919e : SUCCESS in 16m 11s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/4a59988a0b8545848e8e91dd71053200 : SUCCESS in 6m 20s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e3aa727d39e143cbb9838bd7617f5e0e : SUCCESS in 32m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/87f014347c694a1ead9ef5e3178b417f : SUCCESS in 12m 09s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4f871576dd8d4ae3be810b6965e0efb3 : SUCCESS in 5m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fed6d3e2ffba4d138af06a4738c45886 : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d3de51631ab2455a94e951d619af88eb : SUCCESS in 19m 57s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/0f0a1e30ee344b4d8f36b87e906becb3 : SUCCESS in 7m 05s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/094cb43251b94d5587f0bd13a958da90 : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f65cf6a8d88948968f44df3644ca6e7c : SUCCESS in 6m 35s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8aa717890a5f44f0b75fceaf08a1b631 : SUCCESS in 12m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2309a1a6369b42a4bedbd9ac16c84265 : SUCCESS in 37m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c25ab03f325d419aab1ff3778dae503d : SUCCESS in 1h 44m 27s","accounts_in_message":[],"_revision_number":1},{"id":"1ad08246532084e50562781dff6269df167d0836","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-24 18:52:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9b4afcbb937c48e66ce887ea630b70e7bed2a88b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 18:54:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bad63a114fdc42a584fe8a4f28bd37971d454ba6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 21:25:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34740249cc924275ba8644202cd4d568 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/812e1adc98f9452abae093e5793ba80f : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51a8591bcf9341a2963cb556b8f8f3a7 : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a55ba71c6f64bdd8758f994fd3a31a8 : SUCCESS in 7m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/55a822de96cf449a8d585ec120895737 : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/680fe23d148441d0908354d959f46dd5 : SUCCESS in 4m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3e70eeee4c994767882af29fe2d3692a : SUCCESS in 8m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f55959b17a33493bae552226c488106d : SUCCESS in 11m 10s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8b4efcffd2db46baacb390bd904f5c46 : SUCCESS in 11m 16s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/5afab7d0601b4bb69a4d4c195fc52b59 : SUCCESS in 5m 32s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ee024fe1be6c46169be7abd38adc9dfd : SUCCESS in 6m 16s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/080462d9b95746a2be4df4f2b39029b0 : SUCCESS in 9m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/715034bdfe6d46088f7e1e14c18ee737 : SUCCESS in 9m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/11456691348849c788d54453db69911f : SUCCESS in 8m 42s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9caae3c19e1149d38b57c64aadbefc22 : SUCCESS in 13m 21s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b3dd662e1a514bb1a69cf906277aaf09 : SUCCESS in 4m 29s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7a95cfff8048425ba783d8247e395ef5 : SUCCESS in 15m 59s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2e43166d505c40cd820ef2555ac6b8de : SUCCESS in 6m 54s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ad2c7914f7494de4a0269e6dea6e54b1 : SUCCESS in 22m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a5f64bcc773e43a5bd4b6decc748fa68 : SUCCESS in 12m 05s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/531c9e9fcdf54f48bb55d505ecf2f7ef : SUCCESS in 5m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/34c90151e5fb433298c969537d1ea047 : SUCCESS in 4m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/55d5e22cb83540f082654e5e36b98909 : SUCCESS in 18m 23s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/33856b19167b43c881b8e26ab4a7c31b : SUCCESS in 6m 46s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/031e9fa136cb4ae98f8c76c9a4c69a53 : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/68c544395b564fcb91ce7a6fe86daf02 : SUCCESS in 7m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e946c1d815ef4969a23d1f8688c1cb19 : TIMED_OUT in 41m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b194e85caeb043ba8e3bb5215ad9c913 : SUCCESS in 44m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/24953831f05a434eb6a70820425e16b5 : SUCCESS in 1h 33m 35s","accounts_in_message":[],"_revision_number":1},{"id":"2ea30f37d4b3d998aee7d5386ad0ce1e58ddc5f8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-25 19:51:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"edc3a7d1449478dd9d175248cf6a36b4d1f647ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 20:50:26.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3560f5587afc4557b3d6777e7e508e12 : FAILURE in 19m 58s","accounts_in_message":[],"_revision_number":1},{"id":"c7d4cffab8a9ac323138bf67dcb5d8cd45ac7995","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 22:25:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/959d48688f1c442f863940772492fe24 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e773809c3644aea83223909caa6b84f : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b87797a012324888b281837ce7b4c148 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2653c4a09f8240a2a8e9e97a77c32540 : SUCCESS in 7m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/29885d3c3cbf42ac9724c3dd8a992612 : SUCCESS in 5m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/35fcb4c40d1849d1b8f7496912433019 : SUCCESS in 6m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4f7433d7dd8e4913821319bdea21393f : FAILURE in 6m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9ec6e6476fb245ed87f29fc8673dd5db : FAILURE in 6m 02s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/cb10b27ae3ee433d929e707275afbc82 : SUCCESS in 8m 03s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b7f59601d30c4cbdb3ec52bc70aeb2b6 : SUCCESS in 13m 54s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/b1909703038848a6a96fae87253efda2 : SUCCESS in 7m 56s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/11581a6938c5496589de58c111dea9fc : SUCCESS in 8m 06s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b24b97f2a2bd4f7a87325a78865ff6ec : SUCCESS in 13m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c013f47104f14e5eb3a6d0dcda2c1eeb : SUCCESS in 11m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/96d1c5151c584e55b8d28877901e2b83 : SUCCESS in 10m 22s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0501a2a43dc74e659825c35e1409d7b0 : SUCCESS in 14m 15s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/db5e68721a5d4293b848bb6acb3ebab6 : SUCCESS in 5m 47s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ec2f8867efc249b9a31b7e0ea23d0a1c : SUCCESS in 18m 17s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e35134e8956e423da92229d649f70cd9 : SUCCESS in 7m 47s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/76a9de45f66941eeba839638ebaecd44 : SUCCESS in 31m 28s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/bf35a3959c694366bcf99575ebd8a77b : SUCCESS in 16m 04s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0428f35169a449048b1cdbce934be5ee : SUCCESS in 7m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f498d422d6e841afb53005567390a6f9 : SUCCESS in 6m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/93f871b577154c7bafbb3d330a632744 : SUCCESS in 26m 51s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3c6c03c766cf4623b2c67a69b6dd7ee0 : SUCCESS in 9m 21s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2feee2f8c51040e6a8c72081babd85f4 : SUCCESS in 9m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1632b0ca1f124550b039e9cfd9a742d9 : SUCCESS in 9m 43s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2fef369154f24165abf6aca0e59470af : SUCCESS in 15m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/783c202977854297b52b36925b50a4b0 : SUCCESS in 55m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cd0a594597a146c1a5b59fc2b2c65943 : SUCCESS in 1h 41m 43s","accounts_in_message":[],"_revision_number":1},{"id":"45ec348171c795d6dea5c7e2a9a4c266385783fb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-26 01:55:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3d67800338987a0b20b0a054cc3398e31f22d238","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 06:08:42.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f75314854ed64d4da8dffdb1848d2f30 : FAILURE in 16m 19s","accounts_in_message":[],"_revision_number":1},{"id":"bdac2ddcc38f223db8f89655167c67a41114cb79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 07:22:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81e45cbc184445cb9937ef8f189b2943 : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d5f66f82c0d44fba52e4728dbf4553a : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/512064f1df24488885da05f8e88f0f5e : SUCCESS in 6m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22dded0e2c2e48dd9313d9e298cd39b5 : SUCCESS in 7m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9033d8a5e9bb44da8e5fcbcd70108445 : SUCCESS in 6m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3aefe8ede5504f1880e5b92c67b38b04 : SUCCESS in 6m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0de606909f9547458254ab355bd10659 : SUCCESS in 12m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f959e6b4c65144858585a158fc7c261f : SUCCESS in 13m 09s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/32d2d10c7f7b47c3840d0ab1dacc6923 : SUCCESS in 9m 40s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4b619754732e4ea784f2e21705b250b8 : SUCCESS in 13m 29s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/439603cdd89645ba8d8c218b1a968689 : SUCCESS in 6m 46s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7fe7e1dcedd24ca28a906ba3e7e512d4 : SUCCESS in 7m 46s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e3f1cfeaf721481e8b0d871a0aa80d01 : SUCCESS in 13m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d01637efefd34d3d9c41a60333aea257 : SUCCESS in 11m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2a6e056bea054b4e9f984c6bba8b30da : SUCCESS in 13m 25s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c8c0f3624bfb4b929e240903663d9848 : SUCCESS in 14m 21s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e882444f20e64ee4ac16f613845197a1 : SUCCESS in 6m 43s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e0b3b73e63c147798dbfa862d375340d : SUCCESS in 11m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e8f1496be1664849a8c26d317fba56ae : SUCCESS in 7m 16s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4919896627af452498480fd6cc4b798b : SUCCESS in 26m 37s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7ec49a24a7e94855879a140f01e24961 : SUCCESS in 14m 06s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a09002fb8db84fafa6d85ba442ae0f98 : SUCCESS in 8m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/beb5bcb18dcb4ae995802ba81d14c945 : SUCCESS in 6m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9fd968558e4f4656b7155274cbea5fff : SUCCESS in 20m 00s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/afe6258bea5f46c1bd7bdd6af42e5014 : SUCCESS in 8m 14s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6a5116ece9cb40849f38b8ea5523c22e : SUCCESS in 7m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/904919e1e8db49d0afd4c0894970788e : SUCCESS in 9m 42s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ee7a38576d8f4ae89f7352c567fa8a80 : SUCCESS in 19m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc89531503b24e4792dcdb137a5b5b80 : SUCCESS in 58m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/682e9561601e4d10b67b9524bd10c12c : SUCCESS in 1h 27m 33s","accounts_in_message":[],"_revision_number":1},{"id":"c49c97afeb57a1c309e5753c78b903b7c7e92700","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 08:59:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b7df475a4a72f492e04a39e1b1be1bf213c7fe8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 11:31:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcbab43e1d624469b281d97d8b457bb1 : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d940950e82849b8b0560d38ff3af4f6 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12a80a28b64048fe945ab8696d3e6a23 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01836a0519d949618620095c3cf0a100 : SUCCESS in 5m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/651cf2dad66549049fa05a28da403764 : SUCCESS in 5m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/513811ae4c674d7eb5598bf94ffe2f7e : SUCCESS in 5m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6d63a06e7b3f4d0cbbff1da718d39167 : SUCCESS in 11m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6cc212f243a84877a23c33c68ba4c562 : SUCCESS in 9m 45s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/aa2f9ca7a6d74b4daa6f247366e7f2c0 : SUCCESS in 13m 32s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/55ee1628fa3b49e3bdbc12f4c2c96440 : SUCCESS in 8m 59s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e1083eec9eee4eff832d1583fe5437f9 : SUCCESS in 7m 41s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4960b84adf61411c9e52288d0aa82c19 : SUCCESS in 9m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d57303ef02d3424c859197ad528b361c : SUCCESS in 10m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/750d2ce42c9b438fb262a52f5cedc379 : SUCCESS in 10m 36s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5396798acd254cb5b772d023f3a9afe0 : SUCCESS in 17m 19s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/47a02e2b89124278b42f433aa19461b4 : SUCCESS in 5m 34s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/17dcb1a5dcd348e89827fc5a11e3bb7d : SUCCESS in 15m 43s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9df3ef55018f46d2bed924f843c46a00 : SUCCESS in 7m 02s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/971274493b1748cbb2936343cfb0a598 : SUCCESS in 25m 23s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8552c587327b4f319c61c6a8c1f2176d : SUCCESS in 13m 39s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/cf56e5a45d7249beb4bc42dd7e8a73b0 : SUCCESS in 7m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/da660c6d13294e44add3925a51673e70 : SUCCESS in 5m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fefa1cace4a04afea4febbb65f411129 : SUCCESS in 21m 17s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1006ac251a7440fdaf63c26451b52113 : SUCCESS in 8m 03s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/39d1b246c2b74c2c9caf3a29f5ed66f5 : SUCCESS in 6m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2e89a470ed344a69883b2310250ecb8a : SUCCESS in 8m 38s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7318c5f46bda4883b2f6001cd4ae2d55 : SUCCESS in 13m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb732d299c104b4ab8a3dc9b6b7b00c8 : SUCCESS in 41m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/453ca04812514b5bbe17be89e4ee61ca : SUCCESS in 1h 41m 37s","accounts_in_message":[],"_revision_number":1},{"id":"6f495e1f0030b4910dd37adac816f2abf1324c8b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 11:31:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3d0b1f3e3f589cc4ab9c2d936fe79bf03cbf8d4c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 11:33:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d4a08e21de94a0590599942143f267d : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2f056ef84372c14af95996e4565ba51b6b097a1d","revisions":{"2f056ef84372c14af95996e4565ba51b6b097a1d":{"kind":"REWORK","_number":1,"created":"2022-02-24 13:03:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/16/830816/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/830816/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/830816/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/830816/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/830816/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/830816/1"}}},"commit":{"parents":[{"commit":"4c26c59194333b0d559c8d494993e15e529e09e6","subject":"update constraint for python-designateclient to new release 4.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c26c59194333b0d559c8d494993e15e529e09e6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-24 13:03:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-24 13:03:13.000000000","tz":0},"subject":"update constraint for python-zaqarclient to new release 2.3.0","message":"update constraint for python-zaqarclient to new release 2.3.0\n\nmeta: version: 2.3.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I5ae00ce9df41b0ddb869953d13ea8674467db9a8\nmeta: release:Code-Review+1: 王昊 \u003csxmatch1986@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Id44b342bf3aed481bcba67b1b1c15a3ab693c3a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f056ef84372c14af95996e4565ba51b6b097a1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f056ef84372c14af95996e4565ba51b6b097a1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
