)]}'
{"id":"openstack%2Frequirements~830838","triplet_id":"openstack%2Frequirements~master~I9a3d025658263da6e44581d78c2fbacf34f89944","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I9a3d025658263da6e44581d78c2fbacf34f89944","subject":"update constraint for python-glanceclient to new release 3.6.0","status":"MERGED","created":"2022-02-24 15:09:59.000000000","updated":"2022-02-24 23:37:24.000000000","submitted":"2022-02-24 23:35:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830838-new-release","meta_rev_id":"54343a56579827e66dd507518685995ec11b6e43","_number":830838,"virtual_id_number":830838,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-24 23:35:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-24 20:21:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-24 20:22:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2022-02-24 15:11:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-24 20:21:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2022-02-24 20:22:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-24 15:11:15.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-24 15:43:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-24 16:44:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-24 20:21:59.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-02-24 20:22:42.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"b2c936fd6c4851569159d4cc1004bc881396ff13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-24 15:09:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b494c134c53be9b6f8c67b8f9f788e06d450fbe3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-24 15:11:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"19259a46b99a931f32b6e562a899baf5ea7540ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 15:43:47.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4ee7c54abcd2470aaf8719bca69a04ba : FAILURE in 16m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d990c1f94df897f930eaaa61ef0535bdb9e533fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 16:44:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0925b967eef847fea36a2a1bc5f82e40 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce674cb1961b42868832ca3bd620bf0e : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6aca1bb4e1bf4058ba7508c137eeb888 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2eb554abfdf94786ae1997c4c7ccbbb1 : SUCCESS in 4m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eebb724775024549bb849871d5b833a3 : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/424c8922ab494915be7f65d98e5b783d : SUCCESS in 4m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/314efac3593a43069cfb8bd5b707ffe4 : SUCCESS in 9m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7db328a38e6e47fbb38f19073c66b1ab : SUCCESS in 10m 33s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f2f7ade226d244e099f499d91bc2358c : SUCCESS in 5m 51s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ca1578d63f1b4f7886783083920d5f7e : SUCCESS in 10m 21s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/0b334e6e97484b199097d87a95191924 : SUCCESS in 5m 13s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/898534d84c214659bb820cd425645c72 : SUCCESS in 6m 09s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f532fc8610ce4f988fa562e16066381b : SUCCESS in 11m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d96932c94ccf4508b354bba89b6df565 : SUCCESS in 8m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/44dc506f7bc742b6b22565b0186f03e0 : SUCCESS in 10m 06s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a47a1e3316dd4153801cf0f28933ef2b : SUCCESS in 12m 35s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/bd20903f5d814653bf1396574dd17c4e : SUCCESS in 4m 22s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/27fd0284b1c54814a7115d805fff0ea1 : SUCCESS in 18m 37s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5dc4453354664ceebc8e9f7cf6ee8851 : SUCCESS in 6m 29s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/25e7017126114498bc63747b0ef450b1 : SUCCESS in 33m 26s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/90e6e0a5e23846b2ba7f9812936617fe : SUCCESS in 13m 31s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/1a3a52e33d004b04aa3a882509d40a6f : SUCCESS in 6m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3a107fbf3ba040299b79ccd4a2708ad5 : SUCCESS in 4m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/312bb5f4f4f7456792c097945014e44c : SUCCESS in 19m 52s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/8f7f33a5cfb74d8e946c7bc2e47879a6 : SUCCESS in 6m 52s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4f518fceb7f14f22ae51bef805520882 : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/88431b0d39a44231800ccaf50c6f6cb6 : SUCCESS in 7m 47s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1fb1ccba33b648449e6f2de9b2b001a0 : SUCCESS in 13m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a65881307c184bebb3292cd4f19de8d8 : SUCCESS in 38m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1d615fa3322a4974920387de44657103 : FAILURE in 1h 24m 26s","accounts_in_message":[],"_revision_number":1},{"id":"191f553564af2fcf4f2d8c96b516ecf066b13e77","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-24 20:21:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"df15b1709a28489ae6ba18d3de16f8bd9040387e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2022-02-24 20:22:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3bd7902986ef2d3c220c1c6c1849cd82f0b65e6b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 20:50:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5e620247683a478d853aeef766ca4944 : FAILURE in 14m 00s","accounts_in_message":[],"_revision_number":1},{"id":"d6ada11d43e574077dbf98de6fb0af950c3d6017","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 21:59:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dfebf21e74449419c58724ba2313a95 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36c1fb7e3bd448dbacd16bc9e68657bc : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7d9c42f231d4f95ae4735336c4f5bd6 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bdce5334f5354ef6a9e65edf77acaf26 : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8c3fd317457e4f6b91fe997483de4dea : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0e0f43d0b2ef4d6aa217c6a983cbb556 : SUCCESS in 4m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f249addd0af7492688ff8b699eff874c : SUCCESS in 9m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a6ab6232059740b8afca13be15b174ca : SUCCESS in 9m 29s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ddc113faf2464df8964f9df0eb079f0c : SUCCESS in 5m 47s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8f349eba93b9476bb21da3a78aab7c7e : SUCCESS in 10m 41s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a4ca9fd111824494914d3e4349633dd7 : SUCCESS in 5m 12s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f502d562b6a54d03a47c19f952af2f53 : SUCCESS in 5m 59s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8e5d14810ed04370b3c44cfce14350bf : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b26ebceb5f844e4cbb5eb1bde80f2499 : SUCCESS in 8m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/43755e47aa494e6e81a3d224e672fa99 : SUCCESS in 8m 29s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c4120d528a27497b80d18c6da894133c : SUCCESS in 11m 21s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b1dbcdfc3299434ca98ae759908b4513 : SUCCESS in 4m 18s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8e712e792331455aafff64402722e9b4 : SUCCESS in 16m 13s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0aa6f97292f6468998089ec2d7806ede : SUCCESS in 6m 06s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/0f8961686f8f42dfbc75487814e08ecc : SUCCESS in 25m 15s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/85210ab01b9c4732b1720de743e9a46f : SUCCESS in 12m 23s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/361eed58fbed46af81125d442d98404f : SUCCESS in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d883d00a25084980b83286bd2438f4d3 : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/67aecf236555455dbd89eb4658164b90 : SUCCESS in 20m 22s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5788733d51444f9680ce929cc7d809ac : SUCCESS in 6m 08s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/ada4773d2f254c91bde8ea19c4e53b63 : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f62f00cda3124b42a159c8e8cd33d2cf : SUCCESS in 7m 09s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/be8dc6d4138c437c970b935c902dc87e : SUCCESS in 12m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b403e31c8714a0ca8c07f962481cb0a : SUCCESS in 50m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4cf6cce4d9da440bb16b05a429abef7b : SUCCESS in 1h 30m 57s","accounts_in_message":[],"_revision_number":1},{"id":"36bce9f37f9310ff516089c83068d4864da2a871","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 22:00:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ba3520093b8742a5ec6b97c2d155901779846ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 23:35:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdee0f30f63240b6ae783b51ffb221e7 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c121960c56841a3a4e3d45d87065024 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd5d64a089bc4a2aad767523e41e73c6 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ddfb3d91ae9343cd9ac93d223d434504 : SUCCESS in 7m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fdd229f3197d4ef08fa7f3061a8fb7ed : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3e1bab973ef74c208a08121c778bad04 : SUCCESS in 4m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bd074f7c36d84e5293197e3b00c2042f : SUCCESS in 12m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/861ee592ac464de3a3834a53add912db : SUCCESS in 10m 27s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/499ea110ae774fb6a08a57ad562fce92 : SUCCESS in 11m 15s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/e09770ba55764299927689bd45232bb3 : SUCCESS in 6m 33s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d808e74e8d684e0b8be55d314278048b : SUCCESS in 6m 01s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5f5d512a58bf443a8e8457ae6bed997f : SUCCESS in 11m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f9ef9d9bb7ea4b7bad869b58c1b71519 : SUCCESS in 9m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4c64efed8a6a4ec7a7fbca43cffff0d6 : SUCCESS in 11m 01s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5ec7ec07987c47c7a2172e74ad5cc598 : SUCCESS in 12m 23s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/0344ff5f8dc34917a33abe7325f3b0fa : SUCCESS in 5m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f4d01aba7fea407383c04411a9eaf894 : SUCCESS in 19m 20s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/fb9b2a8239b549fcad97cb157b971c4c : SUCCESS in 6m 05s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/0bd7d05038524399bee078bc9da126be : SUCCESS in 25m 20s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/50645c0f554445659b572347eb2b488a : SUCCESS in 14m 00s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ca33edf121c34772bc9ddc0d54ec11f2 : SUCCESS in 6m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/570b9a34f92a489cab36cc78a7a0c01e : SUCCESS in 4m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e4ec7f7612954c889ff8436e0fd35922 : SUCCESS in 19m 48s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5c9f79f3ffdc458298dcda1ce632d814 : SUCCESS in 6m 32s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a750e3a6e808445dac23f6be6da21de3 : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3b23a71a8d92405ea8e1ce6a1645a878 : SUCCESS in 7m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/59f56c8728fb4dcf8b4c94d444bf4b1b : SUCCESS in 14m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9465eb59e26c4f75b1197187fae7c536 : SUCCESS in 50m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/716330b6192847eba22b3c631ca33355 : SUCCESS in 1h 16m 02s","accounts_in_message":[],"_revision_number":1},{"id":"db434d816c7db3e90007fc6e30570152f7b76583","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 23:35:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"54343a56579827e66dd507518685995ec11b6e43","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 23:37:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bdd5563c66e495cb0bb8f9a2e6fb593 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"feff65c8efc3d498695011315bff7c655d67883c","revisions":{"feff65c8efc3d498695011315bff7c655d67883c":{"kind":"REWORK","_number":1,"created":"2022-02-24 15:09:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/38/830838/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/830838/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/830838/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/830838/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/830838/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/830838/1"}}},"commit":{"parents":[{"commit":"4c26c59194333b0d559c8d494993e15e529e09e6","subject":"update constraint for python-designateclient to new release 4.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c26c59194333b0d559c8d494993e15e529e09e6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-24 15:09:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-24 15:09:56.000000000","tz":0},"subject":"update constraint for python-glanceclient to new release 3.6.0","message":"update constraint for python-glanceclient to new release 3.6.0\n\nmeta: version: 3.6.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Abhishek Kekane \u003cakekane@redhat.com\u003e\nmeta: release:Change-Id: Iaaa4188cc856a694ab6745a48f8d3c84da0c089e\nmeta: release:Code-Review+1: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I9a3d025658263da6e44581d78c2fbacf34f89944\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/feff65c8efc3d498695011315bff7c655d67883c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/feff65c8efc3d498695011315bff7c655d67883c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
