)]}'
{"id":"openstack%2Frequirements~830983","triplet_id":"openstack%2Frequirements~master~Idcfc038c687877594840bf0ccf480196cae31641","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Idcfc038c687877594840bf0ccf480196cae31641","subject":"update constraint for python-ironic-inspector-client to new release 4.7.1","status":"MERGED","created":"2022-02-25 12:58:09.000000000","updated":"2022-02-26 16:04:13.000000000","submitted":"2022-02-26 16:02:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830983-new-release","meta_rev_id":"216db3b4168ee61fa85837849707ffe6588b8fa0","_number":830983,"virtual_id_number":830983,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-26 16:02:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-25 14:56:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-25 13:02:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-26 11:44:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-25 14:56:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-25 13:02:29.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-25 13:45:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-25 14:39:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-25 14:56:00.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-02-26 11:44:05.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"12c199acf3574d629a4ab28df59a79dc724c15e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-25 12:58:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e4d1f70fdde9330ab3eac33738b47edc05565a8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-25 13:02:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d57ef97f4e2f50423a70d7c2e3c744423979e498","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 13:45:29.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1a7d6acdc14645cf9b0301dcb966ffdf : FAILURE in 16m 40s","accounts_in_message":[],"_revision_number":1},{"id":"a1657275f51835cc3440afa9601975bbecdb161a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 14:39:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca9d63ce1aa841c69b54eedafb2ed203 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eec198f74124c43b568da93fbe8252d : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82afa1b357bf4c48b19a7d5e95c6ab25 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3c0ada2bef8441b9b555095a8cdff9b : SUCCESS in 4m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0bbf23ba8a5240a5b97c322e5b33b9e2 : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9f513085c47c464d9c29941bf913fc86 : SUCCESS in 3m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3b14958ec4d4458aa0c2cb5fe1ecf52b : SUCCESS in 8m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/69c8e336ce1940a5b0e89463377db768 : SUCCESS in 9m 13s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/49b312a2449e467891a119c184abe560 : SUCCESS in 5m 48s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a70a2371c60e44e984e4767703d198bf : SUCCESS in 11m 21s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/58a260d1acc942d0a1c800d2a1d97055 : SUCCESS in 5m 02s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b60c77fc7e2b48d98633b78c9671a589 : SUCCESS in 7m 54s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ed19ff53446945e48fb163a17e107ebb : SUCCESS in 12m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7f4937cacea44bde860970e7e6216269 : SUCCESS in 8m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d834c1b5217344c4bbc3614bc0c14bbc : SUCCESS in 8m 23s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ba8feaf5099e42ddadf6188e84356aae : SUCCESS in 12m 49s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/556d771686e54e43aa3697b16da392f1 : SUCCESS in 4m 20s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/609823ad364b41058b9b7792ba23678a : SUCCESS in 25m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a9f9257febb0438b8912a75aca470d65 : SUCCESS in 5m 31s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/5e77b4a33db04fb89ff64fffb5d8eaff : SUCCESS in 26m 37s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c75686ab8b3f46b49cec885afae3a70a : SUCCESS in 12m 18s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f4622f193f10471aa6c0f74f5b281055 : SUCCESS in 5m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f6d03ffb85794e358e355ef4eb825d82 : SUCCESS in 4m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/81936a82a53345a494bbadf821dab79e : SUCCESS in 22m 14s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2febabba4f9d4d7da0c944db7f35de47 : SUCCESS in 5m 57s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e29d59e0eb7c4a52a3abe0fa65fe0c51 : SUCCESS in 5m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/85dae7bce234453187abb1335e8db8a8 : SUCCESS in 7m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4104361c636f420e8b61e06fa41e4980 : SUCCESS in 12m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c98f31a11876406dafa84c06ad625a32 : SUCCESS in 48m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b1b93490608745be8320de6a28b8a39e : SUCCESS in 1h 32m 33s","accounts_in_message":[],"_revision_number":1},{"id":"e58a7ad7f1539bef40e9791f2a652087c591ead6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-25 14:56:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ce7e4773c279e293d4c7c9257c7df95f0586ed7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 14:58:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d09bec69e137040ba59d22c728d2770e10d50ade","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 17:39:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9bddc75739f4aba9fb2840788ca3892 : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d02a9e6e69194d0bb434526d8cdd43fe : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c8bf29066e047768b538e92faf2d3d7 : SUCCESS in 8m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f0dbcabd7934024ba84b105d3071afe : SUCCESS in 7m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0200f3e4b5024493ae497bc2aeecd733 : SUCCESS in 6m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/149ce2bc02774b96aaa7e6352e7c905b : SUCCESS in 8m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/606e651b9a874d5e9853332e46271353 : SUCCESS in 11m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e239b0d36bf64de2b17b77a806c6d92f : SUCCESS in 10m 33s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/471f7b828ac94e8e9b4045e04d9379a1 : SUCCESS in 12m 45s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/31657150f7324497ac50f8ad5c35aa52 : SUCCESS in 8m 29s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b8b729eb6a464694a17505cf3da8fd88 : SUCCESS in 8m 15s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/998c6ca50fca482ba21b619d89bbf9c9 : SUCCESS in 11m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/996550db8fde4507a66390cc86aab04a : SUCCESS in 10m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2f31d2f2df464c46b856995d0a8ef747 : SUCCESS in 12m 21s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/15b37825b2364de484dc7ef202f817dc : SUCCESS in 13m 34s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/bfec060610664cc48a4d66de57b2b858 : SUCCESS in 6m 43s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7db8e60a18f9411da28b4824532e1652 : SUCCESS in 19m 11s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d274a96f528940c791562a4033d62d56 : SUCCESS in 7m 16s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2a269376630943d199f6599adff159fa : SUCCESS in 26m 02s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/de42cfff1b7c487a9a4302a0f0906564 : SUCCESS in 14m 56s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/94d44d3da07248368b3c84e92ce1dbca : SUCCESS in 7m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d2caa93196ce4b81bc61645e0c8ab461 : SUCCESS in 5m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9d82708e3247437db3d5065102d5883a : SUCCESS in 22m 23s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1f1b2aab85b04712ae46b4998dafc029 : SUCCESS in 11m 36s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/add085e406a341f899c15afc40704935 : SUCCESS in 8m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6bb24e73fd6a40ffa2d3ccd63d1bfb67 : SUCCESS in 8m 47s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/36d0a8430b2d47d2955f7102533d86b7 : TIMED_OUT in 42m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ad80545d8035431bac1c0a8bfb322f21 : SUCCESS in 59m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c42edeaa53134a0e9bd18b7fbe2ed02f : SUCCESS in 1h 05m 32s","accounts_in_message":[],"_revision_number":1},{"id":"8f6418c26231bce806d00481cd325972fc5513bf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-25 19:51:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7fc7e9238f069657c8e1539c354ad58c93ccfbfc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 21:30:57.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/34a45952c6a140bfafb97927b7e0d1b5 : FAILURE in 15m 49s","accounts_in_message":[],"_revision_number":1},{"id":"d4704fa96ed0c7b201e47bd2593f9b30101ce2f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 22:13:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50f571dbed884db293fa4f916804996f : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dca716502b45475181cdf9f247f7d3f5 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ed4a9e3623946ecbd2dea53bfa21096 : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/109abf4bfc0947bdb1f995d67272c584 : SUCCESS in 7m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7843f8ff6eef47c3b03511084edc2ba4 : SUCCESS in 5m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9502d220c5cd48c3930be713aa067dc4 : SUCCESS in 5m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1d45adc001094cadbd653e9b332d6f10 : SUCCESS in 12m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/578bbb8b82fc4ae8ae4337acbfb03f65 : FAILURE in 5m 50s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/921b2779a57c452f82a1bf55c939d120 : SUCCESS in 6m 53s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/eadb5fbcbab346e8ae53756e35bd1f90 : SUCCESS in 11m 27s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/b6ff62921d984c9fafce9a3c96c99866 : SUCCESS in 6m 08s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f866fd614587438bb27b0cf99b04bdbe : SUCCESS in 8m 10s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ee0d08120e1341dd9a82612c3554471a : SUCCESS in 12m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9a3806fc0f544939864e1490220a8f88 : SUCCESS in 11m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a72ddb9bac5247f599fbffdaa8ad9b34 : SUCCESS in 10m 54s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e46a37af892b46a1882058c42c5b2507 : SUCCESS in 16m 46s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d9a9d8380c8b47678e7b3192e12d3ecd : SUCCESS in 5m 05s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9adc72dac7d245f1bc8e1edc0d4edcb9 : SUCCESS in 21m 20s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/fd4aae40848141808eb13c3cfa168287 : SUCCESS in 6m 07s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/51e9332b86ed4733b9bc73801fbf41d4 : SUCCESS in 28m 42s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7d8056d7c0664bed9e45233d2a71e7b6 : SUCCESS in 12m 56s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d37504241ca3403d98c2ba38dbda9c32 : SUCCESS in 8m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/65d886acc32e4644aa5c6d9accf881f4 : SUCCESS in 4m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/464f3f88703d4dd8a9110a6889ecbaae : SUCCESS in 21m 56s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7d2536f974984c718b5db15c8b12972c : SUCCESS in 7m 49s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/548f2107ddb74001869ebf909358e245 : SUCCESS in 7m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/80b4513f587542b2aff08cd3c74669fd : SUCCESS in 8m 02s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6a772c5cbe8548ca8ceacb0a64700237 : SUCCESS in 14m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/81d38f6e3dfb489d92f09178808c476c : SUCCESS in 51m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4f6774d41e24a10bd8384b42bff2571 : SUCCESS in 1h 27m 10s","accounts_in_message":[],"_revision_number":1},{"id":"d9dc021425f1dcf413ad066e16920836c730cc8c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-26 01:55:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"76955c80b8c3087d215ac7545fd6d28227eabb03","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 06:26:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a38f68cb9c9640aeb2302e3eb2349dd7 : FAILURE in 16m 14s","accounts_in_message":[],"_revision_number":1},{"id":"2cdcc8cad1fea2e538aad9567db0a65a2e2aedc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 07:38:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3c6644ceaa2418598ebb62cea11822d : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95efc273111f45228622a54e4e7982f6 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4a73791bb0640c9a156e57b287d241f : SUCCESS in 6m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c02598477c9440f9faa50f0e518e1a2 : SUCCESS in 7m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d9e83ec317e04a49a9d09565c3606b45 : SUCCESS in 6m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b027f9beb4584f9e972fe28f8d14d798 : SUCCESS in 6m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fb5edab9557b4c7795e237a9dd144302 : SUCCESS in 11m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6dea48a3d9114d5fb27e1ab1afca5fa7 : SUCCESS in 8m 17s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/8c1c03a557ec43f0921df488c43234dd : SUCCESS in 6m 22s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d9b85f7be55a4d808daeaa3ad879fc0b : SUCCESS in 13m 56s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/bf178d8872a446b8b61bcbb47fd5b550 : SUCCESS in 7m 55s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b192b69afa88453c959512e59a703256 : SUCCESS in 8m 17s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/172387580af74329956e79288339726b : SUCCESS in 14m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2f0424a58c714a7bba6473e8cf023dcb : SUCCESS in 9m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0d315a08aafe4404b3643d423b39e034 : SUCCESS in 11m 06s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/85f8eb3ff2434d4e9b5df0324db74c4b : SUCCESS in 15m 19s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/82b5ed1c6eb848dbafe3d8952f3f0219 : SUCCESS in 7m 26s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/74efa999c1194b40b7aa8f3b37044e66 : SUCCESS in 11m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d93b30f0570c49d6b61630ee57d9fdf5 : SUCCESS in 8m 47s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ae7a368dbf1f483c90375176599a628e : SUCCESS in 28m 22s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7957d4225147440cb989d851820bcc00 : SUCCESS in 15m 24s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/adbc8dbd66c94678a7220a95ded7c819 : SUCCESS in 8m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c78a1a8dc0e14b9d9cd834d3862dab13 : SUCCESS in 8m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d5788e7a568d4599a6840d755f410a95 : SUCCESS in 21m 01s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/746b59fa640d45cca34f85c0b34defc3 : SUCCESS in 9m 44s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e99f202b95b143989ca787936e3663bc : SUCCESS in 7m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/11f246efa4a6423296588371742b6c43 : SUCCESS in 7m 11s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/175d338c964746ce87b3e9da5850449a : TIMED_OUT in 42m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e225cbc136c45909a3842d456c4a387 : SUCCESS in 48m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bd8008d3469845edbe94567c2461f90d : SUCCESS in 1h 34m 58s","accounts_in_message":[],"_revision_number":1},{"id":"61a6599c4b9dcc117ee641c70ebc6475d1a80872","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-02-26 11:44:05.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ff580d586737a6f30b07371809866df529076e23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 12:01:28.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ff724b670ede4eee8b085e4eedfffa6f : FAILURE in 15m 33s","accounts_in_message":[],"_revision_number":1},{"id":"42b9b08af7691c10a8bfcf9ec27f9f42fcf39012","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 13:38:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2e95b784a5e4251ba35f794c65a4697 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9057bdb6d424c59a098ffbd2cff2388 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5b4177013a24c408d3c9536cf2f8f91 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9f43db63b894b4aac68084bd19f1d3a : SUCCESS in 5m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fbb3635befbc41d1a9da3a707308db61 : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3f2516c1eea04991b42feb23b4375318 : SUCCESS in 5m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1d73f8e5a3604ae481e5d74900ae777d : SUCCESS in 11m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7d0bdd19d886468a978beb30a9f225ad : SUCCESS in 11m 38s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/38e1a699032c4cabb215ba6e82da6ff6 : SUCCESS in 6m 56s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9185bf4ff5e349b8aac16501199320f9 : SUCCESS in 14m 40s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/86c7de9dc8f044a7bf5857dc215c31d3 : SUCCESS in 6m 10s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/cb7b960c7d554c629b223a23e0ebb600 : SUCCESS in 7m 12s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c4031521c52146bca527218ccb4874c7 : SUCCESS in 13m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/84188e3f1516410a9b2733e3a5dde920 : SUCCESS in 10m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f445422c29904004a3b13d648ec36581 : SUCCESS in 10m 45s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a44cfb32a8874a4298f2d72cb871a0fe : SUCCESS in 13m 25s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/5e3c2559842548caaba6dac1a6b504b3 : SUCCESS in 5m 39s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c5417429f14c42669ad0e0a721be39e2 : SUCCESS in 11m 58s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5c9a471aaac745d1b641c6bf9f2dbff1 : SUCCESS in 6m 53s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9a84f327f73d44c4af531d353042aaa3 : SUCCESS in 28m 35s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/14369f367a604f7ba4d73bacd4cd71ab : SUCCESS in 15m 01s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/79334f9b28c841308d45e989a080c7f3 : SUCCESS in 6m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/44b10dcd74cf4e4ab5150c081a489d64 : SUCCESS in 5m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/779365ae919b47629d85eaf23eded214 : SUCCESS in 23m 08s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/8d97b59d82fa4ae1a3409337a16ad335 : SUCCESS in 7m 22s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b205a2d9ed1843ac9c690e18d1ca84a7 : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b6648461379149079361efd7e20ba414 : SUCCESS in 8m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cc688eb564ef44ffabd17cd47f350866 : SUCCESS in 13m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/62d3481e864d4931afcb1cad3a4c8817 : SUCCESS in 45m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/371a7e9fa2db4f0380a7291aaa7e0e6d : SUCCESS in 1h 46m 52s","accounts_in_message":[],"_revision_number":1},{"id":"2e3d9f8715e6224d6733d45a7dcbb15bb9e704f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 13:39:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5d2d005ea18d6337da845abf0be046904f547fa0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 16:02:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65306c55e6fa4ee6b260028af60fa33c : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03095eaabfd944e4bb3339da2bf09626 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/455d03ea32d64bf3974b54b0d1c366d9 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34d194f54319460cae3aba06fb166e59 : SUCCESS in 5m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b7a0ad4620f34d159ca0c47ece390796 : SUCCESS in 5m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0e267797d0b14a2bb60aff28c1484e1f : SUCCESS in 5m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/121300cf876b410788a114fcdb2aff20 : SUCCESS in 10m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cbeb9e8ce5404c53bc931decac5eec59 : SUCCESS in 11m 07s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5316c290fbf440ec8db57307c6a9d763 : SUCCESS in 12m 08s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/9c17e21f435449118ef3bf05ca56dc3f : SUCCESS in 6m 31s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c9ad47235e624bbea1fb56e1c1c09659 : SUCCESS in 7m 24s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7bb2bd9a6aa94f96b58fa62fca964257 : SUCCESS in 11m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3695abedb0e04d8bbeb6a78949ed3f81 : SUCCESS in 9m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/651754045d11427b8d4d5c80d55b3523 : SUCCESS in 10m 17s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/81076ab24c2047a785ba9b1beff51025 : SUCCESS in 14m 05s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/623df93e0ad649c59312305d65a1d9b3 : SUCCESS in 6m 04s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/085979964ed44ae397604852cdac4ff2 : SUCCESS in 10m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/39df970f7d734decb66b72402cc09ac8 : SUCCESS in 7m 19s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9d7889b756af4d588f28fae6678e9ba5 : SUCCESS in 22m 19s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/82f98fd8b53546329305bcffeb27f2ee : SUCCESS in 13m 43s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e17f83c30e204a0caf96909d6539585b : SUCCESS in 6m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fd558c98cb094261ba56b519b8447096 : SUCCESS in 5m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a7076e879b814f5ca8efa3bb678c3de1 : SUCCESS in 19m 44s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5f1d556f471e470d8600a220ab6f7c55 : SUCCESS in 8m 52s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/eca9b3708b0d4bb58aebfc6d57cbf2cc : SUCCESS in 6m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5de1f70520f2491e8d2659ea32342a14 : SUCCESS in 9m 23s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5670e96426bf41039ef9700385ada585 : SUCCESS in 13m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04d830a841164a0d93082bfa3bb5e821 : SUCCESS in 46m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/28484f9408874050b01c23c6f68dfefc : SUCCESS in 1h 21m 09s","accounts_in_message":[],"_revision_number":1},{"id":"fab608fc76836597a06f6d422813efa13887de81","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 16:02:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"216db3b4168ee61fa85837849707ffe6588b8fa0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 16:04:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73bed2c21ebc4bfa9f0270bbb40ae408 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"79c7ef7e61734394cc8d497872daf55953fba4d3","revisions":{"79c7ef7e61734394cc8d497872daf55953fba4d3":{"kind":"REWORK","_number":1,"created":"2022-02-25 12:58:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/83/830983/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/830983/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/830983/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/830983/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/830983/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/830983/1"}}},"commit":{"parents":[{"commit":"39e0f20f9f3f15fb434f64bf9b4fddae895c6777","subject":"Merge \"update constraint for python-swiftclient to new release 3.13.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/39e0f20f9f3f15fb434f64bf9b4fddae895c6777"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-25 12:58:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-25 12:58:07.000000000","tz":0},"subject":"update constraint for python-ironic-inspector-client to new release 4.7.1","message":"update constraint for python-ironic-inspector-client to new release 4.7.1\n\nmeta: version: 4.7.1\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Iafef13594960668cd5b2d46e34bb5bdfb00adc93\nmeta: release:Code-Review+1: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Idcfc038c687877594840bf0ccf480196cae31641\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/79c7ef7e61734394cc8d497872daf55953fba4d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/79c7ef7e61734394cc8d497872daf55953fba4d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
