)]}'
{"id":"openstack%2Frequirements~830991","triplet_id":"openstack%2Frequirements~master~I01973c9c80532c49365deecb223b0266fc518478","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I01973c9c80532c49365deecb223b0266fc518478","subject":"update constraint for python-heatclient to new release 2.5.1","status":"MERGED","created":"2022-02-25 14:13:53.000000000","updated":"2022-02-25 20:58:36.000000000","submitted":"2022-02-25 20:54:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830991-new-release","meta_rev_id":"f21ee918fb1a874c6b7c95d9f9be874dcb08acfa","_number":830991,"virtual_id_number":830991,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-25 20:54:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-25 14:56:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-25 16:25:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-25 14:56:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-25 14:50:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-25 14:56:02.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-02-25 15:00:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-25 16:25:25.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"aad953c2f20c0b241aa66f8b5804f8dd8771f70c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-25 14:13:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"198b0b161b0fd1181c6a3b6097214daffd0dd325","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 14:50:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/bd75684703404a8ea7aa92db125bc45c : FAILURE in 18m 56s","accounts_in_message":[],"_revision_number":1},{"id":"465898cfc8bacced839f47e19ebb41defe96c372","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-25 14:56:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"77349ea4667b0f094c5ca381f4e134f4b3fe1989","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 15:00:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c9b9f41ecdc4ba4a49c986c37ba1906 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c13320d26321491d845486adcdc51b04 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7514e875062e439788b56005458b9b0f : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3c3c0982356404485f479328fb54f4b : SUCCESS in 5m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6f5e8e024235415f8f5e29448410a746 : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3eefe50bafa94755a16229e4eb0b7c60 : SUCCESS in 4m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/707f315ee1014404b46f68ae7d564d7d : SUCCESS in 10m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/907845c0e6094550895878278fd6cec5 : SUCCESS in 9m 27s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/96ad09ca2c2344ca9765da6a06ca719c : SUCCESS in 5m 49s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8fdbc73743cb484086eaa078d2ac9b61 : SUCCESS in 10m 31s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/01bda8f71f684393982cac02e95941ed : SUCCESS in 5m 22s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d6ba0afd3a194ee7afbb2738cfec5deb : SUCCESS in 6m 00s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7ffc1965c5d9484c9d6bc70e0c7d0289 : SUCCESS in 10m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0d9fccc5a9a0425d9c1e7f86100e7fe7 : SUCCESS in 7m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9fbb6f00a3b346ccbdb9489451e45488 : SUCCESS in 7m 50s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/00fe750c4c124f11a2a78ae30696d96e : SUCCESS in 10m 59s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/fe152d6c68c14ae7bd28c0c9cdcb8af9 : SUCCESS in 4m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/821c9d1432814a6786b77d65b8ed73fc : SUCCESS in 16m 06s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/951575407979472688cd0bf3f76b8374 : SUCCESS in 5m 38s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/46d220dd5d6a4ae39be071699eadf65e : SUCCESS in 22m 46s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3ac47e9b94894262a8677e636694c61d : SUCCESS in 11m 31s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/acdea62e33a94f4e8831b782fbc8827a : SUCCESS in 6m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/08d98901c9024a3eb0b4522df758a29d : SUCCESS in 4m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8643b36423dc4a4faddd46ab5d6a2569 : SUCCESS in 17m 55s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b62f2acf05c04826b3f203293c0d843c : SUCCESS in 6m 02s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/62f56127549a407a9d062a38f25b2f26 : SUCCESS in 5m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8b47f9ee6cfa426daaa9fca39c335740 : SUCCESS in 6m 54s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/adabeb16360e4dab9f3839b98414d341 : TIMED_OUT in 41m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/33cb733940f34289bd7a4454df153c04 : SUCCESS in 40m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4fdd239997024348b33748405b244e61 : FAILURE in 20m 50s","accounts_in_message":[],"_revision_number":1},{"id":"807dc6b30effd751732204c06becd93aa40c0ed9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-25 15:06:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cda58d5131630dc00ef6a0aabca96f41a29e3704","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 15:39:40.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6cb34f8d0ec74cd2853d2631c100ac2a : FAILURE in 15m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b1c8a54457076065c91a36465b1e706b67df6e8f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-25 16:25:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1b968d2b9f161ce5ea53f1be15c7fa5bf87009a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 16:38:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc55838948264828a41168875305d49c : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3ed95602ae74449b8796fd928a77e36 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9218c8020f7433db7f9c4bade39a17c : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebc7cd5d61df41d29e9c6d8ec08769a0 : SUCCESS in 7m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6f12aec5edd247efa302fbd28d6a9a20 : SUCCESS in 6m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cb785b41d81440abb7f17ec903651abf : SUCCESS in 6m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/40c4152064854038891353c30dca9a52 : SUCCESS in 9m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c8b46a21f9bf4f6e88c1d3a89ace31c9 : SUCCESS in 9m 54s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/cd9b39ad8e0b40dd96fc48da31d8e968 : SUCCESS in 7m 39s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b3f9f24544b7445391fe1eed24429ddc : SUCCESS in 12m 47s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/8384853a03bd46c686cf8d6bf6f617d7 : SUCCESS in 7m 38s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/bae0a3fd3d024e63aad649899d3fc8cf : SUCCESS in 8m 05s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6695fda9c6964b87a778a99789c62385 : SUCCESS in 13m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9a9e95f4eca74257a9e119990bba8c9e : SUCCESS in 10m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/228708827f724f309759982dcd7dbdea : SUCCESS in 10m 55s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/908ca286f3324e259047f2863263d4c6 : SUCCESS in 14m 07s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/59b87f92778042949cec45e7555dc4ff : SUCCESS in 6m 34s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7f142a55791448839fed9038ad5bd968 : SUCCESS in 18m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/517b6eb98ef845f49ff507870009e7f5 : SUCCESS in 7m 37s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/75a3160ca83742caa4d7a249852bebce : SUCCESS in 38m 18s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/595f262025624b5bae8104b7267977d0 : SUCCESS in 13m 50s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/39597b1bb3894763a8d911e8708ec49a : SUCCESS in 8m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f9d4c7a560134cbfbea0ab3b9fe4ce11 : SUCCESS in 5m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bf146b3f267648c6a0e0b1336e8c2b2a : SUCCESS in 21m 34s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/36e0d55a36f8488eb88cfbc9a724020c : SUCCESS in 7m 43s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b115ea785c9a4223becdc0d7a4ebcdfb : SUCCESS in 7m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dd1979ae3d0d43b481fe1da4e80e7810 : SUCCESS in 9m 24s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/77b1e6c1d6a74ec08d54c25770d793b3 : SUCCESS in 13m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc2dc194f19a4dfa92e91a8293afe28e : SUCCESS in 53m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be3df10b868f45c28f188b6d1b8a78b9 : SUCCESS in 1h 21m 15s","accounts_in_message":[],"_revision_number":1},{"id":"864e788734867c8fbc29e15650de5584ef55e704","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 16:41:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"675b7195ece38c462727f1d15ca0cf76511d100e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 20:54:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc01454ae0ec4ca2a1563442ad548b1c : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8133cbead524df69e50ea1ff1165877 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d136f0b8adfb47b7a659916d5099cf60 : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1f77237155c478c9dd1a8414c39d436 : SUCCESS in 7m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/856078828ebe45ddb6e10289ddf7587a : SUCCESS in 5m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/16625423a9a64f3282a4815126d91c85 : SUCCESS in 6m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d35b2037ba8b464287cf2505be55d2af : SUCCESS in 11m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5d6a885930d94e4f8bb6125e1c258697 : SUCCESS in 10m 28s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7956d31c778d4bfab192fba26019d9c0 : SUCCESS in 12m 56s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/1a647945a1fb40ee963219759d0422c8 : SUCCESS in 6m 52s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4c43725c115a487aa667a8b9d3040e5f : SUCCESS in 7m 51s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7ec2cce6ae89495ca7fc00bc66d41733 : SUCCESS in 10m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9d219262d50d49e09e6033908c354e7a : SUCCESS in 10m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f20db2d9d4154f3da4c05c345ac12c47 : SUCCESS in 10m 14s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1cd42ae048e647d7a40e8f175009a652 : SUCCESS in 16m 49s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ae7a8af6b4b644c390567ca57133f224 : SUCCESS in 6m 06s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f96b0f9c08d644a8aadd7d2bffd11f91 : SUCCESS in 17m 17s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c2986a218bee4f8fbefa3f25157fc2f0 : SUCCESS in 8m 17s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c936e004e6a74720af33ac38f7285092 : SUCCESS in 25m 57s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/60434e15849d49d4803c9ee0276ba096 : SUCCESS in 15m 09s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/482ce0f7662645fcb735eba372015db1 : SUCCESS in 8m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dab385854f5a465f86a01c54f3f27653 : SUCCESS in 5m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/369751763b6a4e6dacfa7c28b02c04c4 : SUCCESS in 29m 27s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/74dbdf5e38a84336857dfede0e8641ea : SUCCESS in 8m 26s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/30d01a7f938042418ce6a7a381484ee0 : SUCCESS in 8m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a37bb2c7b44140c6a2ecc3c8fe4244fc : SUCCESS in 9m 14s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c8dd71fe66314f05ae2e69f7178b61ed : SUCCESS in 13m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/120e82a19d8c4755aa65f28b13053345 : SUCCESS in 51m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/652eea5bddd44b39b7d1abddda890930 : SUCCESS in 2h 08m 21s","accounts_in_message":[],"_revision_number":1},{"id":"ae54d551c1d73f0b6f14c39c30f3a383c83db650","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 20:54:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f21ee918fb1a874c6b7c95d9f9be874dcb08acfa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 20:58:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4f00943d82f4609a1fb47d69bf47dbb : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f239d032ae8e48c3b8055ad20322673f0e67d9c0","revisions":{"f239d032ae8e48c3b8055ad20322673f0e67d9c0":{"kind":"REWORK","_number":1,"created":"2022-02-25 14:13:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/830991/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/830991/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/830991/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/830991/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/830991/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/830991/1"}}},"commit":{"parents":[{"commit":"39e0f20f9f3f15fb434f64bf9b4fddae895c6777","subject":"Merge \"update constraint for python-swiftclient to new release 3.13.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/39e0f20f9f3f15fb434f64bf9b4fddae895c6777"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-25 14:13:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-25 14:13:50.000000000","tz":0},"subject":"update constraint for python-heatclient to new release 2.5.1","message":"update constraint for python-heatclient to new release 2.5.1\n\nmeta: version: 2.5.1\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ieb79b38d315170d72f3499d7ca2b9b6891b848d0\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Rico Lin \u003cricolin@ricolky.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I01973c9c80532c49365deecb223b0266fc518478\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f239d032ae8e48c3b8055ad20322673f0e67d9c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f239d032ae8e48c3b8055ad20322673f0e67d9c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
