)]}'
{"id":"openstack%2Frequirements~831003","triplet_id":"openstack%2Frequirements~master~I9521640106bc85e79136b2b3c6e6f2be301662da","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I9521640106bc85e79136b2b3c6e6f2be301662da","subject":"update constraint for python-magnumclient to new release 3.6.0","status":"MERGED","created":"2022-02-25 16:13:31.000000000","updated":"2022-02-26 11:32:03.000000000","submitted":"2022-02-26 11:32:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831003-new-release","meta_rev_id":"c845b333775ef6f46cbc7793b4e27cce504dc675","_number":831003,"virtual_id_number":831003,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-26 11:32:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-25 16:46:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-25 16:25:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-25 16:46:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-25 16:25:14.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-25 16:46:30.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-02-25 16:51:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-25 18:03:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0f520da462310f98df6dc1a95d7da41228a97167","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-25 16:13:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29bf034b6c73c54f44c60dcb2cf07a6d392f197d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-25 16:25:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"82d630372586fe03f6306cda6f227cc4399ac319","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-25 16:46:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"01005044dac0abfe699f27f29cb83bbfd00caa48","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 16:51:53.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/45fb12ee53c74d729a6a5a710a2ee3c7 : FAILURE in 23m 19s","accounts_in_message":[],"_revision_number":1},{"id":"01173593d5425093181c8f0a5103e5c215d82abb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 18:03:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4afb232ebf5452593bff695b28ecab9 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9566c240ec024bdca277d7dfe39baa87 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb8e445082ea479ab77c8d52e7b7738b : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68265df231b642c4a5de8ed09c57e495 : SUCCESS in 6m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b32ca73612624b84bbb4c97f6b37d6be : SUCCESS in 6m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5c6aaf473ce746cfb427d2af6a51156c : SUCCESS in 6m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2a635b49782e47378c25a4d03cb53f98 : SUCCESS in 13m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8c0cde5ca8464276b0b16437fc7e22e5 : SUCCESS in 12m 12s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/00820b27d0614fdc806c9c3251a67656 : SUCCESS in 8m 16s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8aa1824f63644a0db21f1b8455128bf3 : SUCCESS in 13m 40s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/8b829e1fa4924b9e8ebea6fb43e13acb : SUCCESS in 7m 21s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/237de041a7f84c0fa73e7c048421cdb9 : SUCCESS in 8m 08s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/548596ec79304f4597d4f081643fa541 : SUCCESS in 12m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/488c4f4a033b4ebf99634c8fad00d4dd : SUCCESS in 10m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1618edcdcc1e4bfdbdff75b11de3791a : SUCCESS in 12m 34s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9a4b2441fc554cd08eb19f3402ca9352 : SUCCESS in 14m 08s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/52044488052b487ca20152a49bc99d56 : SUCCESS in 7m 30s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/dc3029158e9b4fdb8caa9cfd96222b0c : SUCCESS in 25m 42s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/eefaed3fb79d49b6a8b20fa48d1fe99c : SUCCESS in 10m 07s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ac4157f20ef24471af9c9d34f60be348 : SUCCESS in 26m 16s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/06a6cd03d0c34119b7d520a4bebd3422 : SUCCESS in 16m 19s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/563ba895c36247c9b499fcb441aca4ff : SUCCESS in 6m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/abbbf0dbcb81476b973d5168889bc739 : SUCCESS in 7m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5a0a1cf80dd645259cbb4316efc6c404 : SUCCESS in 21m 56s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/9c7fde653b264363a5087ed68332cd09 : SUCCESS in 8m 48s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4d0269cdfc324ce9bb9c251577ffe5ae : SUCCESS in 6m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/efc291f1a2b249529d3cd3eb97119d63 : SUCCESS in 9m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/98f4dc6bee3848899eb3d597282c3e6b : SUCCESS in 14m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a96239dad12a4fdba019a3933445e930 : SUCCESS in 56m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb076347f9294e209d056f2571d56695 : SUCCESS in 1h 32m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b2c25afd431c1c52ed0ec706a5187f9efc02da29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 18:05:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7472eb9f684d76343cc0a8ee2381e2f7e577f1f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 21:44:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e25d97c8806f4d148bc59dd74ae896e5 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86b214f4416e48929356d84d8cfbec46 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dec3769b2e4488fbebca0b1baeb86ac : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2138cfc5f3814fd1bfe1c895828ae0a2 : SUCCESS in 6m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f6800b486dc04702a6badd90c1eb4fb6 : SUCCESS in 7m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1947d27911a54c03a6f3a7d07efae3d1 : SUCCESS in 6m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ac830056e9944d14a6749a216dcb1393 : FAILURE in 6m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e0b5bb4bbe7a40359497f5b838c0ffb4 : SUCCESS in 9m 45s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/66e166c7766a4268b2098279c86a239c : SUCCESS in 14m 13s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/dee6c507b7df4ee18d3ab76f3eaef198 : SUCCESS in 7m 47s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5f47a3cf80d549d2be1857ea6ec0af87 : SUCCESS in 7m 07s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/09c8e974a95e4b818b18de0c2cdd0292 : SUCCESS in 12m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/71fb9c4a6ece47baa3137bdd121b7e0c : SUCCESS in 11m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/11ccd9de23b54a458f95970da5c6f46b : SUCCESS in 12m 49s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/bd5930b58cb94b568c6cadcd6c78c571 : SUCCESS in 13m 23s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/07d1256e0c3f4df6a87fb54f1937a1aa : SUCCESS in 7m 50s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4e60adbba00b4371ad9311d30f63bfe9 : SUCCESS in 19m 28s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/03aa735703834b1287348e7a3e7b961d : SUCCESS in 6m 50s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/978f0757d45b4a0292ebef47b31cacd3 : SUCCESS in 27m 45s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9e9ddbdd0421489abfb49ddeda0fbb10 : SUCCESS in 15m 45s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c32b47de57184c25a73b6e8af68d4e00 : SUCCESS in 7m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/efd3dcd2bdea4814a1c9dc1fe38a53a7 : SUCCESS in 7m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/666d0c20f139498999a28f90008163bb : SUCCESS in 23m 24s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7815f06498a348288d1fe1fecb50f53c : SUCCESS in 9m 04s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/189cae3684114835a2666ad96fd6d24e : SUCCESS in 7m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e220308012f24df7838b16e51ce575af : SUCCESS in 9m 48s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d67b36dd153f41e5924f80d0c366aad4 : SUCCESS in 13m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5a839e608e2d4b4381a29eb853da8252 : SUCCESS in 52m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a3febaecd4fe457993e1261109616d73 : SUCCESS in 1h 09m 58s","accounts_in_message":[],"_revision_number":1},{"id":"746747917e5adbcf42358406ca1b688a077f9c62","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-26 01:55:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"23e018e5adba5e1e5ec0caf50576bfa1657a2509","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 06:26:00.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d398f97eb91f480fb6021202092a0409 : FAILURE in 15m 31s","accounts_in_message":[],"_revision_number":1},{"id":"f112790f1070128aec6d285dbb26086d0dbd955f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 07:40:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8d3485e9d99481a97b7e05ddc93d24d : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b39b07c6042414f9638036a8e26fae8 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ebe7a4a2f86409280fbf87f9c68df9e : SUCCESS in 8m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cfe31095ad054e73ac6ab88563af1c40 : SUCCESS in 7m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8223a4be7ad349819de9b168b5834626 : SUCCESS in 5m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/73c9da1e2db34fc1ba16169bcdfd3d63 : SUCCESS in 6m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cc34c8c49679478785b263e81c25644a : SUCCESS in 10m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4a083bd29fc54d3cbe8e555d8035b7e0 : SUCCESS in 13m 55s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/9ebcc353828648519f6cb146eaa29fa2 : SUCCESS in 7m 45s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6cd91390f64c40a28c884c6672265500 : SUCCESS in 12m 07s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/b7327eacf83a4526b26a4663c3845e68 : SUCCESS in 8m 29s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f016c22313524689987dcf3391aa4864 : SUCCESS in 6m 47s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/beea3e4d714c406f9ffbd45de979619f : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/18f19bdde6334a9ca0ddddf86be61ab4 : SUCCESS in 11m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5e4bef429c5a4e85821191314655d453 : SUCCESS in 10m 48s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/868552d9465646f38b769c8eba51a59d : SUCCESS in 16m 54s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b44f535a724848d5a6477838789ca9a5 : SUCCESS in 6m 08s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/502c192803e14af7b284994d023dc84e : SUCCESS in 10m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a13c24e54fc04197908dc36c0a5ecba7 : SUCCESS in 8m 37s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6a533bb966cf4ec58878e148bfc65105 : SUCCESS in 23m 39s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d804f4ac2e834c7ca401017c9fac4c30 : SUCCESS in 16m 34s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d55587123ed04cbab9c64b69b124891d : SUCCESS in 8m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/92a99eb5607a495abaf7b5917130c6a8 : SUCCESS in 6m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ba5e503ecaad49f78b518b20868344cd : SUCCESS in 25m 55s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/6418cc8f5bd84bae84a70012d6be6ce3 : SUCCESS in 8m 06s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/70aff2157c0b424f9d4e876e95b2347b : SUCCESS in 7m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/012c444087de47da97d0a2a89045aa09 : SUCCESS in 9m 04s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ca30aca78561492da51f227def5a712a : SUCCESS in 11m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26795b8a4c684c55968e2ba016928fcf : SUCCESS in 35m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/daa9ca430da1427faf5416f0e3676048 : SUCCESS in 1h 45m 19s","accounts_in_message":[],"_revision_number":1},{"id":"6296d7c56c5451427aeab58aa3e7edfe45fadb89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 07:42:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"15a0b5bad2f8cc26f6920163e82dc075ca5d3203","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 11:32:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/499ebb8b626345f5afb6d63b7ba253ff : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b143969a51a46368d1a686561790604 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d247067816ab4da6bd0539cfc7794543 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd7d42d252d04fedb1c7543c5814d1bb : SUCCESS in 6m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8ae96c3a9cd04e039e308b5eafb2d686 : SUCCESS in 5m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f86ce69fc34a4e6daf7a39d692108d84 : SUCCESS in 5m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d220334d0b40444aa82a074a830f24a1 : SUCCESS in 10m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fb17fd66a8c44d179b0b9d4b801bf94d : SUCCESS in 10m 08s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/57d7884281cc4c9c989e96b85a02e763 : SUCCESS in 11m 17s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/61d386e5605f4d749aa4ed3e3568bba7 : SUCCESS in 6m 44s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/180b62dfcd0246ff9c02f6b7f6198c04 : SUCCESS in 7m 13s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/bcdb9054ec734de8b786b451c447cdcb : SUCCESS in 10m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1c6ac6b10fbd40ef8f1d6b5b21a2af9a : SUCCESS in 12m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8bdf7aa4ee5043328e933699a4da6ee7 : SUCCESS in 10m 30s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/85c7e22d319a42f79cfdfd80d28263db : SUCCESS in 17m 53s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e848a1bc460446c4bfe47c74ab1b564d : SUCCESS in 6m 10s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5995d4f80b29456b872ea779ff9d55fb : SUCCESS in 10m 19s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a43959bcdf9f4041994b5f59c946c1c9 : SUCCESS in 7m 22s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/455caac8c52642268fca8d2ceeb4025f : SUCCESS in 26m 01s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/226362b0a2ad4ab3bccdc18241fff2bf : SUCCESS in 15m 34s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/3f678e213548442d969a584079cde4cd : SUCCESS in 6m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1e829b4d920b446d9b3a3fc63d3f9f3e : SUCCESS in 5m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d4a118d2469d4dcd9c09292f0ba03b6f : SUCCESS in 26m 11s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7a8ef863d4654403844dd8466d7870cf : SUCCESS in 8m 28s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1258d1f88b824eecb4c087e9a13eb00d : SUCCESS in 6m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7ee71b1291e548279fcf3600956f8edb : SUCCESS in 8m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2b4703b92bf54247961e3ad73f5888b4 : SUCCESS in 13m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b667a3e7faa34479b1eef66793082639 : SUCCESS in 55m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/466cc8512be945128933bd29f8068cdd : SUCCESS in 1h 14m 16s","accounts_in_message":[],"_revision_number":1},{"id":"c845b333775ef6f46cbc7793b4e27cce504dc675","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 11:32:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ccaafc31ebf4e4d890b9cf2d9ddc685abb50afc0","revisions":{"ccaafc31ebf4e4d890b9cf2d9ddc685abb50afc0":{"kind":"REWORK","_number":1,"created":"2022-02-25 16:13:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/03/831003/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/831003/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/831003/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/831003/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/831003/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/831003/1"}}},"commit":{"parents":[{"commit":"39e0f20f9f3f15fb434f64bf9b4fddae895c6777","subject":"Merge \"update constraint for python-swiftclient to new release 3.13.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/39e0f20f9f3f15fb434f64bf9b4fddae895c6777"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-25 16:13:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-25 16:13:29.000000000","tz":0},"subject":"update constraint for python-magnumclient to new release 3.6.0","message":"update constraint for python-magnumclient to new release 3.6.0\n\nmeta: version: 3.6.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I06048ff656e976977f16e689c3a67f83abb120a4\nmeta: release:Code-Review+1: Spyros Trigazis \u003cstrigazi@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I9521640106bc85e79136b2b3c6e6f2be301662da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ccaafc31ebf4e4d890b9cf2d9ddc685abb50afc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ccaafc31ebf4e4d890b9cf2d9ddc685abb50afc0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
