)]}'
{"id":"openstack%2Frequirements~831004","triplet_id":"openstack%2Frequirements~master~I411b6353abe6c7123e387505354937d171162189","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I411b6353abe6c7123e387505354937d171162189","subject":"update constraint for python-tackerclient to new release 1.10.0","status":"MERGED","created":"2022-02-25 16:28:17.000000000","updated":"2022-02-26 19:19:03.000000000","submitted":"2022-02-26 19:17:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831004-new-release","meta_rev_id":"cd1ed5ed5886dcca38aefb0ca8fa8ad48f7aafd0","_number":831004,"virtual_id_number":831004,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-26 19:17:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-26 16:08:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-26 16:08:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-25 16:30:38.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-25 16:46:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-02-25 17:25:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-25 18:19:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-26 01:56:36.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-26 12:14:51.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"433da8295fbc2b1222fe48df7d1d2d21c24b679f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-25 16:28:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac0d5890cf8f06914f22a6ff59764a3fd67577a2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-25 16:30:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4903f09994858549901de7b9b568e4af94381013","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-25 16:46:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1edb51deb14943c6436c053b629eebaf72ea9ca2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 17:25:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/56280d74b8124f368e16bf5049dc35ed : FAILURE in 16m 09s","accounts_in_message":[],"_revision_number":1},{"id":"3e722c6cdaaf21d919dd5e365c3bc4e961474959","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 18:19:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0336709003314d1299145e6406144118 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53fd6dc271c7403989ed93751e102ea6 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80543d9a157543ec8bdc12aa5c708124 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fab46fdca8624a66bd0bee2eff55f55d : SUCCESS in 6m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cec376495434497b9f8d7d7bd97c32cd : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5fc4d286631746c99d2ae22a801687f6 : SUCCESS in 5m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2624172f747d4eb8a00c77b6f8fc78e3 : SUCCESS in 10m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e70d5607c0ff44d9874155d446437195 : SUCCESS in 10m 48s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/2fac025550874151a68ac735a172663a : SUCCESS in 8m 18s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a5e26a015a7d45fe954082e084f8715f : SUCCESS in 14m 05s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a6e0b39249b94953b552496196608f63 : SUCCESS in 6m 06s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/45883ba87e9d451bb1df5f8b89944cd3 : SUCCESS in 7m 51s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d1537b9f0a424673bc515415e677c557 : SUCCESS in 11m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/77f36efd2716434190d8b39e8dc176c4 : SUCCESS in 10m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ca1244c17685443dacc566c88d746935 : SUCCESS in 10m 54s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/63bcbdd722644c05bed8a4c8b045c8e5 : SUCCESS in 12m 47s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c0bdaf1f285f4c3d91ace8bc1d82b1e7 : SUCCESS in 5m 42s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9455f7334b0640dd9714175526aa5814 : SUCCESS in 25m 35s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c38cc87d33824a1b9f863f02f807cf5d : SUCCESS in 5m 45s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/84bb8958f4ee4d73b20d9e19c33cf1b0 : SUCCESS in 25m 30s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/dd1c1d9a33854b6094713c5918c4b4b1 : SUCCESS in 13m 27s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/87dd1e8e6e3e49e9b33d8931051761cf : SUCCESS in 6m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/951e398f6e7642838561664da889545d : SUCCESS in 5m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cca9435eb76a49598a9f55354daab9bf : SUCCESS in 21m 28s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/62f894f32346408c921194aee851dd35 : SUCCESS in 8m 11s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/eadc9ed54b42423d86b105f0d76145a7 : SUCCESS in 6m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9188791a810442f78c8702ed7e08b1d1 : SUCCESS in 8m 34s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8d91a3b26c864fc3b3fd73002bdc50bd : SUCCESS in 12m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1732578a19f240398b8632694af4e58e : SUCCESS in 48m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb670bb6434b4868973200c9121ef7e6 : SUCCESS in 1h 25m 50s","accounts_in_message":[],"_revision_number":1},{"id":"7c6697a3f68669645cf43773f05f50fa960e3a0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 22:30:01.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"32d6bd0156b8c8d7be5236c8db9eff265286c441","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-26 01:55:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c99a1cebd03c9e5ba1d3885cecfb3ac42dd5e37b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-02-26 01:56:36.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"f9ac84faa4793771a4de6af342c3ac0f2a929f5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 05:49:11.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 831004,1","accounts_in_message":[],"_revision_number":1},{"id":"8f16464e3dcb994786117d085452262dce041761","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 05:50:22.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 831004,1","accounts_in_message":[],"_revision_number":1},{"id":"2ce8edbb5060516e34b58f43ac96581e474229e7","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-02-26 12:14:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0c56fbf03419d1277892451249c9cf7dfb0b98d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-26 16:08:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"101a55f476d061ee2da740012d3e3cc1d1f7f3aa","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-26 16:08:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8ef09e96fe027cf86b495b04d396b4236721e340","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 16:24:34.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7c2c7729e2b84bba8b6899f06876dd58 : FAILURE in 14m 20s","accounts_in_message":[],"_revision_number":2},{"id":"98666ebabb6ae3fe99240fad75b1ce2448b83aa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 17:26:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1aaa6beb821494cb602be836104fde1 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9575bb15abe04610b7be875b2c20dde5 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdca88d7d3464844b21a28f3b3902e4e : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74aad91860dd4d8f96a19bb0f1df3980 : SUCCESS in 5m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c72d990789264c0792f93efa59c375ac : SUCCESS in 5m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ab62097eb496486abc74db15dcc0f05f : SUCCESS in 5m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a1a328a4c4a0410b823e43ab41215186 : SUCCESS in 8m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9920b902093a498abdc6534477ca77af : SUCCESS in 10m 38s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f7de9f3c9eca44a5849b12fdd3b0c5d3 : SUCCESS in 7m 08s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4e439b4ad102458eac1a1736e7a1dd34 : SUCCESS in 10m 17s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/f3faf23f3f37490ab0ef9e47828b7359 : SUCCESS in 6m 10s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/461ef053a91145b38d878e9124cc3776 : SUCCESS in 6m 25s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7cb0d7e204034ff9bec41a66b53773d9 : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4c359058912b4b1094b787c69a0de33a : SUCCESS in 10m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5e837d77bab94400b37287bdccf1a0bf : SUCCESS in 8m 51s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/48d9b6e4a27c4e91bf7e0013a40c33fb : SUCCESS in 12m 03s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/66a8609d38fc4a5f9cfd569d6ef21398 : SUCCESS in 5m 43s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5ee838ca7bb948f9abcfce70e1e2df95 : SUCCESS in 11m 17s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8867e69c5f84448483cfcc07bb1c9e2d : SUCCESS in 6m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ecb4c21fe9ce4a45a937794835032362 : SUCCESS in 21m 56s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/25ded1440bef45edaacdbf0616d0072c : SUCCESS in 12m 54s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/1ecb3838543043f0a377320ddfc3d0fa : SUCCESS in 7m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/633cee02263043319fcf1f7e71708eb5 : SUCCESS in 6m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7968c168e8034dcf9549465bdbfe4f3d : SUCCESS in 18m 56s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e6edd55921914bf9b3f70e4f4f194f99 : SUCCESS in 7m 41s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/99db397c82fc4d0bb04715e3585a5b5d : SUCCESS in 6m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d636730038b24e8985ec6eba910b20e0 : SUCCESS in 7m 04s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/24dbbd0736194239abb7c92fb106c1a9 : SUCCESS in 11m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91ec81d3fbe547959516cfee318cfe79 : SUCCESS in 35m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e7f584b316f24c6a8302dd7ac1548072 : SUCCESS in 1h 04m 02s","accounts_in_message":[],"_revision_number":2},{"id":"8476fe5d5c2906fafc304331f6e5cb959322f7a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 17:27:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"43c52a3d264ec9311d0f2fb39d6440f29a03932a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 19:17:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f56290bde5544e495aa9428cbaca218 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f432fc4f400d4414a92e7fcef105f663 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ce8929b850c40b480c03c75989d6f8f : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2bc3108a1e44030b85f8ecb83fb8836 : SUCCESS in 7m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f0af9c7ec8b74737ab111b0f5e9dc1d2 : SUCCESS in 5m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/899213c1a0144f5fab02dd314b4975d8 : SUCCESS in 5m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f704ca2334f34c14ba393b1e9b522a0c : SUCCESS in 8m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9080c54418614223aec6ec4b3ae6199c : SUCCESS in 10m 43s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d60f6ea0189e4fa5a1253473fa320353 : SUCCESS in 12m 44s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/36bd5cb4c6314b8c9a730296e0c5000a : SUCCESS in 6m 41s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/439651f61834458eaf0b221502890106 : SUCCESS in 7m 20s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ca50766a133043abb75b4d8e8fb35e41 : SUCCESS in 9m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2ea847c437724a33be188a615f146fd6 : SUCCESS in 8m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1864f59cfcd14e39b38310eb94b9fbf8 : SUCCESS in 8m 03s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/15cc60a75efb4215a691f9d3f7754b69 : SUCCESS in 12m 01s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/8ad3a51f50964f17b95ae41eb31c315c : SUCCESS in 5m 08s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/eae9850a069543c5a4e6f38f632c960b : SUCCESS in 10m 36s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c0bd61361d5a4cbb84914a7db5540e0e : SUCCESS in 6m 05s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2f2dfcb4ce774538a88989589afa480f : SUCCESS in 22m 21s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/32a93ca1b32b4e67851df44c3d72e77b : SUCCESS in 11m 53s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/7ed45659a45544488759b90b62692761 : SUCCESS in 7m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cf60a64dd1d341e3a5507a717a40a948 : SUCCESS in 4m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5db321624cea4b44946097bf6c76c197 : SUCCESS in 19m 38s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/20926f6d74484182895d554579240c63 : SUCCESS in 7m 00s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a7a3d77c0d3f4135923ef14d6f2a71ad : SUCCESS in 6m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a7fdbc457bc14883b97ab56e2fea6d6c : SUCCESS in 7m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5fb9286d4c64467d939284c0cb5e4e7f : SUCCESS in 13m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7789c32c6fa1447495b4105220f05687 : SUCCESS in 1h 04m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d53c90445b3b48c38faa10baaa8a64e8 : SUCCESS in 1h 45m 26s","accounts_in_message":[],"_revision_number":2},{"id":"dc263d173c048ad8b85459f534b52b185aff713e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 19:17:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cd1ed5ed5886dcca38aefb0ca8fa8ad48f7aafd0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 19:19:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3e7b2f2953b4b0c8d8fa09ca9568f08 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"94e06ae8ac1c4c0aa00e787dbf140a0ef10b3bae","revisions":{"18a204a300aaf863b90b5d6eecc02424e8fd7f70":{"kind":"REWORK","_number":1,"created":"2022-02-25 16:28:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/831004/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/831004/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/831004/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/831004/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/831004/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/831004/1"}}},"commit":{"parents":[{"commit":"39e0f20f9f3f15fb434f64bf9b4fddae895c6777","subject":"Merge \"update constraint for python-swiftclient to new release 3.13.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/39e0f20f9f3f15fb434f64bf9b4fddae895c6777"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-25 16:28:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-25 16:28:13.000000000","tz":0},"subject":"update constraint for python-tackerclient to new release 1.10.0","message":"update constraint for python-tackerclient to new release 1.10.0\n\nmeta: version: 1.10.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Yasufumi Ogawa \u003cyasufum.o@gmail.com\u003e\nmeta: release:Change-Id: If70bf0935108696900bd84cecd44c635036c2fa9\nmeta: release:Code-Review+1: Yasufumi Ogawa \u003cyasufum.o@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I411b6353abe6c7123e387505354937d171162189\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/18a204a300aaf863b90b5d6eecc02424e8fd7f70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/18a204a300aaf863b90b5d6eecc02424e8fd7f70"}]},"branch":"refs/heads/master"},"94e06ae8ac1c4c0aa00e787dbf140a0ef10b3bae":{"kind":"REWORK","_number":2,"created":"2022-02-26 16:08:15.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/04/831004/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/831004/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/831004/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/831004/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/831004/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/831004/2"}}},"commit":{"parents":[{"commit":"d78bb8c7f23ba52c429e97e1fcc58ae7b974c67b","subject":"Merge \"update constraint for python-ironic-inspector-client to new release 4.7.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d78bb8c7f23ba52c429e97e1fcc58ae7b974c67b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-25 16:28:13.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-02-26 16:07:59.000000000","tz":-360},"subject":"update constraint for python-tackerclient to new release 1.10.0","message":"update constraint for python-tackerclient to new release 1.10.0\n\nmeta: version: 1.10.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Yasufumi Ogawa \u003cyasufum.o@gmail.com\u003e\nmeta: release:Change-Id: If70bf0935108696900bd84cecd44c635036c2fa9\nmeta: release:Code-Review+1: Yasufumi Ogawa \u003cyasufum.o@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I411b6353abe6c7123e387505354937d171162189\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/94e06ae8ac1c4c0aa00e787dbf140a0ef10b3bae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/94e06ae8ac1c4c0aa00e787dbf140a0ef10b3bae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
