)]}'
{"id":"openstack%2Frequirements~831006","triplet_id":"openstack%2Frequirements~master~Id8c4a280ec8414a130f23b6e2678f7ec3a893c87","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id8c4a280ec8414a130f23b6e2678f7ec3a893c87","subject":"update constraint for osc-placement to new release 3.2.0","status":"MERGED","created":"2022-02-25 16:29:49.000000000","updated":"2022-02-25 23:15:15.000000000","submitted":"2022-02-25 23:13:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831006-new-release","meta_rev_id":"5468b99d02dfde3a07814cf4880c464faba33641","_number":831006,"virtual_id_number":831006,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-25 23:13:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-25 16:46:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-25 16:30:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-25 16:46:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-25 16:30:26.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-25 16:46:32.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-02-25 17:26:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-25 18:26:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"780a13a1891ff1398ffb263225e015738203c7d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-25 16:29:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b11652779fffa19333a3b94737427af872c4b44","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-25 16:30:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"14617b773c90b6d8f9b60de38756acae03296792","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-25 16:46:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d89093a570ecb3919d5baf8c6a0070f7d66e7f54","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 17:26:56.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9259673e729b489d811a46ce4554ad18 : FAILURE in 16m 08s","accounts_in_message":[],"_revision_number":1},{"id":"0b893877b9e6d810dfc1e92e10638821482953fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 18:26:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0ae141d7cb442e08336eb7cf40d7245 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8de4f8908e94624a1e6069e9e1ce123 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5636c0a2310d45f7a4cd321e8554709d : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73172101fa8149ab88f49cba12f5df05 : SUCCESS in 6m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3fde9ab891164fd59c743127132f3ff3 : SUCCESS in 6m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5d1782563a4f4f7dabd4813805c8e1ca : SUCCESS in 5m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/96da196eb53842d58efd0dce0569d538 : SUCCESS in 10m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7e13c0ed3d6943bc807a1320fa275c8b : SUCCESS in 11m 02s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a3eef24b754645b197ad570d3400e713 : SUCCESS in 8m 50s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7ac658f638d74bf1b16ed628e1ce1698 : SUCCESS in 13m 07s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/fb290cdda2f94e6a96e2a488a1c76e3e : SUCCESS in 8m 22s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/681fac27632f41979000a41fef7619c7 : SUCCESS in 8m 00s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f0b9dc82cde544478162ae8425dabd60 : SUCCESS in 10m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/83365e6ce8654276b611615526b8f1f7 : SUCCESS in 9m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/af9be02ee04d401ba6f1188d3faf594f : SUCCESS in 9m 00s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/69df3267c9084c4480fa2049c222b29d : SUCCESS in 13m 18s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/1648d079224d4af886ceb09c14e9c375 : SUCCESS in 6m 38s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/930a3f3703dc4e30bca246c666a82ce6 : SUCCESS in 19m 01s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6eba06e9ae7646b39c3616d529743e75 : SUCCESS in 6m 26s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b4d81d50657041bdacdbdf12420b1799 : SUCCESS in 23m 57s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d5bf1416e5ed498bb9242b32687bd497 : SUCCESS in 17m 14s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/08ff0288347d4a508eb354ac00e39a61 : SUCCESS in 8m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/daa68fd1a8f54fb8a1930d7a4ba9a528 : SUCCESS in 5m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9fd9a0ebcc48463ea53c386352a70b30 : SUCCESS in 21m 38s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/78e7f25091db4b89953b86ae79dd3a03 : SUCCESS in 8m 06s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/fc0318c341ab4573b8ce3c2dad9d31f8 : SUCCESS in 8m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/96413e1b418e487697a3f7f181677267 : SUCCESS in 9m 10s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b0e9867ec47a4120aebb54c72de340e2 : SUCCESS in 13m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df6ff64db3904161b64fb1c3482beece : SUCCESS in 49m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9a390eac5dd149ed814df480f6f9ad9d : SUCCESS in 1h 31m 03s","accounts_in_message":[],"_revision_number":1},{"id":"dfb353f33c49314280a5fb356347538bb753c90e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 18:29:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"49eebf24a93e4213fd5fbdb851c339df1d8a33ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 23:13:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31b2a19d47994f88bd3ddd9ecbede479 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03653439595d462a991ed97111708ae9 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fcde8239e654c4d91fec98599036d3b : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6ad3968feb94b619095731210757f14 : SUCCESS in 4m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c27c168484574dfab15f884ba83f432d : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/de5e9e4925b94bdda6bce6693ff32830 : SUCCESS in 3m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8eb65ea77059420296291c29b37e3484 : SUCCESS in 9m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c3cfcf06d68746ccbf3cd741df850fb5 : SUCCESS in 7m 47s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/cefbb13c41a94435aafa9d831c7c74c8 : SUCCESS in 13m 40s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/675a23b97b0b4e69878fa1c3d9b087cf : SUCCESS in 5m 41s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/32565e00a1eb478fb8c23d5aafcbc48d : SUCCESS in 7m 57s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4194fd1af4b74d63888345e381381ff8 : SUCCESS in 9m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f04f3f98189044d78bd7423b51fd02f8 : SUCCESS in 9m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c32798fdb7f94f8f884659ec1cba536b : SUCCESS in 8m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/544a401517b541ecbb6cd60345c80e48 : SUCCESS in 12m 48s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/740eed712b614ff49ee213caac40a3b4 : SUCCESS in 4m 12s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e7e11e1211294409839a3bac48e6cbe5 : SUCCESS in 17m 05s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/69d0d7501b1f444286f5afd1223c78ec : SUCCESS in 6m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/5c2e76efdfe74da090219a376db3d060 : SUCCESS in 27m 18s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f70a5026263d4387bb8af3ea56d64def : SUCCESS in 15m 10s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/24a6daf408894aca804d260f902daac5 : SUCCESS in 5m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e30308fa26304e01ae6a5ae61c4ab33b : SUCCESS in 4m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/08293abf86b044aa89fc3d1df8469a38 : SUCCESS in 19m 26s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/499a7331929f432885114f6a5738be25 : SUCCESS in 5m 34s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/cd80348f4f204edebbd6b840f8178e53 : SUCCESS in 7m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e58b6497de5543aa85607ae4eea36c5d : SUCCESS in 7m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7e353ab3c9564004a641e655ca1f82a0 : SUCCESS in 13m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03f8627db834454cbb2b421ad1ce530e : SUCCESS in 43m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e549cb17f4a4d318b1ea4d08c5c9668 : SUCCESS in 1h 18m 34s","accounts_in_message":[],"_revision_number":1},{"id":"d94ffd06ec38e9b40ad1f1355c301e30dd4678c1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 23:13:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5468b99d02dfde3a07814cf4880c464faba33641","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 23:15:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba3cc76226b44d2d95393580efce00d1 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b3992dcc5d4ba910ef242804fb04aaca3e06d91d","revisions":{"b3992dcc5d4ba910ef242804fb04aaca3e06d91d":{"kind":"REWORK","_number":1,"created":"2022-02-25 16:29:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/06/831006/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/831006/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/831006/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/831006/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/831006/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/831006/1"}}},"commit":{"parents":[{"commit":"39e0f20f9f3f15fb434f64bf9b4fddae895c6777","subject":"Merge \"update constraint for python-swiftclient to new release 3.13.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/39e0f20f9f3f15fb434f64bf9b4fddae895c6777"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-25 16:29:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-25 16:29:49.000000000","tz":0},"subject":"update constraint for osc-placement to new release 3.2.0","message":"update constraint for osc-placement to new release 3.2.0\n\nmeta: version: 3.2.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Icf8fa3ed6085f7a524305d24273b361b0f8d0633\nmeta: release:Code-Review+1: sean mooney \u003csmooney@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Id8c4a280ec8414a130f23b6e2678f7ec3a893c87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b3992dcc5d4ba910ef242804fb04aaca3e06d91d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b3992dcc5d4ba910ef242804fb04aaca3e06d91d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
