)]}'
{"id":"openstack%2Frequirements~831009","triplet_id":"openstack%2Frequirements~master~I39099e3ae494817495260c893e4bf8ce994a36ad","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I39099e3ae494817495260c893e4bf8ce994a36ad","subject":"update constraint for python-novaclient to new release 17.7.0","status":"MERGED","created":"2022-02-25 16:34:04.000000000","updated":"2022-02-26 11:34:54.000000000","submitted":"2022-02-26 11:32:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831009-new-release","meta_rev_id":"ceb5456d61730bdf86e5fa3a17d64b574246ae4a","_number":831009,"virtual_id_number":831009,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-26 11:32:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-25 16:46:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-25 16:43:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-25 16:46:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-25 16:43:33.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-25 16:46:35.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-02-25 17:43:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-25 18:26:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"28d889bba7d0a3699e36ce00c045435b1fb66afb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-25 16:34:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"705383c4780ca91ce7eb403f864ce632a8e69ed9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-25 16:43:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5074a4377c64b660dbb20357c4163a7c1e854195","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-25 16:46:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"44b594d98eaea5cea829a0e465a4dd34d52d7321","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 17:43:33.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/763167ebbda04d0987e83d795f6aa8b1 : FAILURE in 15m 32s","accounts_in_message":[],"_revision_number":1},{"id":"66c9ae79af7875188182534df2a2d501d99880fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 18:26:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/980e61adfa1b410ebbb2965c4da33e85 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d4c894e4d7a46b585b898ed05e547a6 : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3cd1790c440a4eb189fcaf31330d9147 : SUCCESS in 7m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b2d3eae6b7041588042a25ef28fe873 : SUCCESS in 6m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/928183ffaef34395b2d7a7a5e3e006f3 : SUCCESS in 5m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9c73d092f9f9482685570e46324c9db0 : SUCCESS in 5m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d23a909e673e49ca9f5da06d39c00d74 : SUCCESS in 9m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/775c5f830fcc4f958a19a781a9cc03af : SUCCESS in 12m 48s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a3119325369a47999e7a59ef5ab5159c : SUCCESS in 9m 22s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b2b7b97a3820462188a3b1de06e8518d : SUCCESS in 13m 08s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/535c526912ce4301aea10afd91587920 : SUCCESS in 7m 48s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/64e691a3f7004aa999fbe5da10e2c378 : SUCCESS in 8m 57s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/15fddfb637f54b968c86abc339b5ef76 : SUCCESS in 12m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e5d070ac7655493794a869531dd5759b : SUCCESS in 11m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/84b460ef6ba444fe817df3cf691efcbf : SUCCESS in 12m 01s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/388039fce14542f5b2dd0d70a99e8131 : SUCCESS in 14m 27s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/9522657381654f329c05588ab83dd87f : SUCCESS in 7m 17s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/afb0e62aa47245f3b232fd9accdfea42 : SUCCESS in 19m 25s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d3796626e4c34b3eadace4e6bb4d54c3 : SUCCESS in 8m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e9e593a2e4924adbb8ee3144ad4ca752 : SUCCESS in 23m 29s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2a63a3f90e7c461888bd9637672a8bfe : SUCCESS in 14m 45s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/04b36000866949bb9930f034d7fe0bf3 : SUCCESS in 8m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e33504a996244e969890b6102fe6f256 : SUCCESS in 6m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/13511330d04044c49f1ea43099f4ebea : SUCCESS in 21m 25s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/032704facc8643759af4569bdc721ab8 : SUCCESS in 9m 40s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0414a5fd5e774b4dac5d8fcf88a04b00 : SUCCESS in 9m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f9ba809d67c74c7db444bde98ff63f2e : SUCCESS in 10m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8fe17975f1ca4cb1894aef4993c42f88 : SUCCESS in 15m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01e3bc3d1a7a4f6cba0f0decfa6e3bd6 : SUCCESS in 48m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3ce4149912524ec09e71a674a72cb1b6 : SUCCESS in 1h 25m 58s","accounts_in_message":[],"_revision_number":1},{"id":"708a73c554f7a8a58839c5581e28555f8b45ccdc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 18:29:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"939aec069d1c7d8c566643235f212051cd882fa1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 23:17:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ceb39be3ff64d14a46e779a58f13b3c : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/167abd105b22425e8b15abb55dd3db8d : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91772166f1a744ff94e47c25c3250176 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26d9844641844a35b53b191785be96b0 : SUCCESS in 4m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2a83d761ba4745988cc07450a279d22f : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/78a6c030c26748069af1dba93d45a7a5 : SUCCESS in 3m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/31fe73370fc74a9ba03f5668a0521fc1 : SUCCESS in 9m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fc963c1b094c40c084e154c02429cca8 : SUCCESS in 8m 43s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ce975e6b3dcb43ae9b8a8e9e387999a0 : SUCCESS in 11m 30s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/5fd6ca9493a544ebbc029141607f29d0 : SUCCESS in 5m 17s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2e1a6fcb474b48f2b1f24dd29e8966b1 : SUCCESS in 6m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/abb6b9e3f7bc4f2fa3a2019eac2650ba : SUCCESS in 11m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9ce4db60cdd240d38f98b17bfb4f0f61 : SUCCESS in 9m 00s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8d5f5bb7b90c454094986a80de05ce20 : SUCCESS in 10m 25s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2a96abfb78414943b5e6ea88d4bf390f : SUCCESS in 12m 40s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a3845dc804d54486898e49173e6203e9 : SUCCESS in 4m 39s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9542442822a14c4f811ac6300f8d1884 : SUCCESS in 15m 37s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f90c8da2ff1c44eb9f72bfbe5f07c207 : SUCCESS in 6m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/de8f8470255441d3ac5f9266821a5d98 : SUCCESS in 29m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/aa70769ab0794900afc48b5462524d7c : SUCCESS in 12m 03s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4329ef4b4aef40939d31c115430aa028 : SUCCESS in 5m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b9b50f97462a44ada67813aac203ece8 : SUCCESS in 4m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b288d380704b4984b991327a317b1a4d : SUCCESS in 18m 29s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/8cda96a32d0048e3b7738403efa9fbfa : SUCCESS in 7m 03s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9fb6e245015b48e79d017530961ea51d : SUCCESS in 7m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ff606b7173e34baaaa37c7ca40694a1e : SUCCESS in 7m 06s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/75c1fe3af4fc4e849067c3a793a69b08 : TIMED_OUT in 41m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34f227c03af54a63958f6ce8f6672b48 : SUCCESS in 46m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/45ade9294f154393a549eb8d1f248236 : SUCCESS in 1h 25m 14s","accounts_in_message":[],"_revision_number":1},{"id":"29291d435d1c5b1b977c3019e9f8029707189c08","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-26 01:55:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3d922cbfd5d0c6492d1d7c180d503ba11e350c2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 06:22:24.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7785b85112f244c8850834b576a3849f : FAILURE in 18m 36s","accounts_in_message":[],"_revision_number":1},{"id":"4b7b9c24297ef63a81c41c60c589b59257d0ebe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 07:49:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16f1cebd0c424f79b2278ce5ceea798d : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd608ccae3ec41a5b809ca28f42df533 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fcb2740232e4d4280099cee1ea3bfe3 : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92faf6532d0e4861ba989a0dee92b33a : SUCCESS in 8m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d48a8afa236a483c9fcbd3a752e1321d : SUCCESS in 6m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ce066e0e627c4737979f74f9dbb5866c : SUCCESS in 6m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/44234c768bf84f09bfb0264012839c6d : SUCCESS in 11m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3182d006e8744eb5a10e9aa0cf53023e : SUCCESS in 11m 42s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d251b939776a4c91a9d6964b66ce3774 : SUCCESS in 8m 58s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/78b86b81666c467ea6fed9d39e92853f : SUCCESS in 13m 56s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/334543c52e084ab384240de4172c32c0 : SUCCESS in 6m 28s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/011a3df26d4741a1893d39db2553d91a : SUCCESS in 8m 13s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/507cf4dd33554f8cb6153fcb517e92a6 : SUCCESS in 11m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6e253966f3ad4a71be2a774b7aa0dfee : SUCCESS in 8m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/13221c3f40ac49218c2d8bc0a650bfba : SUCCESS in 11m 52s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/939265ef13cc4ab08184bf86c644db34 : SUCCESS in 22m 38s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/5060152bf6414e38b78824664a772541 : SUCCESS in 6m 42s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/502e1c6a89b24df38229516486cdfc08 : SUCCESS in 12m 02s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2ce150ecde7b48f9b4c52a21b4d75a7a : SUCCESS in 7m 33s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c88710430f48486995e821a523fdd907 : SUCCESS in 23m 12s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/fb33eed5f8ef44b9825d1da30e633745 : SUCCESS in 14m 44s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/19db87f0e32140eabf59bbb26e299560 : SUCCESS in 8m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/21f460c6ea2042d09d392ecc384673f1 : SUCCESS in 6m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bcdfd5fe0ae44290bb7e99e763e5c622 : SUCCESS in 24m 36s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a1d236c10b3e420799fe10b74da56c91 : SUCCESS in 8m 01s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/ec7b82801ef6460eb29bea94c71359b6 : SUCCESS in 7m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/77c8c49d5a044ca1b874291d1ab22a55 : SUCCESS in 8m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f88ff85f502b46d38c9fccb7e3da8fe7 : SUCCESS in 12m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/025ee32669564d7989b4415e769b8353 : SUCCESS in 51m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bcf6703ddf0d4e8fbdc4710fd5226ef0 : SUCCESS in 1h 51m 51s","accounts_in_message":[],"_revision_number":1},{"id":"d89fafd754de80f7ee6d594f5e466df487c6c461","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 08:30:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"219086f12ec58c6dee5ed5ec5d1728619e18a992","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 11:32:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/884f304808764e2abe4c43d21043e189 : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ee4395b8b014e0aa1fe88bf709b1305 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cf76ae4e07c42d990f6550dcfab808f : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/02ba2ac82d5c4936bd9dfe7bcdfe180c : SUCCESS in 7m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/580a293f4fee46bb950c2e883d3b5684 : SUCCESS in 5m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b8fd979bdc1d4d3591c6a013830c4de4 : SUCCESS in 5m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dc84572e370b4d07aa116a1fdb3cab7b : SUCCESS in 10m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2113b7960bf3455191846437c72e474a : SUCCESS in 10m 06s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ecc21f257dac46fab933f5b4d36fa24f : SUCCESS in 13m 37s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/166438a6c48943e086b44872116c1d18 : SUCCESS in 6m 46s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/00fcc1ab11f24fa7b77b9c0513c2f583 : SUCCESS in 7m 10s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/33ea5d0daa1d4439985187b560647259 : SUCCESS in 11m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e1be505ed43c4edfa63829adc13756dd : SUCCESS in 10m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/36315a8e55ad4d93b00e5d6ff896d4b7 : SUCCESS in 12m 00s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/62aac9c5076f4e73a37a081cf63bad17 : SUCCESS in 14m 50s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b1a49dd3bf5a4390ae0c0bb50b274096 : SUCCESS in 6m 24s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a48a2f717873475789608735fd6183e8 : SUCCESS in 13m 54s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a43e87944e674087ae364cfaf714ba97 : SUCCESS in 9m 02s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3faee51eee91449eac2330259d6c38e5 : SUCCESS in 22m 20s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/96d1d11937534109ba80d011a7145e8f : SUCCESS in 14m 38s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/874b73674ddc42b7981c17095d429885 : SUCCESS in 7m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/efad546a32ef42e088dea32b083a65df : SUCCESS in 5m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/397f3b4b3c5c4de7a7204bea9eddd8bb : SUCCESS in 21m 09s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a7b41f6c2862443386d03cf207bcc1ee : SUCCESS in 8m 34s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5d459efc202941c49e3a54ff2f1f8ff3 : SUCCESS in 7m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e6ab76ddc019413ab4fd4ef92dd24aa9 : SUCCESS in 8m 06s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1ec911fbd72a4f08962f5f85e0e2836c : SUCCESS in 15m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3419d54a06e6429c83ac86dfbea5c7f6 : SUCCESS in 43m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/441fa977243e43ce9a72b6953e89fb83 : SUCCESS in 1h 29m 45s","accounts_in_message":[],"_revision_number":1},{"id":"31879175932b8a4e8ca4a05b20a3214680f142ea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 11:32:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ceb5456d61730bdf86e5fa3a17d64b574246ae4a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 11:34:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5318383a1c3e4ee78b7b915ffdb7c474 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b487441c73362939823a37f527b1dceb719387fe","revisions":{"b487441c73362939823a37f527b1dceb719387fe":{"kind":"REWORK","_number":1,"created":"2022-02-25 16:34:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/831009/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/831009/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/831009/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/831009/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/831009/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/831009/1"}}},"commit":{"parents":[{"commit":"39e0f20f9f3f15fb434f64bf9b4fddae895c6777","subject":"Merge \"update constraint for python-swiftclient to new release 3.13.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/39e0f20f9f3f15fb434f64bf9b4fddae895c6777"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-25 16:34:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-25 16:34:03.000000000","tz":0},"subject":"update constraint for python-novaclient to new release 17.7.0","message":"update constraint for python-novaclient to new release 17.7.0\n\nmeta: version: 17.7.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I98239b7361db77609abf353aa3c8911e439ff418\nmeta: release:Code-Review+1: Takashi Natsume \u003ctakanattie@gmail.com\u003e\nmeta: release:Code-Review+1: sean mooney \u003csmooney@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I39099e3ae494817495260c893e4bf8ce994a36ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b487441c73362939823a37f527b1dceb719387fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b487441c73362939823a37f527b1dceb719387fe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
