)]}'
{"id":"openstack%2Frequirements~831028","triplet_id":"openstack%2Frequirements~master~Ieafaf237f8c735622aee3cf7b58982a40abcac50","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ieafaf237f8c735622aee3cf7b58982a40abcac50","subject":"update constraint for ironic-lib to new release 5.2.0","status":"MERGED","created":"2022-02-25 20:07:24.000000000","updated":"2022-02-26 09:34:47.000000000","submitted":"2022-02-26 09:33:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831028-new-release","meta_rev_id":"7e54c120ccf660ca207e9192689804f205fa137c","_number":831028,"virtual_id_number":831028,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-26 09:33:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-26 01:55:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-26 01:55:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-25 21:49:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-25 22:50:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-26 01:55:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"230adf16e0e6ba298f02a174c9625edd8915aafe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-25 20:07:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aec84d0c17dda73f4b1f2f9bd4c9492b0b16a92a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 21:49:00.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/34ba82cde35d4942afad58cd2ad94892 : FAILURE in 16m 16s","accounts_in_message":[],"_revision_number":1},{"id":"ce5eadb0853e38d7697800a4123498c49349e8a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 22:50:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bd9d37e74eb46ee8d0e9c249abd2c6a : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6883a1342e034bbe961aad3912ae7cf6 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddb956b029d74543875c1baf8e726a2e : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d05459bca7204ab786f5a67c8ef5d460 : SUCCESS in 6m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7fc8739420654ef68672d8094393d697 : SUCCESS in 5m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d19d8a5eab294aebb69b76ada59a5e88 : SUCCESS in 5m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a0b032ee97ef4e85826677b8c7615918 : FAILURE in 6m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/afd72c1bfa44456e8cc8f9e46d3fe6bc : FAILURE in 6m 25s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/aa1be8c81ce248f1bf5c052f4e40f767 : SUCCESS in 7m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/48a2e5035e4847adbd5362fe6f8f7239 : SUCCESS in 12m 04s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/536b49a62e364cd781b0803d2d2a7092 : SUCCESS in 6m 19s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c096c73e3b2446e1bc34aac1f3967cca : SUCCESS in 7m 45s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f6db92e92d8d404bb1e29ab542da966a : SUCCESS in 13m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/624880984db94134be0bdbc45f852ef4 : SUCCESS in 9m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f39a9732da034e07b99bccc6e016c753 : SUCCESS in 11m 21s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a048ab9bd71c4fa59ea43ad535582e67 : SUCCESS in 14m 23s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/69cc718f981f449e90463b6a057a80c5 : SUCCESS in 4m 59s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6f36d45f944040c187e10b09b75e0389 : SUCCESS in 19m 16s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/fd5861507e62466fb1928ffece146466 : SUCCESS in 8m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7ddc19c9e4554947b1a8be5d6a6bd84b : SUCCESS in 25m 01s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/66ece0feced84c63b0413fbd2d870ae0 : SUCCESS in 13m 44s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/41f36b8983e74027a996a6453c3547a2 : SUCCESS in 8m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d70a605dd3d740599d41889b04db1d5d : SUCCESS in 5m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/30b0701ca9f34f2fbd4a780f161f7245 : SUCCESS in 18m 56s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/0e5eec4b324149fe958c840cf5f9a5af : SUCCESS in 7m 13s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9cf82e3990964b92952b64da3d6f7354 : SUCCESS in 8m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/26a3d8bb533c47c88e5411c8bf629237 : SUCCESS in 8m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3fcba2f3fe354098aed85ecd6d801622 : SUCCESS in 13m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a22cd22e0b1d413d81a82e7cebcaf88c : SUCCESS in 53m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6fc0617cc33d47f093e1ba9c01655e75 : SUCCESS in 1h 42m 45s","accounts_in_message":[],"_revision_number":1},{"id":"6d56b340adc7792ca0f8e9e6e7e99d7e0e7385b1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-26 01:55:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dc3e655977bc2ecd99d23a183abfe036172f6bea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 06:42:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3641496a79de4b2780ba351c4eb583e4 : FAILURE in 18m 04s","accounts_in_message":[],"_revision_number":1},{"id":"29209cbb67adf2e13cf425ad5709cb63d8e156d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 07:08:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48a391f8649340898ebbb8dc2e132863 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a098314f60bf44699352b6f93cd359ee : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbd443ddcd3643418fe7f331f5096028 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/343cbee6642e496ba281347cdad6dabc : SUCCESS in 6m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2328ca25cae74d679ebd80453b18bcd7 : SUCCESS in 6m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2b21819c515b4c04b167390ee16017fc : SUCCESS in 6m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/626722b304a74fceb6f61a073f5ea522 : SUCCESS in 12m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fe6401fda77c4bc58c040daec4627c44 : SUCCESS in 10m 36s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/141471acb67e4aaa8593fc88a84ba804 : SUCCESS in 8m 19s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/64f0b3ee539e46b69a4b44d86c6ac59a : SUCCESS in 13m 03s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/55b32cc9f35d455e998c3c4be01f3e9c : SUCCESS in 8m 03s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f05c47dd6f584dd7b76cb6954477e495 : SUCCESS in 7m 52s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e886492b5b8c4aaeb78cd0150df29f6c : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e8ea6e47e67543c5964abfed365f6294 : SUCCESS in 11m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/43567a8ba6bb476bab43771273e88ab1 : SUCCESS in 11m 28s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/137c0d36c7264f0d93eda1782bfd36f2 : SUCCESS in 15m 15s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/3c7b2876b92d47349544995f38c68d31 : SUCCESS in 6m 48s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8e117da1a3e54ce6bc3097621b6b3ed9 : SUCCESS in 11m 54s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8841631604bf402e80478f22c810da5e : SUCCESS in 7m 30s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7e83e905f0784ceb8254ce774b324589 : SUCCESS in 26m 43s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1b8e37eda7584158a56e078f9704c2ac : SUCCESS in 14m 31s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/7ae80f9177e54a62a4c4a5666b048332 : SUCCESS in 7m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3ac9d527eec34c07954f7f8b47a524e1 : SUCCESS in 5m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/55984d44a59245c59c6fd1c7c801c4df : SUCCESS in 18m 11s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a53b0719596d412c8c63039e58b9ebc4 : SUCCESS in 7m 56s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c2e0149e56a942e29e16a9a10bb26f6c : SUCCESS in 8m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/26cdb88ec6f042d49bf920f1369793e5 : SUCCESS in 7m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8bc8a1e0e1994f72b9855698e539243b : SUCCESS in 13m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb2b0bcf488b4107ab35649a77bc7b63 : SUCCESS in 49m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b5d27d60663d4eef8b1afab079a1fb89 : SUCCESS in 1h 03m 30s","accounts_in_message":[],"_revision_number":1},{"id":"6d2ed95162bd16d279e8d941e9dc3c26a8f6f5e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 07:10:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"948d8e5ddaf6b29d04316b6d14285e77f91e47a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 09:33:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/679fa0d89c634ff1ba421011e9a81a0c : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e142beb3b658470d90b2a15d0f764941 : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/747b2d5dc4ab497f9895cc22216b933d : SUCCESS in 8m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12ab05a56b2941bf9254a19f6dad0076 : SUCCESS in 9m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a71c6b4e2094f7d9972be8c144b299e : SUCCESS in 7m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a443e02e827b4ddf9a6e088183d0e52f : SUCCESS in 7m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a91cd2566055487499915b6d741bc30c : SUCCESS in 15m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/880ad18be9fe416da2d2f8b612fba13e : SUCCESS in 20m 07s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/eb6cb8414bc240d59bdc5cb21326034d : SUCCESS in 12m 19s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/eb779e8908f8448b8b74af5395957a3c : SUCCESS in 8m 23s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/86f1223c749540eeb70f5b7be674b104 : SUCCESS in 7m 28s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e95b0969bb264274a7ed9a4de8d65bc0 : SUCCESS in 13m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/51a305e2839247baa95843454c558afb : SUCCESS in 10m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/575fb5df6e584e0ba697733a25aca37c : SUCCESS in 11m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ffa6ed321cca4faa8f0e21f2fc931115 : SUCCESS in 14m 05s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e09fde768c2d474581a04d01ef8fa7e3 : SUCCESS in 6m 12s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e2715d2d0bab459799c557264bdc3145 : SUCCESS in 11m 06s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/3120cbe6d3e34554a464e89fb9a1d27f : SUCCESS in 8m 07s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e0e914dfff3240378ab7c667564b2740 : SUCCESS in 27m 56s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/77e164f248304d8898bd69e5bdae8094 : SUCCESS in 13m 12s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a7ec382582d748ceb0c0880ba5435256 : SUCCESS in 7m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9eff2e00347a48c88298eeed17c63cbd : SUCCESS in 7m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5163ceadb69740e3b01a445a02b580e1 : SUCCESS in 19m 50s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/68ebb6b653b549a99a42d7240d9bddf6 : SUCCESS in 7m 33s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/92ff3ba0d6fc4f5384c4c1db4b35e395 : SUCCESS in 8m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3061c25ae24e42aea9f70d8623dda153 : SUCCESS in 8m 59s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6c96eebaf5a8476d94c6e783ba9a6d08 : SUCCESS in 14m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0001b2ae3038423dbee75774991b2b5b : SUCCESS in 55m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/53f87b3d50d8454f903e7ce25cda0746 : SUCCESS in 1h 18m 18s","accounts_in_message":[],"_revision_number":1},{"id":"35428b9e5ba5d3a1412f8077d46a9f68af5cd942","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 09:33:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7e54c120ccf660ca207e9192689804f205fa137c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 09:34:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e6669d91b4b4d05aed62b4f87d42f49 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ee0fcece8ac36888721d6de8964c27596194e23","revisions":{"9ee0fcece8ac36888721d6de8964c27596194e23":{"kind":"REWORK","_number":1,"created":"2022-02-25 20:07:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/831028/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/28/831028/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/831028/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/831028/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/831028/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/28/831028/1"}}},"commit":{"parents":[{"commit":"39e0f20f9f3f15fb434f64bf9b4fddae895c6777","subject":"Merge \"update constraint for python-swiftclient to new release 3.13.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/39e0f20f9f3f15fb434f64bf9b4fddae895c6777"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-25 20:07:23.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-25 20:07:23.000000000","tz":0},"subject":"update constraint for ironic-lib to new release 5.2.0","message":"update constraint for ironic-lib to new release 5.2.0\n\nmeta: version: 5.2.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Change-Id: I393721dd1a06d094391a257b12685e60e2d2ff7c\nmeta: release:Code-Review+1: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ieafaf237f8c735622aee3cf7b58982a40abcac50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9ee0fcece8ac36888721d6de8964c27596194e23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9ee0fcece8ac36888721d6de8964c27596194e23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
