)]}'
{"id":"openstack%2Frequirements~831050","triplet_id":"openstack%2Frequirements~master~I60e84e9c4ed0015c970e7e1ccfcaceae7a9e9e32","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I60e84e9c4ed0015c970e7e1ccfcaceae7a9e9e32","subject":"update constraint for python-manilaclient to new release 3.3.0","status":"MERGED","created":"2022-02-25 20:52:01.000000000","updated":"2022-02-26 10:32:28.000000000","submitted":"2022-02-26 10:29:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831050-new-release","meta_rev_id":"18d11d01f3a6ad047d05754d13aa8ccf8ac80d1e","_number":831050,"virtual_id_number":831050,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-26 10:29:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-26 01:55:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-26 01:55:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-25 21:54:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-25 22:35:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-26 01:55:50.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"20fb4aaca9556045106ae04bf10caf56641e5ca5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-25 20:52:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e93780a7d61ffb6e69941a70b999218c9e632d61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 21:54:46.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c5fa406c914f4bb1b703b2cf112b2203 : FAILURE in 15m 16s","accounts_in_message":[],"_revision_number":1},{"id":"dec4a5b652cf5679667cdfa01fd4fb21a730f22a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 22:35:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23212480a494480fb6af11c35f4aab8a : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9a7bfba4c4f4111932b27324ea78fa5 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fadebcefd8ab49fca9ba3ee8c12d375c : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/442059d3bc4c44638e611fe18ec46104 : SUCCESS in 5m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9a9ddeb765dd470cb521cda72cc79e0a : SUCCESS in 6m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0bf891be78eb497ab77d7f584603de25 : SUCCESS in 4m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d497a8baf3c74dc7b50eaaa913d9a385 : SUCCESS in 7m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/59fd25290ff74a92803986fa927d39ce : SUCCESS in 13m 49s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/c2cb75f4201742aaa33770e32b658a29 : SUCCESS in 7m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4b18dfd533db40c4899e308e6696cf76 : SUCCESS in 9m 31s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/10bcc299698e4a3a95b7dcb1cfcf9e12 : SUCCESS in 5m 40s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/fa3314c9c3654926a1733a0b45fd0dd5 : SUCCESS in 6m 05s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f13a0aa7f8b04aceb8758783bbd9c86d : SUCCESS in 14m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/014edbaa3850488586e6708760151c4e : SUCCESS in 7m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/43eed8ab9448419d910e616790c90f6c : SUCCESS in 7m 25s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/cc2e9d95742d4a3b858dac4d628da0f3 : SUCCESS in 10m 30s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/cef97fe555cb4b5a8ea0643296a3f200 : SUCCESS in 5m 48s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9900f6ee8f404ea1bb99af0521f9f82e : SUCCESS in 20m 31s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7d14dcda95de44dba049b9ca3559cb3e : SUCCESS in 6m 37s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/dbab9a9e30f4445e9d345b37665baf94 : SUCCESS in 26m 22s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/fc4f47d6a6704ea1b57aa332fc670801 : SUCCESS in 13m 12s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a4abb11d7707440fb853694fd6d6044e : SUCCESS in 5m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a4cc553d11654c07a93c0f10a292dd11 : SUCCESS in 4m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9eb2787c5867443bb19ea773ce12f5ee : SUCCESS in 20m 18s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e0127d2012f74ea0bc71f41c6dda6605 : SUCCESS in 6m 25s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c7bdf869c97748e59ae89e4045784b68 : SUCCESS in 6m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/46651b34a58446e585543b0191b274de : SUCCESS in 7m 48s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d4374f2d410244e8890c0ce2e0467581 : TIMED_OUT in 41m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/886062903b1143dd8a0227bc12e180d1 : SUCCESS in 50m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c639f5f707cb44828c15b5aeb474d481 : SUCCESS in 1h 21m 51s","accounts_in_message":[],"_revision_number":1},{"id":"16c3795113a7856be3dc55007246c0786aff719b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-26 01:55:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0ce0905d9cbabf5692b1f0ce2dbb1d3a20a436fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 06:44:53.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/53cf57d3f5924e9d94504c653d95deae : FAILURE in 20m 39s","accounts_in_message":[],"_revision_number":1},{"id":"777f7bf571657cae42516bf9db3635aa9e1f3326","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 07:14:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3c4ad22411f43a7a63373cb846b8636 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af005422d06142ec82fb1943e71bc2ed : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b397e941104643c1859ec33a760a1060 : SUCCESS in 6m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05c4a31491fa41be97b375484ba1d7d5 : SUCCESS in 6m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/599f078b831d40a7a1aa6e5ef0623348 : SUCCESS in 10m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e8873087bbad4f289355b6ede2308289 : SUCCESS in 6m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7015c0c722bc4d4283144ad6bb8bcd8d : SUCCESS in 11m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/93027b1e8705467c8b67341754c5cbb1 : SUCCESS in 12m 13s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/703b6d483061499282ef51711a1884c8 : SUCCESS in 7m 48s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/37fabe4964914eb59c1a8ae6eb2b8081 : SUCCESS in 13m 18s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/6b161283af1546a7867c2e2b46bdbd80 : SUCCESS in 7m 33s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f17ec88b562d4f3fb95d2c5e64cc3cff : SUCCESS in 8m 48s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b9be1125cf0044eda92af2edc455e604 : SUCCESS in 14m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e3d208de8018414e9f79df485797ec14 : SUCCESS in 10m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9e9f48236e044319adc993fd67f76b8d : SUCCESS in 11m 30s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/369baf90a6874419b12b370d7ce51459 : SUCCESS in 15m 21s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2d2acbf3172341cdb571c6cd63f17b4d : SUCCESS in 6m 14s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/456a275e9a274c888b5e6ebb880ee2da : SUCCESS in 10m 03s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/4699bdaaf27b4ec689d52b49ce9f9ad8 : SUCCESS in 7m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7f5f7d02a9ae48628d35223d40c7403e : SUCCESS in 26m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/31a83e1b8a2a4883bc435be848bd2aa0 : SUCCESS in 15m 02s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/7830f08e11dc4ceeab124d6da7011849 : SUCCESS in 8m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9dc6d1272aa3421a9f85d114b0fbbd2e : SUCCESS in 6m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2320e02107224d65b962496e67d3b10b : SUCCESS in 23m 57s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2f4245ce6c004c09a0a930ef07dd03fe : SUCCESS in 7m 30s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5e03941984b741b9afca23a91555e59c : SUCCESS in 6m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/26ed311af75449abad60ccb238900316 : SUCCESS in 9m 49s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3305ddcda0504c03b95de750e63fbf93 : SUCCESS in 14m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5199dfc0e454ad7a743eaac14e7126a : SUCCESS in 44m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/88c1135f6d4d45b28fbc20f442ed903b : SUCCESS in 1h 22m 06s","accounts_in_message":[],"_revision_number":1},{"id":"8a8bd293c9800a4391cd7f8b25be2f50c845003c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 07:16:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dc7f6f93e38bf8d470cc5c170331bac3833a690e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 10:29:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70085104f7dd4614954c66da633aba09 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e25595f6c2754f60bc36b46d08185e48 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb382dc8e6a94b6eb2bf78b54058b956 : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3861eb0e10ab467c828e9958de960f54 : SUCCESS in 7m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/650411990626406880004866f9f989e9 : SUCCESS in 5m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1e93d225ec4141d7b31a26f60d082983 : SUCCESS in 6m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d3f1273cb53b49a8967530223042bfcf : SUCCESS in 7m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e6b6ef2ecb174726b0b1455612c545cf : SUCCESS in 13m 18s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/49384652876c46bb945e66be3e06f293 : SUCCESS in 12m 02s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/4143b10d656149d1bba71feb636c7f4d : SUCCESS in 8m 16s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b614c37772fc45539239b1c03bd3afbc : SUCCESS in 6m 42s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/cf63d9ce401e4dcb89972f61b66e8ab1 : SUCCESS in 12m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/75356d9988124808b656b488d6f94d7d : SUCCESS in 9m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5991fdca65bf47e1af593e813fb947ac : SUCCESS in 9m 49s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d409f68ef302426aa50359185e5b1348 : SUCCESS in 13m 26s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/7c4fe959664045c798944d68dacd13e7 : SUCCESS in 6m 45s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/09742ffe12eb423cb42bde5306f1c4a8 : SUCCESS in 11m 16s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b58610353e464ff590698397dfbe915c : SUCCESS in 8m 15s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c2478a18228e49deaa79484ab16fdb28 : SUCCESS in 22m 57s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d6dee20c85984feaba80a1a5e7840b3a : SUCCESS in 13m 03s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4d7228c0acd8420fa74cf8aa3d70b2ee : SUCCESS in 6m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a2944b5bad0148d791781d20fe047fa3 : SUCCESS in 7m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7f2442b872ca4fd5ace5680ce389c257 : SUCCESS in 19m 04s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/8f2cbb1d10dd44909b634bd93f59585c : SUCCESS in 10m 27s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a34d9b15a24f4ad781d0790efb16adbf : SUCCESS in 6m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2a38f0865ca541d79487b3abe86c7719 : SUCCESS in 7m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/da0a24cada4d42ffaa7524db3fd466a0 : SUCCESS in 11m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ebc284b244749e69490ab76fb646937 : SUCCESS in 53m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/31729c8aac3441f08cc13ea58b711fdb : SUCCESS in 1h 04m 31s","accounts_in_message":[],"_revision_number":1},{"id":"03468f0114e460d4242342fffb7a3a08876a20fd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 10:29:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"18d11d01f3a6ad047d05754d13aa8ccf8ac80d1e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 10:32:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aca6e65ad8ed409fb4234f7b484ccbbf : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4110f78c15882f075f557eff654a95fd16c236ad","revisions":{"4110f78c15882f075f557eff654a95fd16c236ad":{"kind":"REWORK","_number":1,"created":"2022-02-25 20:52:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/50/831050/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/831050/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/831050/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/831050/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/831050/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/831050/1"}}},"commit":{"parents":[{"commit":"39e0f20f9f3f15fb434f64bf9b4fddae895c6777","subject":"Merge \"update constraint for python-swiftclient to new release 3.13.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/39e0f20f9f3f15fb434f64bf9b4fddae895c6777"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-25 20:51:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-25 20:51:58.000000000","tz":0},"subject":"update constraint for python-manilaclient to new release 3.3.0","message":"update constraint for python-manilaclient to new release 3.3.0\n\nmeta: version: 3.3.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta: release:Change-Id: Iae0a6b5d89982550f86a72b51b541f406b8b2b2d\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I60e84e9c4ed0015c970e7e1ccfcaceae7a9e9e32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4110f78c15882f075f557eff654a95fd16c236ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4110f78c15882f075f557eff654a95fd16c236ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
