)]}'
{"id":"openstack%2Frequirements~831332","triplet_id":"openstack%2Frequirements~master~Idc18764505a64c7b287f2782b09f4d4dd2de8d5a","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"Idc18764505a64c7b287f2782b09f4d4dd2de8d5a","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2022-03-01 02:45:39.000000000","updated":"2022-03-01 22:41:39.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4dd6be9de546d186147a718f85591d8f80335bea","_number":831332,"virtual_id_number":831332,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-03-01 04:22:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":-1,"date":"2022-03-01 18:53:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-01 03:02:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-01 04:22:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-01 18:53:18.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"c6bb632033f9e77e2b015c776b6ff1ee53bed337","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-03-01 02:45:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10091b4322bf7fa6a7c161c22962161b3b9b6b13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 03:02:57.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9de1fc2c77f545e3a5cca798f37452fe : FAILURE in 14m 25s","accounts_in_message":[],"_revision_number":1},{"id":"7768581f9e23e529c2501b1d2d6ab96b3d57cb19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 04:22:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce0b2238a5e041fb80c838e46ab4b4a7 : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cea0d1fe97874229bb518122ee18df22 : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35074e338212473d8e1eb5099e83896e : SUCCESS in 7m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/926b5d932e424e6495a9723d27d4d2fa : SUCCESS in 8m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/43a099b116fb4ca496b6626bce6809de : FAILURE in 7m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/44cb8c8b861846229e3cdf2045d6c3b5 : SUCCESS in 7m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a461d65b55f74609be470542cfdbdd05 : FAILURE in 7m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/312dd99253af417e80cb8a4330d5ac8a : FAILURE in 6m 44s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d7e2417158964de5be531eced5c740f2 : SUCCESS in 9m 12s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/846082134a2b47ada747a1c1fc83a209 : SUCCESS in 12m 29s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/60639ebca62743a985b35e06c6e34d21 : FAILURE in 7m 47s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/1ec0948087d24438bd6682eb05ead272 : SUCCESS in 8m 23s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/10bebf0e4b17443db5e19e6d12f4673b : SUCCESS in 13m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e26b0f5cb2a843dc82d5bf57c127dcf8 : FAILURE in 8m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/875c269d57aa4874b54a43a5c77ab406 : SUCCESS in 9m 37s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/30e0a608c35f443388ebde394c514bae : SUCCESS in 12m 33s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b262980ae494432c980a3a847ec8e3bf : SUCCESS in 7m 04s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/99ce26e3bb5d49cb89d932a1e79001bd : SUCCESS in 15m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2632cdf479b74955bc7d8a36fce57367 : FAILURE in 7m 47s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/492eab53dc33453da2fa43363eb544ce : SUCCESS in 26m 22s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8e715eaf8a9d43818ef79d04f499fad6 : FAILURE in 17m 04s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/daed84c72736424e8e3b38773e73b649 : SUCCESS in 8m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cc72c5f83abb4a0d97ba2359ec96b90a : SUCCESS in 6m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/10042bbb3b664c52b690ea03548414cb : SUCCESS in 21m 24s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d9b001bb436a4ee9939fc195563d2935 : SUCCESS in 8m 50s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/86d6979abee847a6bc77b54adbe90a56 : SUCCESS in 8m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7e7d124d023d4f4990a1f472ea3187f9 : SUCCESS in 10m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e86684c1bd484082b36733ce99f797d6 : SUCCESS in 14m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b40786f095c4789b322d19527141f6b : SUCCESS in 39m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb403876b7ea4617be97379c9efa9c4c : FAILURE in 1h 30m 44s","accounts_in_message":[],"_revision_number":1},{"id":"6ffdda13fe3502b2dab9c0e9e494f5f2f9b96742","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-03-01 18:53:18.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4dd6be9de546d186147a718f85591d8f80335bea","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-03-01 22:41:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ba8851fd10bf1021e67a280a4681375dc087097a","revisions":{"ba8851fd10bf1021e67a280a4681375dc087097a":{"kind":"REWORK","_number":1,"created":"2022-03-01 02:45:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/32/831332/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/831332/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/831332/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/831332/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/831332/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/831332/1"}}},"commit":{"parents":[{"commit":"48fb5c24764d91833d8ca7084ee9f183785becd6","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/48fb5c24764d91833d8ca7084ee9f183785becd6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-01 02:45:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-01 02:45:35.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Idc18764505a64c7b287f2782b09f4d4dd2de8d5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ba8851fd10bf1021e67a280a4681375dc087097a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ba8851fd10bf1021e67a280a4681375dc087097a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
